)]}'
{"id":"openstack%2Frequirements~702941","triplet_id":"openstack%2Frequirements~master~I904fa65a5fc677f513df92b0afac233e78607bc3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I904fa65a5fc677f513df92b0afac233e78607bc3","subject":"update constraint for sphinx-feature-classification to new release 1.0.0","status":"MERGED","created":"2020-01-16 17:59:23.000000000","updated":"2020-01-17 02:52:44.000000000","submitted":"2020-01-17 02:51:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702941-1579229460095-46a14a57","meta_rev_id":"22d3404760544f31b2288f551d2016477f575846","_number":702941,"virtual_id_number":702941,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 02:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-17 01:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-16 18:02:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-17 01:15:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 18:02:25.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-17 01:15:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-17 02:50:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e393d285260b60caca3c977096e6aac19505f795","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-16 17:59:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87b359b1c1a179f40a58d8ffaac8e3a7e1d09826","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-16 18:02:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"821e67715730524c49352c11b8bd4ed991eadfef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:24:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ad811674f64f8a81bdbb09a47711f8 : SUCCESS in 4m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8d7b4cc7ef1b4481bfb89422355e44ba : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd508865064495b8e47c51dd167e230 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4038bf7468fe40c39d8a6181c3eb6bd3 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6dfa8bcf5414c5fa7875a41ad83acec : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bba984aadac40a9ab918a9aa2bf5248 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69d4075689f149eebf325aa46a4d5350 : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c085d4cac4b3429295e28938d3930f8b : SUCCESS in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1f50dc774b0846688f6138f5b563134b : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7b6c4948a0144aaab46ea7501a7c9d09 : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6b149ff95f5e47e28fda2f7b1cf1d3b9 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22c5b957406c42c78102027682f2daae : SUCCESS in 7m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d94cac5e4d4c49fbb194a62f46fa8cfa : SUCCESS in 14m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/449e684defd64e5ab05a189c23ee100b : SUCCESS in 28m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/722f9f1a6e524f1e9ce1a32843bd9d5b : SUCCESS in 12m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ae1bd3735ed643d589a79fcd70f65779 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9b0ab0f77414f9ab6a1713288be8591 : SUCCESS in 16m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa4fa0f7e7a944c2afeb253f94bf6bd8 : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/95b427348a42490983cc1d9788329111 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b61756ea9eae49beb5d6eac7012a2995 : SUCCESS in 8m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7152ea1ed4984edf98eb3c282d902c1e : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e40b9b4d57d540c2946f864f46974993 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3ea0c9377694854b894abf3a519d616 : SUCCESS in 1h 13m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2e920b7274f4fc89927460a8868d8d5 : SUCCESS in 1h 11m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ee3a2ea91d384c15a967c40a088212bc : SUCCESS in 1h 50m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6d514c0da0c24fd39fb18d082677895a : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":1},{"id":"34c207779bb66d6704e69d855ffbb9695eaac34a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-17 01:15:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e575efa3559f0e9c13547b0b848e689deb11a52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 01:15:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3a89243f5a01bbf0ef5eec3b8e08907b29aa5d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 02:50:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf823eaa114445a9b9ce825f51a983d : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bcfab8ea81d4302ad3b4a8b72bdb84e : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d942db74e004e03ad5887ac9f1031f8 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee8bc2299f62483298a8118cb1856950 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd00d827cfec4992931d1e988a2fb76d : SUCCESS in 7m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac15fbdcb04847c4a5efdfa3d26f377e : SUCCESS in 10m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db9ee2150e1747a29fdc3ebdbd5202b6 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/447285cec266409dbf804a9f1da11fbf : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/303026feae03492e8965c8bf58e65d2e : SUCCESS in 4m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb87f84000594d20b50d14f34adf6dec : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72e4319ca10449e4a748478a345f2e42 : SUCCESS in 6m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f8344abd9af64b9d9a3069e50d4022fa : SUCCESS in 17m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c47ec6fff194ed389d6fdcddb0bb494 : SUCCESS in 30m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab834717c4a94d1697e1290b7dfbc0d1 : SUCCESS in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/250b52eccfc94b3aa29dfa8ead7ecfa1 : SUCCESS in 17m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43e4670ec5e14fa389237319bb33e067 : SUCCESS in 19m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6f2644986a4a41c5a3d3f991447b882d : SUCCESS in 4m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e0ce92c3efec4b8eb4aafb3715e78817 : SUCCESS in 9m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f5b5d9881c6469f8c96089375094a38 : SUCCESS in 7m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ad544d64a20846f58471bb0f647fb694 : SUCCESS in 9m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/679f1205a57c4ae8bb5f33c5c644f986 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd7972a6f34d475db4dd6fa17af29840 : SUCCESS in 55m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cb3e873e7f745a4a83be4fd8e396ee2 : SUCCESS in 59m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a1d38653c044e78a6719da0337d4217 : SUCCESS in 1h 28m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e56c84732eaa4ea7b65427570a2a93c4 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f4b185baf801e50e91c3e04500bda7fe20757443","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 02:51:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"31bd60dd4e2bc5b4acc5577a5b64c4ce421861cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 02:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5386dd7e234537b9d8eef30462c661 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49a4f96a69e5de21e07b51584f06493681cff4cf","revisions":{"49a4f96a69e5de21e07b51584f06493681cff4cf":{"kind":"REWORK","_number":1,"created":"2020-01-16 17:59:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/702941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/702941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/702941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/702941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/702941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/702941/1"}}},"commit":{"parents":[{"commit":"9ff06cb821f2875d7523963e0f394921a52e2709","subject":"Merge \"Remove trusty wheel build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ff06cb821f2875d7523963e0f394921a52e2709"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-16 17:59:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-16 17:59:21.000000000","tz":0},"subject":"update constraint for sphinx-feature-classification to new release 1.0.0","message":"update constraint for sphinx-feature-classification to new release 1.0.0\n\nChange-Id: I904fa65a5fc677f513df92b0afac233e78607bc3\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: I5c80a42fc61f2af7f18510f8275e25693b36e09f\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a4f96a69e5de21e07b51584f06493681cff4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a4f96a69e5de21e07b51584f06493681cff4cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
