)]}'
{"id":"openstack%2Frequirements~703023","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I7420dca0b3a1bd11db634629ef92e275368f50bc","project":"openstack/requirements","branch":"stable/queens","topic":"fix-queens","hashtags":[],"change_id":"I7420dca0b3a1bd11db634629ef92e275368f50bc","subject":"Fix testsuite: pep8 and requirements-integration","status":"MERGED","created":"2020-01-17 07:17:58.000000000","updated":"2020-01-18 01:49:57.000000000","submitted":"2020-01-18 01:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703023-1579312197451-9981773a","meta_rev_id":"b04a7ef43737aceda9f0869ab63b9da8a2646c02","_number":703023,"virtual_id_number":703023,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-18 01:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-17 11:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-17 15:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-17 15:26:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 11:07:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-17 15:26:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-18 01:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"963561c195f0e03b52e4affb752ccc838e45fe81","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-17 07:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd9ddb429ac6950e312479ac8fc7d65e7959deaa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-17 07:28:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0195d2b7f2c85b703196943b7e75900e20b5cf3d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-17 07:51:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37011a9188b1da532a34fc1bd3b27e7d0263cd5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 09:23:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/549d887910c44945b77e4ea914751954 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a18f06f46064f91ac7ea1943f1baf89 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49b41c0e2bcb45b4a1df843cf58d64e0 : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/98ead753a6bc4a3d8f622038fe3c6941 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2aa7caca4bcd43829ea0a13e4269ce65 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/395cdfa6436c4cfca0801b32c6ebf1a5 : SUCCESS in 3m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8ef547a4c219447d822bd4ae8b22deaf : SUCCESS in 1h 01m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb0a8b88ce70428695a9faeb89deac72 : SUCCESS in 1h 01m 35s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/38662faebaa44aeeaa916615fb177bed : SUCCESS in 1h 25m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/628007e700ae472683e9c5b92c75a5ef : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9d63e902f959e276f785feffc9706a8d5796ced0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-17 11:07:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08908c68eae7dbf0da0732eb425a35f60348d14f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-17 15:26:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"112e2878bb8d1e82d030efcb1d4b84587f045ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:26:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f184b09ea0f44c5c2ecfc7b802b85cfb7a4cd866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 20:09:18.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f46d873a24b8dc769bed9b40daa110be1f2b86f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:49:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e6d445ef474ba384f9ab3e22ccdc58 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a10e17305d74efaa4c0620717df2e4c : SUCCESS in 6m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f73154e31fea49d69f9f22e2bc6cf42b : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c8054ea5dd44ba7a28f4b62798cea8e : SUCCESS in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfe13383ca114e72969036463412f32f : SUCCESS in 6m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1e9ff9de2f5a49769beb54ca2dfae9c0 : SUCCESS in 58m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/382f7da52d12463d98e206f47f36c425 : SUCCESS in 1h 10m 00s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7ef93f65a185488fa0e0af502e7aaafa : SUCCESS in 1h 32m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ad4c2c30030c4e7e93eaf87c78e88510 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":2},{"id":"acb404376c22cdeea043f59be19e483880eaad3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:49:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb70fa451f366db180202cb6a778d6208713a2cc","revisions":{"97678b269ecbc2557b85bf6bae8f2b0306b761e7":{"kind":"REWORK","_number":1,"created":"2020-01-17 07:17:58.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/703023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/703023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/703023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/703023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/703023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/703023/1"}}},"commit":{"parents":[{"commit":"8811182c31cea08757ac70d8491fbb632c05dac8","subject":"Merge \"update constraint for ldappool to new release 2.2.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8811182c31cea08757ac70d8491fbb632c05dac8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 07:07:12.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 07:17:38.000000000","tz":60},"subject":"Fix testsuite: pep8","message":"Fix testsuite: pep8\n\nfix pep8:\nCap hacking, a newer hacking version includes newer pep8 and includes\nmore warnings that break build. There\u0027s no reason to update hacking in a\nstable branch.\n\nChange-Id: I7420dca0b3a1bd11db634629ef92e275368f50bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97678b269ecbc2557b85bf6bae8f2b0306b761e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97678b269ecbc2557b85bf6bae8f2b0306b761e7"}]},"branch":"refs/heads/stable/queens"},"bb70fa451f366db180202cb6a778d6208713a2cc":{"kind":"REWORK","_number":2,"created":"2020-01-17 07:51:56.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/23/703023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/703023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/703023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/703023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/703023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/703023/2"}}},"commit":{"parents":[{"commit":"8811182c31cea08757ac70d8491fbb632c05dac8","subject":"Merge \"update constraint for ldappool to new release 2.2.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8811182c31cea08757ac70d8491fbb632c05dac8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 07:07:12.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-17 07:49:00.000000000","tz":60},"subject":"Fix testsuite: pep8 and requirements-integration","message":"Fix testsuite: pep8 and requirements-integration\n\nfix pep8:\nCap hacking, a newer hacking version includes newer pep8 and includes\nmore warnings that break build. There\u0027s no reason to update hacking in a\nstable branch.\n\nfix requirements-integration:\nCap gunicorn, version 20 is only supporting Python 3.5 and newer (see\nalso I81658ed225d6da3b98fc63d8454562a2f27a2995 for train).\n\nFailures can be seen in\nhttps://review.opendev.org/702915 and https://review.opendev.org/702462.\n\nChange-Id: I7420dca0b3a1bd11db634629ef92e275368f50bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb70fa451f366db180202cb6a778d6208713a2cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb70fa451f366db180202cb6a778d6208713a2cc"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
