)]}'
{"id":"openstack%2Frequirements~703642","triplet_id":"openstack%2Frequirements~master~I5ac0a875ebeda9fdccc77eab76780ff9a4be6d3f","project":"openstack/requirements","branch":"master","topic":"bug/1860033","hashtags":[],"change_id":"I5ac0a875ebeda9fdccc77eab76780ff9a4be6d3f","subject":"Limit neutron-lib versions depending on the Python version","status":"MERGED","created":"2020-01-21 15:37:45.000000000","updated":"2020-01-22 17:01:12.000000000","submitted":"2020-01-22 16:59:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703642-1579712364153-a46f390a","meta_rev_id":"074a150e54b279538b6d872e2e4584f95a3fcf42","_number":703642,"virtual_id_number":703642,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-01-22 13:21:42.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-01-21 23:25:40.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 16:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-21 23:42:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-01-22 13:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-22 09:23:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-01-22 13:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 15:38:50.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-21 15:38:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-21 17:40:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-21 23:25:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-21 23:42:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-22 09:23:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-22 13:26:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-22 16:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bfd138cb7672abdcd97e8a1169019add6a357fb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 15:37:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8063650783a212707819c8a631c478e3951b94c6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 15:38:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"90139adecfbb41cf4c9d9235a951c1bfd29aa749","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-21 15:52:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nNeed to pick up global-requirements too.","accounts_in_message":[],"_revision_number":1},{"id":"012c5b4f02c6063c6ee41611f4a59e6a8a58cb28","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 15:58:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0c944969070faaadabb7274b589468c2d07a720","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 16:00:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fa977f95f1c9607c19c7b5c96c7bf7c3ab563d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 17:39:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7446664984a9846489324ba0bb82cf75b4472b9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 17:40:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a7b571cbad3f3a7ed35821d04d71a4fc634e3f25","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 18:04:46.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"87fa3e540f7b26812833877a1fbcff9f3216d07e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 18:06:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24c7f7cb80e61e2889150a5db5e258fc0977419a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 18:10:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab5fc486c5a58d5381754a844ee22b50212fc59f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 18:11:40.000000000","message":"Patch Set 3:\n\nOne this patch is merged, I would like to cherry-pick it up to Rocky.","accounts_in_message":[],"_revision_number":3},{"id":"f24e65007e9df291db4e714c0ec402930e579d15","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 18:11:52.000000000","message":"Patch Set 3:\n\n\u003e One this patch is merged, I would like to cherry-pick it up to\n \u003e Rocky.\n\n*Once","accounts_in_message":[],"_revision_number":3},{"id":"0399edc3e5621625a5099e5ce83809afc4266812","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 18:12:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd3691dcfb95d520da70257f7bf3c782bac5dfc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:45:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dccde73287ea4eab90c329defe1f871d : SUCCESS in 4m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a5b90afd8dd14a59ace5c5b424efb16c : FAILURE in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542c6edea4234f518a18b8195addebee : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc7f9a2b979d445b9486952fee8de26d : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5e6b8db339a4b5494d1648bfe843ba0 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/997e77a0579c42e3b1c1580134532c6a : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9c7b8f4a385489c9b57d19a0391162d : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f9d07ab546e4fbb91dfb5f16837632b : SUCCESS in 16m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3bd7417eddb14baea46a72ccb525d36c : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/34f28b8af01e4b55948716918d306255 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aff2bdc5ccc44fad9fc5313e68b37c3a : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/938107de4751471fa102d95cfca17bfc : SUCCESS in 12m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f0b3e3e7a294898af9f408d9e9d9340 : SUCCESS in 25m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/54db371785454b25b8b3ed4698160020 : SUCCESS in 30m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cfcdbbe8da3443c5baa9733ee7815bea : SUCCESS in 20m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0346976c051746e181d0190a252b23c4 : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/62e5871674554d40a81f0ad6a76e4d82 : SUCCESS in 20m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f61678bf07e94d99bebaba5d6e5e7659 : SUCCESS in 5m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e1ba6b442bab470c89408d1192528f7b : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef0f9c65cff44dcb863dd6a09518b5b6 : SUCCESS in 6m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1e7d332c9a8d4570a6eeaf3a0e9444ad : SUCCESS in 11m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0972707a6fe5436faa4e0af4d28a8e01 : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/057a884e4f864e91b1f569b5fe635dee : SUCCESS in 58m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3baba38f0c62452d87e428f9561f4e42 : SUCCESS in 1h 06m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4451185abff541bdb76143d8595c262d : TIMED_OUT in 2h 07m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7834e62d8a7a4e8db66d118ca6c70135 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3b4a3d548d1f48a68a3a121e859d61a4dc626d37","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 21:47:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"aae6fb303cdf63f7a8a105a8954b3c04e835242f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 23:06:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec5fd747d9442e9a4d679471be058f0 : SUCCESS in 4m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a82ad9f5477046c1948a5cd4e66d3b7e : FAILURE in 4m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954efd8498cc433d808e4563924c5121 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/480478d871984b2f8fac59c15b3e8bac : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db19f7d023564ad286cbeafa1496e40a : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c833066ee8fe471cb344730002490e00 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/31983aaaba284559b7df3d8b1bee3e76 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/34b6901ab4f94e70ad36af10d1f72cec : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dbd4e28900ea4ffba3f9b59b16557a20 : SUCCESS in 11m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b2002605b9384467b8bd2aa92ac5736a : SUCCESS in 4m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b25503a6960b4c0d8dfb1a74c4cf0287 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d3aff21a4cb45a7a6875d13f548eef4 : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2f4f045b15404b0e9e5614996ff5abc8 : SUCCESS in 20m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce96833e4e684da8b0ce205ed26c373e : SUCCESS in 26m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0ce6e1c6181427e934cd5b7acc51c82 : SUCCESS in 13m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8f1b1b0ea960407290a160083dccca62 : SUCCESS in 14m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f418a5b546ac483cb54deb95c259e12c : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f4eaf7a797f6409fa7c2f1b2f94dfb52 : SUCCESS in 4m 44s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/750ebbd04bf747fcbcae543c7d1640e1 : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c0d496d561f5406886919c125bf42a83 : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a91d3cad1af4494cb010b9e8023bfb06 : SUCCESS in 15m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/481b891a59ad422080db80ccfae9aad2 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf066693a1524a42976fc31123cae1a3 : SUCCESS in 51m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5c31012faed4ed0a423fa81b36ca2a5 : FAILURE in 45m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/64032544997a4faf811f6825bfa06a3d : SUCCESS in 1h 18m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7421fc933914469a9efbd7b41cbb7634 : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":3},{"id":"54c1b1d6fb4c2f6343b35eb881816060c59115cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-21 23:25:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"817f9f5be2220a68473ae353f6df481db2141aa6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-21 23:42:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6fe5afe755e21218f3b6488acc20af90ce9ecad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 00:52:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c064a69ac504c0685dab48998e97e3d : SUCCESS in 3m 59s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/db6c5cec0df94f05afde3e18b467c6db : FAILURE in 5m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88e5ff3df2f04b73afb71482830010a8 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5edc7fcce747078bdbb170460ed92f : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2f8750dc6e24887a63ded5f45b57f6b : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4df4f626ffe345daa100ace86e1baabf : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/325b615232fe4178b4b35da24b688e1e : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e518c21f1414d79bc65a0ac25d575e3 : SUCCESS in 8m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/569492c2cfbe4ac0a93a34176c349710 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4f625f0722e407cbf53473a81a7d998 : SUCCESS in 8m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bb0695d242e4b8793966bc6a1007823 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/975359e536364662a2e12ba1bc8fd55e : SUCCESS in 8m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/989699adef91425a84a5595fdbb155de : SUCCESS in 22m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4fd76465b59c4979b56c338533fde5ea : SUCCESS in 24m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09ef779684944dc48796027a81c0a561 : SUCCESS in 13m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/57a27949e7094c028a1774804288cb6a : SUCCESS in 15m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b14041d91294889a5c380928b0e66a3 : SUCCESS in 17m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cf55f652e7dd4588b1454bb9edf7b52f : SUCCESS in 4m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7ecf18f84f154fb5906cb0ae50c84c7a : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12b2cce27d9b4989b28c08f52c067d19 : SUCCESS in 8m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0c81a850917a4be0b9516f3ede516885 : SUCCESS in 15m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75585f2a5bac4f31bef349639c1ec65f : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8ac11dfb45448fa83b1b85c82aad46e : SUCCESS in 1h 04m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f4559a9629e45f2aa60eb310c41370a : SUCCESS in 1h 10m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/40d34f81402c477086ea2f0bd5d9f71b : SUCCESS in 1h 26m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/122d577efabd41b898b2e89e9b00d56e : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":3},{"id":"73783247dbd4688dbe032a40428b29e4ef0063cf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-22 09:23:51.000000000","message":"Patch Set 3: Code-Review+1\n\nIndependently of the devstack fix of bug 1860033, this sounds good to have this listed","accounts_in_message":[],"_revision_number":3},{"id":"8238e9d1217088593f73bfa5f8b39965005eca82","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-22 12:02:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"103945d90f8b3e62b7d21c837a75d872ca0964d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-22 13:21:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f50a68024565fc78d997a15db43f86870c77d90b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-22 13:25:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"959e1ae10763f4d6467529a6a3ebe1299ae27bb4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-22 13:26:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI see there is no branching prior to the 2.0.0 release, so I guess that kind of answers that.","accounts_in_message":[],"_revision_number":3},{"id":"d0632391ff129b4c7bb3833e8ef5aa5e38feff3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 13:26:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ae48535c7723152862491bd601d43ea62212ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 16:59:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b331aec73c9640ff9b68c463fa695a50 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72326cfcd55049b288fe3075be001a33 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc54bc388364289b7eb2fc057e11a7e : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62df07f44b204491b556f96f06550706 : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c9413c6fa3f4a79a22953c511874433 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e76aa0996b3f40fcb2ee272c34b5c30d : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/50c3b79a13da44799768cb278d739988 : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a2b5accad8b41ddb72f44e3be03a509 : SUCCESS in 10m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ba7096e6507418aaaf4ab9474eb53b4 : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f7db7a0093e74b5db9851bf5c8d24b23 : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d37faf049adc413186d3b1023b181902 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e9be02052174ef3be71f5d07a12b4e0 : SUCCESS in 22m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d90abcbff8b241a5b461771b363ee6c8 : SUCCESS in 33m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1da7a06ed35e4e929f2c21e62330a4c2 : SUCCESS in 19m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d8056ccd074e424890b13636b93358e7 : SUCCESS in 22m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/916561e119054224b2e444d2f40ef306 : SUCCESS in 22m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fe623af813a94c5da7dfd2ec20cd19fd : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7fc392e9b94c403e8bfdc535063105ae : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0ccf9c93d5c4a199781d56196c19a17 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f988df991d4c4f309a09209700f6e278 : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee709057d3104bdaa91a0ac9b6aa4105 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c52dd2634f34374b0486d98437d6c10 : SUCCESS in 1h 02m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4369adbb6ae649b8a83904f92d0484b0 : SUCCESS in 1h 14m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7ab02365c48436e8bbefd9d8fcb7d4e : SUCCESS in 1h 32m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/51d398583ac34ae6928af82baae1f8b2 : SUCCESS in 1h 12m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e121f9210b973c81174b0baf412023cd54cd37c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 16:59:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"732b03917eb2c5d727b9487ca6a4c891aeefa852","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:01:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4135dcbf22a4457382129fbe80e876f8 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1f38f1241509dd2fe40983e4c6acf7bdbdd2593b","revisions":{"5ee77bb39f01886d73b87818c3a3304d7a98e743":{"kind":"REWORK","_number":1,"created":"2020-01-21 15:37:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/42/703642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/703642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/703642/1"}}},"commit":{"parents":[{"commit":"d22ceab64af4878eb2def57968740eb15cb16517","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d22ceab64af4878eb2def57968740eb15cb16517"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-21 15:34:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-21 15:37:43.000000000","tz":0},"subject":"Limit neutron-lib versions depending on the Python version","message":"Limit neutron-lib versions depending on the Python version\n\nLast Python 2 working version was 1.30.0 (1.31.0 has an API error).\nNew 2.0.0 version dropped Python 2 support.\n\nAs seen in the related bug, some CI jobs still running in Python 2\nare trying to install 2.0.0, returning an error because this\nneutron-lib version is not supported.\n\nChange-Id: I5ac0a875ebeda9fdccc77eab76780ff9a4be6d3f\nRelated-Bug: #1860033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ee77bb39f01886d73b87818c3a3304d7a98e743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ee77bb39f01886d73b87818c3a3304d7a98e743"}]},"branch":"refs/heads/master"},"8410fc62bcb6f94d8e9d554589d4bb7d73ee5f4b":{"kind":"REWORK","_number":2,"created":"2020-01-21 16:00:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/42/703642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/703642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/703642/2"}}},"commit":{"parents":[{"commit":"d22ceab64af4878eb2def57968740eb15cb16517","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d22ceab64af4878eb2def57968740eb15cb16517"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-21 15:34:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-21 16:00:42.000000000","tz":0},"subject":"Limit neutron-lib versions depending on the Python version","message":"Limit neutron-lib versions depending on the Python version\n\nLast Python 2 working version was 1.30.0 (1.31.0 has an API error).\nNew 2.0.0 version dropped Python 2 support.\n\nAs seen in the related bug, some CI jobs still running in Python 2\nare trying to install 2.0.0, returning an error because this\nneutron-lib version is not supported.\n\nChange-Id: I5ac0a875ebeda9fdccc77eab76780ff9a4be6d3f\nRelated-Bug: #1860033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8410fc62bcb6f94d8e9d554589d4bb7d73ee5f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8410fc62bcb6f94d8e9d554589d4bb7d73ee5f4b"}]},"branch":"refs/heads/master"},"1f38f1241509dd2fe40983e4c6acf7bdbdd2593b":{"kind":"REWORK","_number":3,"created":"2020-01-21 18:10:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/42/703642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/703642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/703642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/703642/3"}}},"commit":{"parents":[{"commit":"d22ceab64af4878eb2def57968740eb15cb16517","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d22ceab64af4878eb2def57968740eb15cb16517"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-21 15:34:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-21 18:10:20.000000000","tz":0},"subject":"Limit neutron-lib versions depending on the Python version","message":"Limit neutron-lib versions depending on the Python version\n\nLast Python 2 working version was 1.30.0 (1.31.0 has an API error).\nNew 2.0.0 version dropped Python 2 support.\n\nAs seen in the related bug, some CI jobs still running in Python 2\nare trying to install 2.0.0, returning an error because this\nneutron-lib version is not supported.\n\nChange-Id: I5ac0a875ebeda9fdccc77eab76780ff9a4be6d3f\nRelated-Bug: #1860033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38f1241509dd2fe40983e4c6acf7bdbdd2593b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38f1241509dd2fe40983e4c6acf7bdbdd2593b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
