)]}'
{"id":"openstack%2Frequirements~703863","triplet_id":"openstack%2Frequirements~master~Ifde31a76818524fe41f07b27f497972707381263","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifde31a76818524fe41f07b27f497972707381263","subject":"update constraint for tempest to new release 23.0.0","status":"MERGED","created":"2020-01-22 19:50:56.000000000","updated":"2020-01-23 03:16:14.000000000","submitted":"2020-01-23 03:14:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703863-1579749281559-5aeb02ef","meta_rev_id":"372129e3ea8ef24d155da2dd416cc97e825b5fcd","_number":703863,"virtual_id_number":703863,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 03:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-22 22:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-22 21:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-22 22:22:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 21:12:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-22 22:22:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-23 03:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc335335e5775ca2238091f619c59c39e07475fb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-22 19:50:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb7a095921f9568e5aa5bf094a3bc488a5baa4fa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-22 21:12:53.000000000","message":"Patch Set 1: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"5e9471c2dd5d618369f351c2fa59b73a0f9995e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 21:32:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ca6d67b9d1486dafcb9ce5864ccd46 : SUCCESS in 7m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/26e6a2dc4ada46f5a2fd73a8c08a0db6 : FAILURE in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3f6c8c22ce43a58b44cd6faa57279e : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3715d1e7fb4a55b707b4a2c9386ea0 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9093e887312d47338d4ea8bc857060dd : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8f87eb5fefe40e98a2ad64db8526fce : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/911abd18a24e4f1eb3c76e06207b0dd5 : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/477954a8151942218687aab086c65eda : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b55355319ab44ae4bc309b8b34f185c1 : SUCCESS in 11m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/47571f436fff4748b48519998793a02d : SUCCESS in 8m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c86c5ee58a3403c8d1ceae50fb83d1e : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fffdd356cfd84cc988a625489c980927 : SUCCESS in 8m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5030cd9492bc46fc89fe335dc551a4a2 : SUCCESS in 19m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc3c9b7e549c45f6a572e8d773c7fdc5 : SUCCESS in 26m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/82f211cb05b54a47bbffbd6b8ca99ab6 : SUCCESS in 14m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/91fc6dbd04bb42248b2db93a6848bae2 : SUCCESS in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da35ab0fac2442dbb1620cd2747db960 : SUCCESS in 16m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cfb50a40ffde40a2b96e0997bdca4900 : SUCCESS in 6m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b396e1bda13e4c0e8a313b39cb0052d1 : SUCCESS in 7m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11fc0bb00db24577aa82e3738c578d3e : SUCCESS in 9m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d9598f6cddbe4bfcad113f0c50aa9353 : SUCCESS in 11m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/61eda870b8184501b743923e89497720 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7bd7e28b0a241ea8ea4c346be013f67 : SUCCESS in 55m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e1708f0001b4c3cbfe708e8125cbb5c : SUCCESS in 1h 16m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c42ba318b194e0eb35d4966dec3d990 : SUCCESS in 1h 26m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2a1f107ad9364fc2b63b27d6b182b221 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2dd17a3572efaea7edc7230dcf13603e5930804d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-22 22:22:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b66fdac845d87746f95b55e19f4e1eeff9d42593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:22:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2be2a49b784db66ffe83b25a0da5c9660349522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 03:14:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699fe9b9e1d14f51b8b7dc66841f6b1f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4745664e144475869f091cacb08373 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1f527b28bf4b3bb695c406672e2c4c : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8543cf8b096b47e7b6c4090f499192d0 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/385ba10ddca64e2ea8fccf395403c2f6 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c165b55bb754ee29c794e3d9b977647 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c75a93e35e04222aa1eab6144cf5e39 : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/49ca3183c9c04ff2b1ec0211decb28e0 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/64dc1089acab4820ace3d9e27a4d450d : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/384efc33b2494d04a99433cba7d8f21b : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/603067acecf74e97a255f8cca3fd030e : SUCCESS in 6m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ccdc921683284388aa46fc6d61de4f8b : SUCCESS in 13m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/39814c4f6743454da74d0be0bf0e1dee : SUCCESS in 22m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/99d03fcaace14e82824101fc6525ac3e : SUCCESS in 11m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/24654a46e95f4f81bd4234e9b1d4e31d : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88335e89b6584317b8f7a34a874b8f04 : SUCCESS in 16m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/639d17aee2464355adb8849416f39571 : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/597dd8718f144753853e6a98cfc4fcb2 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38764793e820497fa73f879486e55973 : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0ba9dc2d152a45dc82e1531ce8c8b8dd : SUCCESS in 11m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bdfed7b0f764e86a270bce9368028e1 : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54d37e30c7cd496eb72e8692be9fbec2 : SUCCESS in 56m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/39a12642af9a4947bcc1a0e05ab72a1e : SUCCESS in 1h 29m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f7f0bfb4493047dc88fd57e8661f01d5 : SUCCESS in 1h 28m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77b89456d60f41a0aa1058106c9b87ed : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8ec2c09df7648f8b38e9ff2a007a2a4eb67a8668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 03:14:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0535b99478725f2ffdb9aacbe60f97024fa6c69a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 03:16:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88aa514ccf884dd780a6339b7ed9dcaa : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"554cc4b8601d0f53984f515235f58b59f309aa1f","revisions":{"554cc4b8601d0f53984f515235f58b59f309aa1f":{"kind":"REWORK","_number":1,"created":"2020-01-22 19:50:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/703863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/703863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/703863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/703863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/703863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/703863/1"}}},"commit":{"parents":[{"commit":"1f38f1241509dd2fe40983e4c6acf7bdbdd2593b","subject":"Limit neutron-lib versions depending on the Python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38f1241509dd2fe40983e4c6acf7bdbdd2593b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-22 19:50:55.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-22 19:50:55.000000000","tz":-360},"subject":"update constraint for tempest to new release 23.0.0","message":"update constraint for tempest to new release 23.0.0\n\nChange-Id: Ifde31a76818524fe41f07b27f497972707381263\nmeta:version: 23.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Commit: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Change-Id: Ie7481aa19e7493082233ef65f8759de41460898b\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta:release:Code-Review+1: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/554cc4b8601d0f53984f515235f58b59f309aa1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/554cc4b8601d0f53984f515235f58b59f309aa1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
