)]}'
{"id":"openstack%2Frequirements~703975","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ia1274cde009d8250cebd8182f9f0b2821e8779f6","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ia1274cde009d8250cebd8182f9f0b2821e8779f6","subject":"update constraint for horizon to new release 15.2.0","status":"MERGED","created":"2020-01-23 11:38:49.000000000","updated":"2020-02-05 23:39:46.000000000","submitted":"2020-02-05 23:39:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703975-1580945986658-e685ee51","meta_rev_id":"e0fefb1622ded8fb5db6556b13259382e4e5004b","_number":703975,"virtual_id_number":703975,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 23:39:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-23 17:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-05 21:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-24 13:40:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-05 21:27:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 17:27:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-24 13:40:41.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-02-05 21:27:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-05 23:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f7b4e43edeecd9bd3caba78fa97f78be90f85aa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-23 11:38:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"062bbbc29de95ec6d12d9f56a939c2b180cde771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 14:18:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ced7dbf39f55488c8ed5084569e214c3 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ed899c021f4eb386c2295621d78ad6 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b0ae7f7011b473e85150580ac97ded2 : SUCCESS in 7m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b73d7b0fbe3431f90e121b8f2076b86 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90dda7d44d0c425982d109dea6081b88 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b9b977a580149f1b532d0729b01eea1 : SUCCESS in 10m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13916708112d4984a2177f009e906ed8 : SUCCESS in 5m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e81578509ee34941bcadef42f2cb2339 : SUCCESS in 8m 54s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f67bc5b83d09442289b6246ec16bc2c8 : SUCCESS in 10m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6944b018b47244c981d6d67dfeb94c04 : SUCCESS in 11m 05s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/91b324d5ab2e42d5a0ba5b69380ffa47 : SUCCESS in 14m 39s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7ffd18218e034006bad6a69a564f50ae : SUCCESS in 11m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aed111e2cb004073bffbfc059f684695 : SUCCESS in 10m 28s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7bc88cf288014e339862d0eb55ffd2a0 : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/914b6a54ae804484b2188cd80cc923c2 : SUCCESS in 11m 52s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cbd3ae0a97b843c785129bdba00ba6c2 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bba29a1b277142fca7f1dfa3e11096f8 : SUCCESS in 8m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/53958f4ff9554dbcaf2862d052545076 : SUCCESS in 31m 45s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/42fef327af9b44fb8fd1ae9d6987ee3f : SUCCESS in 28m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/dd856fbbd2934ea4a2c7d8db4aee9ad0 : SUCCESS in 22m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cc5bff1847f74a1f89f671084a50dca6 : SUCCESS in 25m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/db3ccfcc18b54d1cb09f2364aa04c032 : SUCCESS in 13m 40s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6d44f6e4bdaf4367bf230d11f4de93a4 : SUCCESS in 14m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67e0f59337ce4fca95c502cd2cf9a663 : SUCCESS in 19m 28s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a04d7480bb4e4539b199332083f75583 : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/19f6fd959d40406a861665333aed1ee5 : SUCCESS in 10m 05s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a0f74460919442fa941c676d6f7752e0 : SUCCESS in 8m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/482bb6e5d456494c9c93c072b714f449 : SUCCESS in 1h 11m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e2936088d81d4a21b5ad764213a40467 : SUCCESS in 1h 55m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/36059fadf62942f68ee41059978d9a01 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"167e6f1f1bf3b6e53c349a62b2327961caccaf2b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-23 17:27:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2fbc094127c9d4374fd4adf8089a61d9c9fa4ed","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-24 13:40:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4ffbdeb3759e1dd1d9292bb4ab98188b421fead9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-05 21:27:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7b16cec63a6ec25854304b7b8b6ffc2120c4e4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:28:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a8455e249617bf053c4c8e60c39ad543f834ca7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:39:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c89c388486cf3fd5bb96f7ed69b020598f3ed9e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:39:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1f0395e8b846f4a084f08e902eff49 : SUCCESS in 8m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cb15b857a6340d8a0b0bfdc6d66cf61 : SUCCESS in 10m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe20498005aa4d2ba5a41885092c94ea : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c87ec0d7ced54ee8983259449faaee4f : SUCCESS in 8m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19166b3b358a4dfb9af446d55dbac280 : SUCCESS in 10m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d732e1a6ee23450b97a9d11c759ef51c : SUCCESS in 10m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e68dd45d133041ccb94319b2ef2dc8f7 : SUCCESS in 26m 46s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b212002b97d24b079cd685658d87360a : SUCCESS in 30m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/71f8ff6987f8439dbe3830c036b15cd3 : SUCCESS in 17m 48s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/aa804aee771a4348b49d4080bbad6d0d : SUCCESS in 22m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c94e66976d7c4a9581fa7409deb2d041 : SUCCESS in 13m 03s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/daeddd748bfc4801a82668f641d495c2 : SUCCESS in 11m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ad4420a7350a473c8127c4df4c1774ba : SUCCESS in 14m 16s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/978467fe8c924dd2a91c7f0f81a0ec96 : SUCCESS in 13m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17b4512580714674970a2e0a2f0ec9fb : SUCCESS in 9m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d87ed85be87c4ff19ca493a64ab7492f : SUCCESS in 25m 32s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/75c5aa123c8247efa1282d5de7112b9b : SUCCESS in 34m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/80d38dbdec4d47a9b61d8170f6df1582 : SUCCESS in 33m 53s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/895cbe204b8d42ac8f74ca360417bc3c : SUCCESS in 32m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8e41c0a341fc4a798b9fa1722c6691ac : SUCCESS in 15m 00s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/99fc3cbb9d0842b6a85027536785cb14 : SUCCESS in 19m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14ecde8abf4f4edc83e96ad227345ec8 : SUCCESS in 23m 41s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2f2dad878f004335b592c2bd73987480 : SUCCESS in 7m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3be828ed1f57405ebdc5138e85d9572d : SUCCESS in 11m 06s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/790430149d3846f998146eccae022613 : SUCCESS in 10m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c1edbe9c4ea4f2aabc003501fcb408c : SUCCESS in 1h 14m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/135058ab4d93492e9d6c1dade33c22d5 : SUCCESS in 1h 43m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/71f350435ab64abc97fec6dc0822e36c : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"004628dfd4a86568af03e9624542a5d3f7c56749","revisions":{"004628dfd4a86568af03e9624542a5d3f7c56749":{"kind":"REWORK","_number":1,"created":"2020-01-23 11:38:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/703975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/703975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/703975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/703975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/703975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/703975/1"}}},"commit":{"parents":[{"commit":"c739b9d835d5ca0c3a882f462f7647fe90585595","subject":"Merge \"demarcate python-memcached requirements\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c739b9d835d5ca0c3a882f462f7647fe90585595"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-23 11:38:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-23 11:38:48.000000000","tz":0},"subject":"update constraint for horizon to new release 15.2.0","message":"update constraint for horizon to new release 15.2.0\n\nChange-Id: Ia1274cde009d8250cebd8182f9f0b2821e8779f6\nmeta:version: 15.2.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: I42fcd8fc2702ab8e31e05aac0a3b00e62a7933e7\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/004628dfd4a86568af03e9624542a5d3f7c56749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/004628dfd4a86568af03e9624542a5d3f7c56749"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
