)]}'
{"id":"openstack%2Frequirements~704176","triplet_id":"openstack%2Frequirements~master~I32e746460bb0ea00d9fde15e1afa9b9bc66e5338","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I32e746460bb0ea00d9fde15e1afa9b9bc66e5338","subject":"update constraint for ovsdbapp to new release 1.0.0","status":"MERGED","created":"2020-01-24 16:58:58.000000000","updated":"2020-01-25 08:47:20.000000000","submitted":"2020-01-25 08:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704176-1579941918568-5aeb02ef","meta_rev_id":"e669f73edfdec738f0a8774638a7b2a7ec7c6f0b","_number":704176,"virtual_id_number":704176,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-25 08:45:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-01-24 19:13:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-01-24 20:33:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-24 23:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-24 17:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-24 20:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-01-24 17:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-24 23:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 17:16:25.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-24 17:20:15.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-01-24 19:13:16.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-01-24 20:33:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-24 20:50:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-24 23:57:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-25 08:45:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d86044f36470b58af22ca858a48950a9598709a2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-24 16:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdadcdfaedd4e99a7def5c186444651575d0067d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-24 17:16:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b8537725ebe90866da425ea40736a3e721c11b36","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-01-24 17:20:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c9a75f563cc820c39ea2d02ea7da662b5b0b207b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-24 19:13:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e6f4baa5c3d98543cbcf9284424b6bc44fbccbc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 20:31:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9546b57b48a048a790080530448b52e2 : SUCCESS in 5m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6599f88db81e494bbd12dd7fa4461bba : FAILURE in 7m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76cd34e9ac5847b6b7ea557d8424eefb : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e1fdf69e7884b63a387003d3a76dc4b : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/576123487b9b41dbb9a612aeb797c055 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ff19546dae404683acb2eabe1afa52 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/462c7e7fc0874ab0a13199dfbc6de5a0 : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c1102bf411f491693546ba1833539c3 : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a07ba8a0b70744b39b648d01c43fd3e0 : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd5bd935e37c4278a1320a90ccb0c9a8 : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf5bbc73ee064e479c9cf969374977d7 : SUCCESS in 4m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1bfd8bc7cc0547b7971b680e427f9d40 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76caec20c52a45988733b6d72ae72d2b : SUCCESS in 8m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/832bc7869f294a3ba512905b8cea181e : SUCCESS in 24m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d11f6fa0f6c7490da550dc4d8d98dbb3 : SUCCESS in 31m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c7e21a33a85f424fb88ec0eb8b5d9649 : SUCCESS in 13m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/58ff2999bfec4b6cb9bcb5c7ad922873 : SUCCESS in 15m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f12cf44c4182444b89c34f129144edcf : SUCCESS in 19m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af050c50657448b482b98a62bbb07a30 : SUCCESS in 4m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0247278a913e409eb588db55ef5387b1 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcc81e2e9a464cc497f72dce75484834 : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a0458bf911754c94a0bd3b3c0e775705 : SUCCESS in 11m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef1d3d7699ca496d87879cb0cb6ec320 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90d66087d37f4980b3936ed1cc35f962 : SUCCESS in 1h 06m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32b46f07cb854a8f854c4aac782d505e : SUCCESS in 1h 03m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b3723a086b947c19f94ec42fa3f03c3 : SUCCESS in 1h 37m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eb7b850b990d458288b6e5a2ddc7ac60 : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"627cc30852f6faf42bea3eba47ba8cb5379424f3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-24 20:33:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc0cdbc1bfe1dd3b3ff7f882cf6520a32b530f19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-24 20:50:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a7a31996c2c844a8f1ab659a1b5a7a9b3c9538e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-24 23:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc962f45db6b1914db9fcdc4336fb1297a21a813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 23:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b9943d0aa465b130fe47b08c69a16e9ac5e6f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 03:25:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d56be116eb24613be003d9b0eb77f94 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa3499a2b0f481393846ad0d0788d81 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b1114d2c1849a595cf657a0a29b438 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdfdc0ea2cf54672ba914b3875ae867c : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cca3afcf6a684977adce8637bba50c50 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b64a980590674c7ea764b31e23d2ac93 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81c7d939c4934ec7abd605d15a25f870 : SUCCESS in 7m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2cf5a73e030c432fb06060c9c1e759fb : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/304999e0bcec48a69ecfac76b3d3169a : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5c21f6759673404f891152f6b2c862c5 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf4766d9abc446ac9f4f03a1b3330c05 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9c7bc89e8a5c4f75bef53ce13816741e : SUCCESS in 32m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8910ad653b5742c18d0dd9c4d9265b88 : SUCCESS in 33m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7bc514af21e04ea8ad16763eb58e0acc : SUCCESS in 12m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f8a2771a7bfa4fe8baa0c8a947454d48 : SUCCESS in 25m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88144f5d5ec64342b838832b0e667e5a : SUCCESS in 16m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b12431ef25fc4599b4b49478c6106165 : SUCCESS in 7m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/76b87f6e873f4337acd34632fa2cdb1e : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64134dd0991b419b854d6a0f379970b9 : SUCCESS in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/59640e3bd7df409c911f2beb6bd82a47 : SUCCESS in 18m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d04cc995a6e54c02b48c744b3d2f6489 : SUCCESS in 10m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79306e27d14a4f9887876c584ca14c4a : SUCCESS in 54m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fcec586ba5b4a3fa19bf52c7881aa1d : FAILURE in 1h 05m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5d59808881ca4278b78bd634c18fe2ae : SUCCESS in 1h 32m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/de2f6a8d2536497c8ba8a57552d2e282 : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"06694e4b2dc2a201526cb1e75dbe5d9484689001","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-25 03:54:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d07ce17ee89b17a21a2faa2b5d1f11a536a4b6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 05:26:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95d80ef79a44f518ba09034bbda0168 : SUCCESS in 3m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ef4e721ef5cb411c86380332ce9d6383 : FAILURE in 3m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d10a6d7ed0e4d55b2efe037fdebc5ee : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a6c10e059045b0895af16e1d0ac056 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42ec51d53ffd4d4c8e72c270fbad55de : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be6106fd0d074df39b106573fdeb39eb : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96092596107d44a1b46f202a3e9d399d : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/427fb22c209142c3a549428aff5d1417 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8d3fc9096804a9baf3268feb48c832d : SUCCESS in 6m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bab5d3319ee043d9a5d0c7ede51a4c92 : SUCCESS in 11m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed08ae8074054b0eba533cceddd84bfc : SUCCESS in 4m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e9217a3553f44679c4c917d2437e7e9 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e04a949f7e2d4d70bfb68ab0b0770673 : SUCCESS in 6m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bd7300b416e54097927c164b1ef1e6a3 : SUCCESS in 11m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce2c603f6c0d4177a73e2f96d86bba53 : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc796f8350a04ad0b9dff432e56822dd : SUCCESS in 11m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7942748bad3743d8875cff1d48286866 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7972935ffa084ff58fe6d5a5b109acfe : SUCCESS in 15m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/42ae7de803814baaa37b5356b0bb9f4a : SUCCESS in 5m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3b30027a6d7f426f9efffbf13f438df7 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acac6ce1c12546f593103cd165b458a3 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8ea70be7e41842d8a60076365c4e87ab : SUCCESS in 10m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69030779f0934e30b524a512af2caa02 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0449de9b3d994c0da219cf73caabd7a7 : SUCCESS in 57m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d64a17f7a35d42138a782aae84149c60 : SUCCESS in 59m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/40263ff9a77f4af591ae0914d8cac64f : SUCCESS in 1h 30m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4dfa93b935d449b5ade6c7e8a4c1c1b5 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ee4420fa42311000e2367e6625ea6ced49d9756d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 05:26:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48afac1951fcc5e8cc1e1724cb9cba90e58867e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 08:45:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d4bbc76c074103933f1a5b5c82169e : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0f3819bce746a19908dea8dd5c6872 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/032ae8dbb2164d8482ddd3063a200f27 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c85638ad9cb54e8cb07156cd77c85ef9 : SUCCESS in 10m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c63a268aa42436e96dc7b9b1fecc5d3 : SUCCESS in 8m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e593343853b44a9898b8db2255d289f0 : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64af34bda358427ebf145f4fe757b641 : SUCCESS in 22m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ee8382aa5b464347bfaa8858e466168c : SUCCESS in 14m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/29cbc85aa8cf4a7da2e5e3032e9ece50 : SUCCESS in 7m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac317a1e82ff4155870ab18c3e161cea : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41ab4218dc97447c85909caec3edeaa8 : SUCCESS in 9m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/150388ba548b43d58e56d7788403f803 : SUCCESS in 13m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f7543e3bd74a4eb5bc594e2040dede24 : SUCCESS in 29m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d21c779f689f4c0a95bd274ce45526f0 : SUCCESS in 19m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cf9efde35da640eabd60c5cba581ea09 : SUCCESS in 17m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a00eea772aaf4a4fb901cc3a40ffbf67 : SUCCESS in 19m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ba7fdf2bc1174daca4fd568e91cec2e2 : SUCCESS in 7m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1bd25c5e189b44759e4a49b30b933ad2 : SUCCESS in 12m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11ab8054c50248049b2072833f2f913e : SUCCESS in 8m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6ae032c09ebc4c3ca3daf615a36e49dc : SUCCESS in 12m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0eb7d1ad7ef043e8b4929aa8f88b9b5c : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a52f2741b2a94db49c843e1c56a542df : SUCCESS in 1h 09m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5520eeb83dc74202aef26eb66ec1256a : SUCCESS in 1h 27m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0e7dfe1c355344aa9126680a428e2a10 : SUCCESS in 1h 48m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f954e20f45a4425b8029f39328ce8f7e : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"259ae143068dd86a5b4308ee7259e9982feab501","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 08:45:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b72d2bf2180ad0242034d8202edd4253f256fc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 08:47:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2556048638bd41a194a3ff3b8524f83c : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d2f28e66f54c15e2b0aeafa7fb59e43d43586a2","revisions":{"5d2f28e66f54c15e2b0aeafa7fb59e43d43586a2":{"kind":"REWORK","_number":1,"created":"2020-01-24 16:58:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/704176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/704176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/704176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/704176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/704176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/704176/1"}}},"commit":{"parents":[{"commit":"554cc4b8601d0f53984f515235f58b59f309aa1f","subject":"update constraint for tempest to new release 23.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/554cc4b8601d0f53984f515235f58b59f309aa1f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-24 16:58:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-24 16:58:54.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.0.0","message":"update constraint for ovsdbapp to new release 1.0.0\n\nChange-Id: I32e746460bb0ea00d9fde15e1afa9b9bc66e5338\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nmeta:release:Commit: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nmeta:release:Change-Id: I1c663556b99f61864d365acdc40661aef93ac27c\nmeta:release:Code-Review+1: Flavio Fernandes \u003cflavio@flaviof.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d2f28e66f54c15e2b0aeafa7fb59e43d43586a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d2f28e66f54c15e2b0aeafa7fb59e43d43586a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
