)]}'
{"id":"openstack%2Frequirements~704263","triplet_id":"openstack%2Frequirements~master~Icfd5ef0c2d9db262838ff8de2bc6822fae5d6b8b","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Icfd5ef0c2d9db262838ff8de2bc6822fae5d6b8b","subject":"Use zipp 1.1.0 for python 3.5 and lower","status":"MERGED","created":"2020-01-26 22:45:13.000000000","updated":"2020-01-27 17:39:25.000000000","submitted":"2020-01-27 17:36:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704263-1580146562690-46a14a57","meta_rev_id":"4005f644a44ad142aacdf5f5e38244a2e5483c2c","_number":704263,"virtual_id_number":704263,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2020-01-27 12:39:40.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 17:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-27 13:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-27 16:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-27 12:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-01-27 07:28:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-27 12:31:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":1,"date":"2020-01-27 13:58:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-27 16:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 07:28:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-27 12:31:57.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-01-27 12:59:25.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-01-27 13:29:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-27 13:58:55.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-01-27 16:00:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-27 17:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24cc86ba8bba6a2ac12ebc4905e40df5165b5dda","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-01-26 22:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70f4c387ac71095475b064838ab94cf3bb069ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 00:23:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c67a36fca6f4b87917d7999a6512306 : SUCCESS in 3m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/32fec4d6d1194166b68fa18c021c2d23 : FAILURE in 3m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a714d5ad7645d3b419c93d118bb929 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60a90a0d68004e90b560d3cd4e08b7e9 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0418a86b647e4bdeb190f7af1882e85e : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5adaa30bccda4846a5275daf945a0b5e : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f34aee3fc2a74dddb5dd7882dd86be41 : SUCCESS in 2m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce3bf8f247b94284933a25f67a4bac7e : SUCCESS in 2m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/08b49dc066f542669daf4bf9db2ef70c : SUCCESS in 7m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2d6d94509a343408dd6809be2d0d849 : SUCCESS in 7m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cc7549b8210747ea9de85a805bfe2d2f : SUCCESS in 4m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2ba6e6b0b1114bedbfd3b1fea04ddcae : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0d85314a02d4ce5ae9de398897dd78a : SUCCESS in 6m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0dfde616bccd4dd19998ad43e877ee70 : SUCCESS in 13m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dfc2cf24a5ea4f6b8290971c889c5307 : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d06a53dd39954190b2a689442f1af995 : SUCCESS in 11m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9dad21fb89bc4772988e5e1fe22d2e14 : SUCCESS in 12m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2adb7f0f55e4d57b5ba7a69f9510994 : SUCCESS in 15m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0e7e38965d44481784360cff8d7bd46b : SUCCESS in 3m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e0dd1ae396d44a4a94d3b55c465159e : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68a188def28549ed917cab56b52ef0f6 : SUCCESS in 5m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/764461a12dca40d68781054d76c2ab13 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1de8f3fd42a54fb78c8c795172228a25 : SUCCESS in 8m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31fce09ef5c6492babde0694ef03f759 : SUCCESS in 52m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7fcc26c0d3034702b1c58434d9119241 : SUCCESS in 1h 03m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3c65f7bcef1a4089812e3b71a2a234c2 : SUCCESS in 1h 33m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/843a18280da04ed2b4107c7d3d10aa39 : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dbd8c612653e8cdee1f53a926adc796e5fa67367","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-27 07:28:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fbec66997fef8011e0653aca138142b8b9328c35","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-27 12:28:30.000000000","message":"Patch Set 1:\n\nthe issue with this is that zipp was b0rken by proposal bot in d898b71fc0fb2eb75eec869a4d552a9d75ac9a46 - it might do it again ;-)","accounts_in_message":[],"_revision_number":1},{"id":"31ecb702990292098781aa088df3757c10979fe4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-27 12:31:57.000000000","message":"Patch Set 1: Code-Review+1\n\nfailures observed without this\nhttps://zuul.opendev.org/t/openstack/builds?job_name\u003dtempest-full-rocky-py3\u0026job_name\u003dtempest-full-queens-py3\u0026branch\u003dmaster\u0026branch\u003dopenstack-tox-py35#","accounts_in_message":[],"_revision_number":1},{"id":"42587f2764b117383708d7da677de9a999fdfdfc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-27 12:34:35.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThe issue will affect with pip \u003e 9 and projects specifying minimal python like https://github.com/jaraco/zipp/blob/v2.0.0/setup.cfg#L20\n\nzipp-2 introduced this:-\nhttps://github.com/jaraco/zipp/commit/05a3c52b4d41690e0471a2e283cffb500dc0329a","accounts_in_message":[],"_revision_number":1},{"id":"796d317b4d1481178fc60928dff7a9532d7abc0a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-01-27 12:39:40.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e the issue with this is that zipp was b0rken by proposal bot in\n \u003e d898b71fc0fb2eb75eec869a4d552a9d75ac9a46 - it might do it again ;-)\n\nit might, unfortunately, I don\u0027t know where the bot is running and how to fix it :/","accounts_in_message":[],"_revision_number":1},{"id":"c99a27c25d00dc4def17cb957be74f63afc48039","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-27 12:59:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0b9b4dfde03cd988c3aaf838cd14a0ef91a7c27","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-27 13:29:34.000000000","message":"Patch Set 1: Code-Review+2\n\nNot sure if we also need to cap this in global-requirements.txt, but this looks happy.","accounts_in_message":[],"_revision_number":1},{"id":"4b56411237fd04dcb7a293ce0d17c165dd02a10a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-27 13:36:06.000000000","message":"Patch Set 1:\n\n\u003e Not sure if we also need to cap this in global-requirements.txt,\n \u003e but this looks happy.\n\nbest follow-up but jobs I\u0027ve seen so far consulted u-c","accounts_in_message":[],"_revision_number":1},{"id":"7d41149588fb71c5fe2d22dfdd6afb8c8b2c7ac6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-27 13:39:24.000000000","message":"Patch Set 1:\n\n\u003e \u003e Not sure if we also need to cap this in global-requirements.txt,\n \u003e \u003e but this looks happy.\n \u003e \n \u003e best follow-up but jobs I\u0027ve seen so far consulted u-c\n\nYeah, this will make sure jobs using this upper-constraint will not get anything higher than what is set here, but my concern is the proposal bot may try to raise this upper-constraint if it not capped in the global-requirements file to explicitly tell it to keep these older versions limited to 1.1.0.\n\nI can\u0027t recall - we\u0027ll see what Matthew knows about that.","accounts_in_message":[],"_revision_number":1},{"id":"fb106f5bd06668cda3caecd63934f17d254028ad","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-01-27 13:58:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ceb424e21e94cd288b0039120282020e44a427d","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-01-27 13:59:37.000000000","message":"Patch Set 1:\n\nFunny, there\u0027s no py35 job to test it","accounts_in_message":[],"_revision_number":1},{"id":"bd801d3629ae7020262bce2e5cecfab37253468f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-27 15:44:24.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e Not sure if we also need to cap this in global-requirements.txt,\n \u003e \u003e \u003e but this looks happy.\n \u003e \u003e\n \u003e \u003e best follow-up but jobs I\u0027ve seen so far consulted u-c\n \u003e \n \u003e Yeah, this will make sure jobs using this upper-constraint will not\n \u003e get anything higher than what is set here, but my concern is the\n \u003e proposal bot may try to raise this upper-constraint if it not\n \u003e capped in the global-requirements file to explicitly tell it to\n \u003e keep these older versions limited to 1.1.0.\n \u003e \n \u003e I can\u0027t recall - we\u0027ll see what Matthew knows about that.\n\nYeah, proposal bot already wanted it b0rken, that little traitor.","accounts_in_message":[],"_revision_number":1},{"id":"acd8f86b686580837b7d124f7c95e36e76e1ea02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-27 16:00:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6422c01445f7a594da1e0d70a27eb74f46e68df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ab5f928d8810575cbb330b757b39578a02b384a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-27 16:02:43.000000000","message":"Patch Set 1:\n\nnot in global-reqs, so not sure if it\u0027s gonna be rolled back when the bot eventually does work again","accounts_in_message":[],"_revision_number":1},{"id":"49a13ddac807507bf90d1c58f7fffd32bc187eb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:36:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93838c0ace2d47289c0869efd6b729c9 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a448495fb9424b76973bb552276424ae : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0679fd077da0482cb5de7c9eb658e06e : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29be429b55644858b8733e661ddf81a6 : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d952ce46b3284896917246db8eb11aa4 : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09c585ee1bfa480d875f44e04b43bac8 : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9be577fa6564c698b800abe99980452 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/88f1d7332ada4b63b9d5804fd5734596 : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b1730ff8cc74940a638d224ec4c1f70 : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8b9949abfafb42f79ba3927336025ffe : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f2a020c3ec04a489ffb67ca0d19cc4f : SUCCESS in 7m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6366d3995984f5fa485533175073df4 : SUCCESS in 22m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5758671cfe204907a20d15f1b321566d : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13971572b1d84125b895e518c12457a3 : SUCCESS in 13m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/42cd7c891e914b3787ed7884c56fa7bb : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4842547a083749b58f904cb7b3a75161 : SUCCESS in 18m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db33eda5afcf4633b3f61b497c982f8a : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f25c2c8f0d6547a09c8af14e44774006 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c8c2da38c85484e9dec78e433913631 : SUCCESS in 8m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b042cbb93b9d4e148b1117fe2cb63fa7 : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dfdde706bf92466989fb0627d5903098 : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/188d668f19db4e8899f0f8de719bf006 : SUCCESS in 1h 01m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f16a1def8fa4963a237174a56c12e2f : SUCCESS in 1h 08m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b7834f5f9180416aa0549bd719a0f0fc : SUCCESS in 1h 19m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a5eeb219979e4a38b7fc3b801f7599c9 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1c194e52e38aa46d7b871161ae45ef0f9dae357e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:36:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4005f644a44ad142aacdf5f5e38244a2e5483c2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:39:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00bf906db5e4e1f85c0ac660b7e7173 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95cc881198705a46f131cf2a52f9cb31fc004c43","revisions":{"95cc881198705a46f131cf2a52f9cb31fc004c43":{"kind":"REWORK","_number":1,"created":"2020-01-26 22:45:13.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/63/704263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/704263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/704263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/704263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/704263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/704263/1"}}},"commit":{"parents":[{"commit":"d4a75d617e93ffa4a76d841a86a8bdc942ab48d8","subject":"Merge \"update constraint for ovsdbapp to new release 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4a75d617e93ffa4a76d841a86a8bdc942ab48d8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-01-26 22:40:48.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-01-26 22:45:14.000000000","tz":0},"subject":"Use zipp 1.1.0 for python 3.5 and lower","message":"Use zipp 1.1.0 for python 3.5 and lower\n\nVersion of zipp 2.0.0 and above is available only for python 3.6\nand higher. Therefore python 3.5 and lower should use the latest\navailable version which is 1.1.0.\n\nChange-Id: Icfd5ef0c2d9db262838ff8de2bc6822fae5d6b8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95cc881198705a46f131cf2a52f9cb31fc004c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95cc881198705a46f131cf2a52f9cb31fc004c43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
