)]}'
{"id":"openstack%2Frequirements~704525","triplet_id":"openstack%2Frequirements~master~I5762673966187239ccbdc021215f158faa9efdf1","project":"openstack/requirements","branch":"master","topic":"kafka","hashtags":[],"change_id":"I5762673966187239ccbdc021215f158faa9efdf1","subject":"Revert \"confluent-kafka 1.3.0 fails to install\"","status":"MERGED","created":"2020-01-28 10:08:30.000000000","updated":"2020-01-29 04:48:48.000000000","submitted":"2020-01-29 04:47:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704525-1580273238559-04bc3352","meta_rev_id":"347ef017cc9eb40e633ebe521a456ce90a181044","_number":704525,"virtual_id_number":704525,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 04:47:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"date":"2020-01-28 12:04:10.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-28 15:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-28 16:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-01-28 10:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-28 10:11:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-28 16:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-01-28 10:08:30.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 10:08:30.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-01-28 10:11:23.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-28 10:11:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-28 15:46:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-28 16:57:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-29 04:47:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c9aea64dbcd9ba95b3bd39a1cf87c18083180fa","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-28 10:08:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"023852aacee840c5e6de95d84fd8410069d302d7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-28 10:11:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"306af9732ed3e43e4b833365ec68f388942241ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 10:11:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4276555ca697928703ba2de856685b4854758b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 12:00:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2fee1c7c0f48b6bd2df06d03f5eb05 : SUCCESS in 5m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1201e28769a94536abd08e082e320a84 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef43b2562a154543802b550137ce0728 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943dda0da758475b99a2c7a7fd6f996a : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e0834afb3d54ec2a3e3349b508e4df1 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a9d3a65f5e4cf391fc4321919816a5 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37549cb8d1a0402d8fc401dddd9d2ba7 : SUCCESS in 6m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50daa9e94f7d40d3b35212847072337a : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a8dfd0d6803a46b0bb455ff5e47fce1a : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18ea3db4adf64ee8b14cb19e2f58f5b7 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aae8668d5af04588a43135599330784c : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee915491057c47b38a91d8c19b379489 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92db4e4422b54803bb0c805c1ac8985b : SUCCESS in 9m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f4de2ce12c0b418696e401c067c68ea9 : SUCCESS in 14m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ff368d32845c4cdca56ea80f0aad4a77 : SUCCESS in 30m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a1f3a5a35fbf4a1f890d910f4413fe0f : SUCCESS in 13m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/68b6f607edcf4d55bb9c994b6f57fc48 : SUCCESS in 17m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7639f19c1a3a482294746d816a221bb8 : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6b50499b103d4cd68ebf0c2c221aa917 : SUCCESS in 6m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/66dea9e72dcb416bab582bdd765c5021 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c7f4318a5ee4ab9979f5a4eca17dea1 : SUCCESS in 7m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9a8cfaa0f67a4d7da24366905055d4df : SUCCESS in 11m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2176fd2fc34044209fdef98973a9fd22 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9b321daf6824d738fead05892ff155a : FAILURE in 58m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6def001857244ae082cd2c1fcd714ab8 : SUCCESS in 1h 12m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b2fab513af374f3685f833aa6c229900 : SUCCESS in 1h 45m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be5352f314954475bfd00490b9ff6bf9 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"65455cdb967ed08e2c6a47b69a2fd0c084b1137e","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-28 12:04:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58575089d95b06b092a380e9bad727ba8ad196d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 13:47:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33627ef5c7bc420ba2fffaa581fa2b33 : SUCCESS in 6m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d40fbc5b07334a54a06665c7e197f607 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2659feb2ba4f5da34a242eec919a4b : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9f89109c904ffb91c323078fc8182f : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bd17ad7afb14875b196dab6a6b843ce : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1da3ca09aa84760b0f92cd6e484a7a3 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0460e2bb8481494290ace66216fba919 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab1dc9e1bd9d437b8edd43800daa0dfe : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea7b6ccac452410795584f0b545ec643 : SUCCESS in 10m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/afa5b3191c6b4ed2860d68bf2f44ccdf : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd241d016bdf464b839ddc2bc56c6f35 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7d97b1ce33b94144a0f5a016aac1ed4a : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2151afd49a6c4075add58339e7443e4f : SUCCESS in 7m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8db6776b7bab47a593a844c97a831ad6 : SUCCESS in 19m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae067e2b3093418f8819719ec8afb9bf : SUCCESS in 36m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/241ba91b5e2d48819626fda5435120f5 : SUCCESS in 16m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c2b77fe6e4454dff88e457d74eff0533 : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5bbc83251070494a90aa5092891b2753 : SUCCESS in 23m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/02f2b38295f04092a36902212ca28c00 : SUCCESS in 8m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0591f921e1be4abdb26529315e9329be : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df9518460b014f1f9e102d0e695ce053 : SUCCESS in 9m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a3e8f712d024455b99e402a383185e41 : SUCCESS in 16m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/752e46729e8941698158e88e494e97c5 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61bd49d5c55d42ceb86097e0c7158d54 : SUCCESS in 50m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abbfc15e5e5143149222955ca54da2a1 : SUCCESS in 1h 05m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/465e96f0625b4939b0fc0c50ab1dfabf : SUCCESS in 1h 22m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/993acbd3c55345d4be355186e9d78c0c : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"438330e189c41cc854974e75adf4762a1d5e4ed1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-28 15:46:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb8f554fc227c50e2c94035ef61e6f34842f2a17","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-28 16:57:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f43ad19ee41d8e59c20af0041cd762a795a9c08a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 16:58:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6270797a9d0ef32ca5b2e877bc6f244028cb87f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:23:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9832c86850446292bb5d8c7f85bd1b : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c083322023104c95b5194f26bb40e0b5 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8030a162dc40cdb133da6ca87caf4a : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aaa0d638c294471d9428877b9b750c2c : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8996f9002869461ca13c0a026d6a3cb8 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cac3028ce62041a28726eaab8a0fb42f : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2c8190e5e5e4fa2a0576bf04741a166 : SUCCESS in 10m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c96d8c577a84430fb221a58f2afac5df : FAILURE in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eca276d659ef4d828112bbbbfcbb6054 : SUCCESS in 6m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/793372e246a44a8abe803cfe4fd77ce5 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b76c89d3b9b747069398c9055dc0d18b : SUCCESS in 11m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c2f9323a1d804d8a999c19c759d60249 : SUCCESS in 23m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/640fbac5409544c4b473c5d25d6d1054 : SUCCESS in 23m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9dc6d82a2f2945f695093c480e8a72f9 : SUCCESS in 17m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/73faa8ebe8ae49cea2b039f1037d0fea : SUCCESS in 17m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a97be10ee44c405c8c1656cee183bc32 : SUCCESS in 18m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/016d7a77af5a4b27b6e3d2c5e8f2fef6 : SUCCESS in 8m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e8a03fc4bc2444cfa2ffe00023d59c88 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e63db05838644b2acdf0a46c0771249 : SUCCESS in 7m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8eb8dc047804ca4acd30057d4bcdb3e : SUCCESS in 15m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3acb541efa3640ec93e30655b208d20c : SUCCESS in 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f15257d136454ce3a40a04237db7a0b2 : SUCCESS in 1h 19m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fea23c06a6cc4ab889088b2d34f36920 : FAILURE in 1h 28m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bc28c0952a0a4398946dfb92ca6fb177 : FAILURE in 1h 10m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ffb2fa2f49574f46810272353f8f9945 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e7d2d29e89f58ab7c0ac4201a2ddb75abf3d8409","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-28 23:08:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"636d9b3882fa71d8b5a88c82f67cc04920ecfd40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 01:30:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e68465f931f4a989356d417b32f6bc7 : SUCCESS in 5m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e23c6521826d4878bc03c85b27a4ac14 : FAILURE in 4m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecca8583333b44f28b432be1ada1f620 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b984d8a132414995f366a0a107f5ac : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f5fd392e04a4a63b47e2a3e8416ad17 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2c70619d54047e5a43dff7c3225f8e5 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c3979c26f5e4f12ade84667ccc1927d : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e70e6216c2694538809f7b3f502a9c08 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4960a5c27be041e1ba11b1227f8d37d4 : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/039cac811ba4432fb3b54c1e2a6db20c : SUCCESS in 11m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c8bfe93fd49f492d80d62b21291e2c7a : SUCCESS in 6m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ada40788982a414e8c746b52a3114e4b : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/becfe1345eac44a982cb4a2b593af5df : SUCCESS in 9m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9413b28b5928463da5b114311984bdda : SUCCESS in 19m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cf25741a9a5f40ecb668ec0825896cba : SUCCESS in 26m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36cf6819959742259d874fc5d83ad32a : SUCCESS in 15m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fbd2cc0b852c4eb1ab3b517d2b330452 : SUCCESS in 14m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02bcda24f2ff4b50a2a453e64775b179 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f9fe66ffe6f04af4a2901e4d65ae9ef9 : SUCCESS in 5m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/594bf9044968468ea35f503aaabb5e2f : SUCCESS in 7m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77c5a8df33c14a03865edf36ada6403c : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4d3fa1f08a7c4ac598f343b8bb3866b6 : SUCCESS in 14m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/760a9cf093e648749cc904c5c2574c81 : SUCCESS in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/101d1b0fcf614874b3517613a14d5d14 : SUCCESS in 59m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29307d3d06df43249961faf0c5bfb899 : SUCCESS in 1h 10m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/95ec68f472ec4973bbc6d8390700f74c : SUCCESS in 1h 53m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a256d71622b646c68060314a0fc570ff : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"697cf6c3216ec9404c6969ee571cc612c0bae144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 01:31:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdccdca8b819ae4ecce7471237cafd37fe7df0aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 04:47:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"647881cb273c706c6fe0baacdb76e13c9053297b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 04:47:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951ca5c1989c4baa97b5917367f4cffa : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf00d221c8544bdaed0f69f227e6490 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e29e9180629f4962a2baaad1807eafd8 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcb43588f2414b8099ab0ac72ff78499 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/471e7be82f9a458b91edcbb5bb8913cd : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8279a7eeaa2409aa9842adbc8c9e633 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b0f5d54d1bd146a79e1e82f70ecb3d45 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7bb9073f4ecb460caab0f8c89ae6317b : SUCCESS in 8m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb58b6908fdc49ad9320827df322abda : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4438b27fd649420c9405c41d96d982c7 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a64328cc99df424694a655e4184376e2 : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da1080fa218e41b6a59d620d515852f0 : SUCCESS in 16m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eef7e601abce4f38b136b8c354981be6 : SUCCESS in 26m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68d1e119befd46f8a5e06a64a7a13afe : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/272e42e5c55045c088391115270f1da2 : SUCCESS in 12m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/653ab687403d49088061d025bc5d3368 : SUCCESS in 20m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a46831a47d7440699dca0bd3991dde55 : SUCCESS in 4m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/84bd0c9838de4be58c588f883cda85e4 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d8d929568994e70bcdd7213de619af6 : SUCCESS in 7m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/83fc90d68c3f4a6c8823d7989d35865f : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/58be52070ede4681ae305b5767d571e9 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/942fd7f62bdd47859d9e249cdfd46d90 : SUCCESS in 1h 03m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be0999da6746441996dbc6dc139ad8b5 : SUCCESS in 1h 06m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1627a3de196c482d8ee19894e30f47b8 : SUCCESS in 1h 28m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eeea98040b7a416c838c827a40ca824b : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"04b85277674e80e1856677b164014a1e174f964f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 04:48:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2235c913520243ce94ad95af0e812182 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b399014bac6cba76738b84181f08c12f3a93c56","revisions":{"0b399014bac6cba76738b84181f08c12f3a93c56":{"kind":"REWORK","_number":1,"created":"2020-01-28 10:08:30.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/25/704525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/704525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/704525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/704525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/704525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/704525/1"}}},"commit":{"parents":[{"commit":"9987b849df85bf40e5698dc52e29bd3f21d5464a","subject":"confluent-kafka 1.3.0 fails to install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9987b849df85bf40e5698dc52e29bd3f21d5464a"}]}],"author":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2020-01-28 10:08:30.000000000","tz":0},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2020-01-28 10:08:30.000000000","tz":0},"subject":"Revert \"confluent-kafka 1.3.0 fails to install\"","message":"Revert \"confluent-kafka 1.3.0 fails to install\"\n\nThe missing wheels have been published on Jan 8, 2020.\n\nhttps://pypi.org/project/confluent-kafka/#files\n\nThis reverts commit 9987b849df85bf40e5698dc52e29bd3f21d5464a.\n\nChange-Id: I5762673966187239ccbdc021215f158faa9efdf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b399014bac6cba76738b84181f08c12f3a93c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b399014bac6cba76738b84181f08c12f3a93c56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
