)]}'
{"id":"openstack%2Frequirements~705380","triplet_id":"openstack%2Frequirements~master~I4c877ad186380c6f1f504bcf1806c6972efaa275","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I4c877ad186380c6f1f504bcf1806c6972efaa275","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-02-03 06:33:57.000000000","updated":"2020-02-11 19:58:29.000000000","submitted":"2020-02-11 19:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705380-1581450793320-7c055bf3","meta_rev_id":"35aab0de5fa3ccf9ff8d7b61bb0eb28a5e76cdda","_number":705380,"virtual_id_number":705380,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-11 19:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-11 16:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-11 14:20:11.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-02-11 16:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-02-11 16:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 14:20:11.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-11 16:02:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-02-11 16:20:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-11 19:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"962134fea1b69950c944811c9f5828c4c4450855","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-03 06:33:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc9503accca3fbb1f5b6b2a1caa8730174bdc238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 08:20:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f31e843cf3b4c378b315e6c08107b48 : SUCCESS in 4m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5d163d107ce74228ae20b9cea7976f81 : FAILURE in 6m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5dbfb739984f838ef595c9ace23e53 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9ddcdecf254406ead13dd62d3a978ea : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e170ab576b74499b640a02113e3a506 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7461bfb4d645d5950d6021756f37e4 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85d5a97489024470a98745648ed1d434 : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1350d5f1e3f44a5c8e52683dca81d95c : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90f196159073483294731512edeab380 : FAILURE in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2eb633e2288a4cc5ab2f6d7e8c88823b : FAILURE in 12m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d192458e98e4d099d5a93eaf5ec2dc2 : FAILURE in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab2c7de0ec1c4107999cefd7cf67794b : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b795165887f4c98bbd08d13a7295498 : SUCCESS in 8m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/89e49054a2a34b1aa2bedff04b4d9704 : SUCCESS in 15m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9e0c1598e2ab4460a19ec86f9b7a68af : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3ba346c9e81b409ba4a460d7f321ecc0 : SUCCESS in 13m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/23d97049f8f543bcae594aa7dd60e41d : SUCCESS in 14m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d97390e2c8f8410bab8d02a3c9103769 : SUCCESS in 21m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df40a450bc304faa9940ca1d050c4dea : SUCCESS in 5m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f781d869f8cf4355852db38c9de9af54 : SUCCESS in 8m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe8dad7104c7410abdd989354d07ebeb : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7b68f6719b7b4444a4216f24dc6e9d07 : SUCCESS in 12m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/59cf79d0c4e845e9894bc393927df2cc : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c45eebdf454340cf989dc094da23b804 : SUCCESS in 1h 05m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb8b9c9cfbd54693993ed1c2a86a12c5 : SUCCESS in 1h 15m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/131e3f544952487cb3153c9becb90c84 : SUCCESS in 1h 40m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26a0e258e9c94a37b8572a12b2da4181 : SUCCESS in 1h 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e9d19dfa9c7cfdc624a50a67d63b126cea474e7b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-04 06:25:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbc262a002e9589fce96619a5b5570346416c1cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 08:36:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40bff51b98314cab8b1caf186b3029af : SUCCESS in 7m 36s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0c52bc7426024d4bbe80add0e77f961b : FAILURE in 8m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9b9963c10484a3eb0a3cda45b225b67 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a372e303a4d4d868cd19a40e5abddda : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c412a9e1c34a478ea0b62225ddc5663c : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fc62d709de3435683d33b7300c74996 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98322104d0094fa0b911009a1528cfd2 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/601699b0d4b840b1b11673a22395d306 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5d208be2974418894faf602f73af142 : FAILURE in 8m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea8d11c4026a4d65b98988c7081b9e7f : FAILURE in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/efb42426efeb48f29d1d4c5d4b5dbb83 : FAILURE in 4m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8908192a06449b0a09459ae028812fa : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b35473770a134020abc93ae57066fe8c : SUCCESS in 5m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e4a543b5c9c4d61ae5bee5a474d3490 : SUCCESS in 14m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a85826fade394b71b39b0ca06df8a788 : SUCCESS in 29m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/959f744a06b64f52be5ca5f7a9cba160 : SUCCESS in 17m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3ae6fd916e0b4fec9e6b8dbd48664707 : SUCCESS in 16m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/762ee560fe344aa8a1cc385d20ac127c : SUCCESS in 18m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a59dc3ca7baf45cbb8a2c2bc43779ccb : SUCCESS in 4m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/53d124673a0e4449be3672a593eac7a7 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e25f785e88c74ef3a0c4dd1877946333 : SUCCESS in 7m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32bf48de63894bc8a8c2441bad63e5e8 : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/15b0ee7226a14400b60d26800fd54b40 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffa2f7cba8d24b8fa05146b6ad2ef28d : SUCCESS in 59m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a491208c9444c2796798f0eac73d348 : SUCCESS in 1h 11m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c90669c822a84a56824bc2862e3fc278 : SUCCESS in 1h 21m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f4669311e3b04967b7b8f320fa43fbca : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5863681ec6a70a4170fc0fc863fd1ea5ab700d2c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 06:18:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"13a0ea097fa771100708cab4ea909cbd8aa01a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 08:14:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3b2d2ee2694b0ba3faa2753fe44b92 : SUCCESS in 5m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/da05ab55ccbd4206a266115b99532475 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/610f58a9d8e141009cbd117aa22a9665 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/464d968a80724efeae7f4a29104d2f9e : SUCCESS in 7m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cea567abf8c4202b040561066b7a408 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ec6571cd887471199f9f67b17beae9a : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a72e8987fe9406ca1d908cbe1feab1a : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56c4268c95a243f4a4d28de268df46ee : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de6f388827544beca52f77493c103819 : FAILURE in 10m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/83f889b93dec40059030d0a7afc5aa35 : FAILURE in 24m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e7a6422a7ade462d9d023d711bfb3400 : FAILURE in 18m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5335c8830c3349bbbb3376dacda5a5d6 : SUCCESS in 17m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74dd403fc2ff41c5b6d5f6e187399cc7 : SUCCESS in 24m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa781fde1244473fa923a350e0e2c670 : SUCCESS in 24m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/87a74b782db348a19df4c3101447476c : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8719eb150a86418297359345f13fae9e : FAILURE in 13m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a6e4df7d0e2b44bc8a9b4cf3f8d638f6 : FAILURE in 13m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be194143dd1445b880b02707e97a7aa6 : SUCCESS in 15m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ee3f9efafdc4c81b44fe11a4b19c8d8 : SUCCESS in 6m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/826d53a74d944ab88a86da299f15e277 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/979b4708d3a1433284ddbf3e6cdcc0c9 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a775f28a857e40179b77255ec4ad2489 : SUCCESS in 10m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b2ea4d012a0c4fe7954195164fc8221c : SUCCESS in 20m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f75478c23c8c41acb47eeb1b6ff63a0b : SUCCESS in 1h 13m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a008e5c700c047d0988c98ef8eef6257 : SUCCESS in 1h 19m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/62ea6c5c80a24f0f94b269be995f2f3c : SUCCESS in 1h 30m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4dae08c5965b44969567fce0b72fa135 : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ff7ed6c1e68eea0925ed9ce35b74ff2a3fbcecf2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-06 06:32:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d5ddabfb238c70237336aa216632dc8b5ad0d9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 08:35:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e438643f91f24791a36f3f5956533bdf : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef2236d7439482a9eb90153190da71e : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fef1e0410524be5b5dce19d5cd8ed59 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32a88f4a51904739836823ca69b8c7d7 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/694019320ea145c486747c0f92287d0a : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab919defadf3479897e98bf8125c7c7b : SUCCESS in 7m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f887d222aa14c1b9c69d3e7a8c8f832 : SUCCESS in 10m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ebfbf464fee4ed6b646307e9cf97eaa : FAILURE in 35m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ab6401939eb4bcc9f0f74a21932ff01 : SUCCESS in 20m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/136e04d7ffc841c38a78c971bb41f656 : FAILURE in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/121630668ed048fe888b26b768c7bdd4 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b24b27a98ed342ba8025a1719b0ad4ff : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3888ca7bac5f42799ea610516cf24e4b : SUCCESS in 13m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/58f96384e84e4f21acb371aa2968cac6 : FAILURE in 25m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/33fbb0a185d243069d4ebdf477249255 : FAILURE in 15m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fd8fcb901c0e470e9af216540e84cbc1 : FAILURE in 15m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a28015bcca6d4d39858d4cf09989b610 : SUCCESS in 21m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1dba7aa5e59d41bba819dbecea541074 : SUCCESS in 13m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/78a6cf9478c44ef5a74f391194ae832f : FAILURE in 16m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5053578528041e7b064e865195ef5d6 : SUCCESS in 15m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/018905814b13452fb8966361f41dd1ae : SUCCESS in 11m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7991585bad424500b945fe33af86ef40 : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/241477d9a198476f94edf2b1c50a09fa : SUCCESS in 1h 00m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26b4c43a4efa4589a2df665a2f1393cb : SUCCESS in 1h 08m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f84485a511394d4d8f5aadf5f3d8779e : SUCCESS in 1h 37m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e63498bcf8334290b715f28642b648c8 : FAILURE in 55m 50s","accounts_in_message":[],"_revision_number":4},{"id":"c4e5ae20742b0ab5e9556db83a46ad496a461f44","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-07 06:13:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c20dcedc7b388f273f2e0d4ba8152a32a36f0b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 08:11:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904a9ba704fa4acd9739180e860e770c : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e83fc81d66417cb1ba6a841b77fd79 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e524f992a93c4ec1a187721fdfeba861 : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72a76a171fda4b99928ad5acaceb668c : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9fd1febb5714423aea3b4cf7e6bfd17 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa007ff32f9b47e583e8cf4f17b6a296 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be27398aa49c42c081732689fdc95464 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d5be87994a84addbc97be4840381bc5 : FAILURE in 11m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e70b5000ea14cca8737a0959ba4817c : FAILURE in 15m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f8f67ec85b19406e94d8a1397ccf35cf : FAILURE in 7m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/77e58a9f13644f8cbdb6acb301327a4e : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddefad92043340afb908be253e060ef6 : SUCCESS in 10m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/11c733dc652b49b6b3fa621db210ff5c : FAILURE in 24m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71fb2e58f22a4b6a83f30fdec11905db : FAILURE in 26m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8eddf1d6733f4aa189ffa7b906f8c33c : FAILURE in 19m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a07013c67d474738a225019cb08de7e5 : FAILURE in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15d8b0bacdbc40d7b4d344e819b5e10c : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/460148e6ea07486b9ddf0db21c8cf12d : SUCCESS in 6m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a30ebd6455774d63b7f97a9b4a56f778 : SUCCESS in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5140816d91945ae8e3a045313f8f912 : SUCCESS in 10m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/afea3340f0af43a68fc3e9af35a4878f : SUCCESS in 14m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5226afb8486a4f0eb3f06653c28111c7 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a72a4f8ce8441f79014ac0b857eee8d : SUCCESS in 54m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d995a2e3432d456a8282d135fd374c24 : SUCCESS in 1h 19m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1030f4cf89e54c2fab20aeb77f514cea : SUCCESS in 1h 48m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/464e63f161af46c0a7f76a028f3628ba : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":5},{"id":"4210f2a32b784fc25839cfe76abbb965c0243a68","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-08 06:19:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ba345d18bf203e867622c7acf5f16f501ca0ee8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 08:03:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c60580484614bb9b119b8545c931fa1 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1539877fae5b4dc79c48997f92ea1883 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4d87f37192847d5b8e3dd700839323b : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b693f2733ff0402cb6336f367d2b25a2 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/944d9f3c33aa4e7f8d836ab54a0119d8 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7df45523d56d423889153635dcf5c9d5 : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d00d924cc73c44ab95cd289b705ce4a7 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82a85f2f1cde4dcba279434741e29bae : FAILURE in 11m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4da1c3f2f9c4bad88d05ad57b269fa1 : SUCCESS in 12m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6bd76c42ea17463186a87d37fec0f46b : FAILURE in 6m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a07e6771a72941a8981f916072cb2a2b : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21c0e2d004aa465e9b782d2c25e19d3a : SUCCESS in 9m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/267d119ce28b413c9a6790bbafe0195c : FAILURE in 17m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e3b20e99ab31449ab9fddcda46bf143d : FAILURE in 29m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb476cffc4304e4f89a5a14ebcb9c490 : FAILURE in 14m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dca78c762be54ec49ec606852cba4e9e : FAILURE in 16m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d8fa065f00e4baab79aba361d7a18b4 : SUCCESS in 22m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a74c86fea3504aa2b0e8098531eaeac6 : SUCCESS in 7m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d580b2abeba4440ab9f52a4fc0df76d2 : FAILURE in 8m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0191bc8b58004bc18b77c6aea9263bde : SUCCESS in 8m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f0743e89c44342cc913f4dcfc82ae025 : SUCCESS in 11m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6391c29a2b8048f1be165a90471c359b : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6987c0bac8cf46a3843e4a5af159148b : SUCCESS in 58m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de90685a949242d782d1bd37c18c92cd : SUCCESS in 1h 19m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d0320bec2ac4c91bc22a034ee3c785d : SUCCESS in 1h 35m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3c264fb00779429a8f19169e73f68552 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":6},{"id":"f7c40d3a64d10728ebbc2ee9ea14590a1c6f8748","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-10 06:31:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d0a7f4ce41b817a52741a79d3c65282887d6a2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 08:30:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc18a14982214fecade7d461b4b233dd : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b3dfb9dd834c148855a166a233ff63 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfd6024c4936420db126e7705fd73c4b : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b72fe554b25a44128555350484596166 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/714ba8f58d52499f86f291ec381b267c : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afdf4f935db2444c8bb8324108088783 : SUCCESS in 7m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/58277ddd82c9479b9300709720bdf872 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aec2481c56cb40ffb81f926bb758418c : FAILURE in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/affb2fd7944b47c2831919ecd5562527 : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/450f029d0a4742db8198c208920eb8d6 : FAILURE in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a0d2bd670f7243c098c8b466abae6330 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f448b8df62264983bbc9cc0c039e60a6 : SUCCESS in 8m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4adee216b59f42c282e4d31488aa24e8 : FAILURE in 18m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b102a3361fc472dafe5cd8f5e6b18bb : FAILURE in 24m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9366feb7410748e29b67b4bba886c736 : FAILURE in 17m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c898895670e24251a319bfcb5957ca14 : FAILURE in 14m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ee2b5234b6a49be9408463088ab35ec : SUCCESS in 15m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d84246ec30d748edb36f500c49175694 : SUCCESS in 7m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4390107f21a14658a32983a1f1fe8c07 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/170e00dc747049629880d829e1855f8c : SUCCESS in 8m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/09c4172b38ca4b35a8c52af5ec9dda54 : SUCCESS in 12m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2cf24b9c806648aab394df375571b991 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e0910f047d44088a8cbe9704b97c070 : SUCCESS in 50m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/adfb115a4a0747349b1be5131b257a68 : SUCCESS in 1h 02m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0575ff1ea1604fb4ac44f2fe530a3721 : SUCCESS in 1h 30m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a1b5743df3c545c794d32ff76ee8d075 : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":7},{"id":"c01baf626ecf283de15433bcb9b250b14732df8c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-10 16:38:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"26fd5d270f0f5e9fed7e19427a5fe2581ad650c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-10 16:38:27.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":8},{"id":"bc62b624bb4d05d4a866d4a577ee989d12477d97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 23:57:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489b2b7e9f9e46cf912abadb05abbd54 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a04b52e400ae41fe97b55a70b3a1e904 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da0ab760bbca48c2a796e4cff1b5a952 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4392899c5e29459abc045252d34a2799 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31bf000549f54478ab72b9752b85de88 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/183d1d07ef8249278fef91262aa8af66 : SUCCESS in 7m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbd955ee94b6479a89c735047ebeea15 : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d27468aa82924c80b50d60b177a64eaa : SUCCESS in 11m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fbfc7313cc694e3c8fc6a5a751bcaeee : SUCCESS in 14m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/699c1298c9a548c7a260eae6222fdb84 : SUCCESS in 9m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66351e552fea41269e71ee19cdced96f : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8258b604ad324c16ae0747d8b2f955b3 : SUCCESS in 9m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be121a3a8ba947ef8d938a375eaed520 : FAILURE in 21m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/50d92696493f4efb99282caba08f9173 : FAILURE in 25m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1dced2ed55904c70abe607dd488bac69 : FAILURE in 17m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a7811eeb7ef425581968804a7b355f4 : FAILURE in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99cec8004c3b4505871a6aab903bef87 : SUCCESS in 18m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bf520bb1c5494a4cb9d0e98711b57d7f : FAILURE in 7m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/81009b7d161e404cab48d7cbae35aae3 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f11622a6b658458fac72e0f993a93b0b : SUCCESS in 10m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/387f638009a34427af5df2886edf82c4 : SUCCESS in 12m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c681ddfb1f949d48522d5a6df2f5c1d : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f54de1c0725480ab53871d4f86f3839 : SUCCESS in 56m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff6b514e98ba43c1bb23053507f41619 : SUCCESS in 59m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/03717135354d4f57952d6b4fb63282c0 : SUCCESS in 1h 24m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0dc11552c1b7417e8557e5637ab81bbf : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":8},{"id":"63e7b1e370eb78066982c0e2ad90a7072a326e68","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 02:37:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"af21d4d0057b194a3c56a9077dcd937908d3537c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 04:18:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e34bcfc868a4395b311a432c41763d9 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199acb445757447eadfc61d8416bd2e7 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/487bdee44701469daf628de62afd38fa : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85be74f094394d25890be2e5b9ea6a09 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca814a75a8bb46d084b6a2e6008bc861 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38f1d5dbabf842bc8eb2e3c0e57caa4c : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da56ccfab9ae409cbc1cdac039ddbb41 : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9695bc77cc3401cabc3351029415dd3 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7181b508b28a419592f2f17d9cee3595 : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5c6461b846248408f4131b7065c4aa5 : SUCCESS in 6m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4ec0c939d6c4656af2a6ead120c415c : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf322adf2f3f485ba458a2b439e5582d : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8290a286fd243be93751a3ae512b362 : FAILURE in 18m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/29e33e9f3a444d9fbc68483e13013f9b : FAILURE in 26m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b5ccd2c9e3ab40ec93e50c6f91463ad4 : FAILURE in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/97e4d8c5e3d04d77b88c33533cf9d87b : FAILURE in 13m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1379829ca95f4c1892e2e5c35136afca : SUCCESS in 15m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/07bb9d1320cc4ed9b836a8d4f29c1a12 : SUCCESS in 5m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/def93d16303544dcb01c2eb2f5ff4aa7 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/309a017690fc4f88a6b7a332ee829887 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a62d1c0b692342809f60ec58d3464a4b : SUCCESS in 11m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ce5b41554d9745c58973b8965b83f85a : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d309bc2eed54d368983fc73b95c5c4a : SUCCESS in 57m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ca9307b3e2940c1b809bdefab3b44cc : SUCCESS in 1h 03m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eedff7c1bf5745bcb2529dcd2d349c01 : SUCCESS in 1h 27m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1ffae7dea76a414ca3ebdfcc8cd68c7d : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":9},{"id":"d87636a58e19393eb683f63011eefc4dd2664750","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-11 11:34:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e1697332168fd65f566c1a287ac6aff94fc09b92","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-11 12:42:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1a366e3bace568306f797818fc1b077bd7115788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 14:12:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b35e36d28b4a98ac5b1c8e26fe85a3 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5393793c75f4789afae2310ef29e665 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255b8e0455c34839b0ec2b6f3bda6773 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bcf5eca7fd04d0e8e429b7961b9a7ce : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/479496bd407743fc85626801a151b3f2 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12acec314b234ef5af117e5c5921a88c : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8fbc2df132614aeea37e5bbff2979eb9 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44b308c1125e4c129ff21e3b43b3b8fe : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4e52435a4ef489b894d6465c86a25f0 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1c82472b4f5b4f5bbc35a1e07a4cd9a4 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a2339727af54edeaba89fdaf7d9016c : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/382c078a7cbd44bfabf21acb54459149 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19358d9c4b014aad8cad1b252603b171 : SUCCESS in 13m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/35f7801c23264546a94a7ddb3ad51602 : SUCCESS in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f58dbcdc283d475a86bddd3fc141bd72 : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4891dd1cee6d4565ac84d75e53a3bccc : SUCCESS in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/736eeaaf9ac246ca8f734aff5e3619ee : SUCCESS in 17m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2c1a8af996e94ff6ac20146ffc45b07e : SUCCESS in 4m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ca67b25ffd814c2687bc12f55deecc97 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bf7df98237742168120d141de12fc22 : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/455df7e9d64545109280c602b5759267 : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e99b54e861c542e1aef7cb8445474e8b : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18d10298018b430c8acfbc392b6a5bc2 : SUCCESS in 59m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9fee8940e994a44af6f5924039c543c : SUCCESS in 1h 05m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1afc0f4147484800afea364359a452d6 : SUCCESS in 1h 25m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b95f7d731b948a999b9627aaf4e22a5 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":11},{"id":"2895270fd2fa0ad887baf01c0a6ddc7b78917e88","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-11 14:20:11.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"77ad96edd8de4c66f0b4d0ac0f03cce263e60c53","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-11 16:02:42.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nFast-approving in order to unblock projects stuck on the mox3 issue.","accounts_in_message":[],"_revision_number":11},{"id":"3eccd8f0ba0fa57423f7491e1ed5bbdf7f9b5788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 16:02:54.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"f320554fb4a711d890e8398c7fd47c55daa51720","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 16:20:10.000000000","message":"Patch Set 11: Code-Review+2\n\nthanks for working on this overnight","accounts_in_message":[],"_revision_number":11},{"id":"0145c3441cf6ef40f37e7e8eb36940aea5558bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 17:52:11.000000000","message":"Patch Set 11:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d2f0536b81b573a5c1079a970a5b5519f554229f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 19:53:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"8f8617dddabc320c49339ee97e0419072145c545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 19:53:13.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae81b46fa724a1385bbf7e974819bfc : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6710b15d7f4e4f9d9f9c24d38dc5a0 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0baaded23d264e88a342a0cdbc581d49 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26c5cae13f42440e9e02d5e7ba2c323d : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d10bae43ed44b13a08ac74f7931ee64 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77b24b58a6d14c859047ffe15b975718 : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e147fdbe1e54749827df52477aa6c08 : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a6d86ddbde443b785a8780a8e9f9993 : SUCCESS in 10m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/73a28a39392a46dd83705bbdea18b662 : SUCCESS in 6m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c38e2fc66ec14144a08d4270d1a51d4c : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a70396cd2b8494581eb25bd7fea7ebd : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/848f7e899859498a99e5d5f953d9e010 : SUCCESS in 14m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/33191564dc5c47548dd9c1fd0c734694 : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d0dd3d447cd405293a142aff9e74044 : SUCCESS in 13m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/002a762f21754798bcbc9049b2e1ddff : SUCCESS in 14m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbdd136bfcbe4e8f9d5132c1004189ba : SUCCESS in 19m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5aa2984fb1243f08325849e7361122d : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6e4affd2211d45fb8b27cfd0819ee75d : SUCCESS in 7m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/832bda68eb954d2783d31aedf95a4395 : SUCCESS in 8m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e0bbc55ab8384a29a57353ececd24ecd : SUCCESS in 12m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2faf8ea698c94e12a661c5592784b8e6 : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e296031fa0843799b5345dbb3d3aa3b : SUCCESS in 1h 02m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/24c37c36fa6d4cb7956e7f4218059961 : SUCCESS in 1h 10m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/60fdbe6c8c8d4cb397d80bf6e304855a : SUCCESS in 1h 53m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6ca37cd4968479994a1c9c75d664237 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":11},{"id":"35aab0de5fa3ccf9ff8d7b61bb0eb28a5e76cdda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 19:58:29.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f27858d3894db696767e7f4f2b0f2e : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"46917162cddfb606983bfda88d983f066cae6871","revisions":{"b94f84ffb9cfdf5c60c0c63c58c239aabc7e5d03":{"kind":"REWORK","_number":1,"created":"2020-02-03 06:33:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/1"}}},"commit":{"parents":[{"commit":"ba1a1cbb36fb962e48299a0d2a72641751207f79","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba1a1cbb36fb962e48299a0d2a72641751207f79"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-03 06:33:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-03 06:33:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b94f84ffb9cfdf5c60c0c63c58c239aabc7e5d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b94f84ffb9cfdf5c60c0c63c58c239aabc7e5d03"}]},"branch":"refs/heads/master"},"5d981a8840ace65bd3ae0d176f4347191b3c6064":{"kind":"REWORK","_number":2,"created":"2020-02-04 06:25:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/2"}}},"commit":{"parents":[{"commit":"ba1a1cbb36fb962e48299a0d2a72641751207f79","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba1a1cbb36fb962e48299a0d2a72641751207f79"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-04 06:25:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-04 06:25:18.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d981a8840ace65bd3ae0d176f4347191b3c6064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d981a8840ace65bd3ae0d176f4347191b3c6064"}]},"branch":"refs/heads/master"},"f1b150a6e205f557afe9bee1988d925fcabf42ea":{"kind":"REWORK","_number":3,"created":"2020-02-05 06:18:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/3"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 06:18:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 06:18:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1b150a6e205f557afe9bee1988d925fcabf42ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1b150a6e205f557afe9bee1988d925fcabf42ea"}]},"branch":"refs/heads/master"},"f326c4d18a92e6eba40c59b54f929a06d5b146d7":{"kind":"REWORK","_number":4,"created":"2020-02-06 06:32:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/4"}}},"commit":{"parents":[{"commit":"8a73e436b723fe5a80bb12eba2a7b763435b36e4","subject":"Merge \"update constraint for oslo.db to new release 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a73e436b723fe5a80bb12eba2a7b763435b36e4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-06 06:32:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-06 06:32:50.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f326c4d18a92e6eba40c59b54f929a06d5b146d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f326c4d18a92e6eba40c59b54f929a06d5b146d7"}]},"branch":"refs/heads/master"},"a71ef6f33ae401c2e1508c47bce0ec98472a4fae":{"kind":"REWORK","_number":5,"created":"2020-02-07 06:13:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/5"}}},"commit":{"parents":[{"commit":"6851e4620aafddc5e2917d48be039db92aa79ba9","subject":"Merge \"update constraint for reno to new release 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6851e4620aafddc5e2917d48be039db92aa79ba9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 06:13:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 06:13:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a71ef6f33ae401c2e1508c47bce0ec98472a4fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a71ef6f33ae401c2e1508c47bce0ec98472a4fae"}]},"branch":"refs/heads/master"},"f3532d072bbf6853b5b367f01e1121f403786c9a":{"kind":"REWORK","_number":6,"created":"2020-02-08 06:19:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/6"}}},"commit":{"parents":[{"commit":"f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8","subject":"Merge \"update constraint for oslo.log to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-08 06:19:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-08 06:19:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3532d072bbf6853b5b367f01e1121f403786c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3532d072bbf6853b5b367f01e1121f403786c9a"}]},"branch":"refs/heads/master"},"23f524ffca8f095f9244701c2bca404a4bfaff01":{"kind":"REWORK","_number":7,"created":"2020-02-10 06:31:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/705380/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/7"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 06:31:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 06:31:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23f524ffca8f095f9244701c2bca404a4bfaff01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23f524ffca8f095f9244701c2bca404a4bfaff01"}]},"branch":"refs/heads/master"},"7fac0e3ede24a3de551806df4d1eb256916ccbeb":{"kind":"REWORK","_number":8,"created":"2020-02-10 16:38:18.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/705380/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/8"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 06:31:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-10 16:38:08.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7fac0e3ede24a3de551806df4d1eb256916ccbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7fac0e3ede24a3de551806df4d1eb256916ccbeb"}]},"branch":"refs/heads/master"},"37219350b7c999f87d6ac532c56db002a70d16e2":{"kind":"REWORK","_number":9,"created":"2020-02-11 02:37:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/705380/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/9"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 06:31:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-11 02:36:42.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFlask-resful held back for (keystone at least)\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37219350b7c999f87d6ac532c56db002a70d16e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37219350b7c999f87d6ac532c56db002a70d16e2"}]},"branch":"refs/heads/master"},"b4bce2c0fbd033f67562a8f1d2ecfd748bdcf821":{"kind":"REWORK","_number":10,"created":"2020-02-11 11:34:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/80/705380/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/10"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 06:31:35.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-02-11 11:33:40.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFlask-resful held back for (keystone at least)\nmock held back for Neutron\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4bce2c0fbd033f67562a8f1d2ecfd748bdcf821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4bce2c0fbd033f67562a8f1d2ecfd748bdcf821"}]},"branch":"refs/heads/master"},"46917162cddfb606983bfda88d983f066cae6871":{"kind":"REWORK","_number":11,"created":"2020-02-11 12:42:26.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/80/705380/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/705380/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/705380/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/705380/11"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 06:31:35.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-02-11 12:41:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nWerkzeug held back for (keystone at least)\nmock held back for Neutron\nalembic held back for Nova\n\nChange-Id: I4c877ad186380c6f1f504bcf1806c6972efaa275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46917162cddfb606983bfda88d983f066cae6871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46917162cddfb606983bfda88d983f066cae6871"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
