)]}'
{"id":"openstack%2Frequirements~705685","triplet_id":"openstack%2Frequirements~master~I3f2ead92cf8ff27b7c65aca952ec49eda3070173","project":"openstack/requirements","branch":"master","topic":"fixoslo","hashtags":[],"change_id":"I3f2ead92cf8ff27b7c65aca952ec49eda3070173","subject":"Keep oslo.concurrency to 3.31.0 for py2.7","status":"ABANDONED","created":"2020-02-04 13:59:33.000000000","updated":"2020-02-06 09:08:29.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"682ec4a430cdeffdd2a25b37e1108dfe29755657","_number":705685,"virtual_id_number":705685,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"date":"2020-02-05 06:30:46.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":29755,"name":"wesgerrit hayutin","email":"wesgerrithayutin@gmail.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-04 15:49:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-2,"date":"2020-02-04 16:46:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2020-02-04 14:08:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29755,"name":"wesgerrit hayutin","email":"wesgerrithayutin@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-02-04 14:14:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-02-04 14:06:42.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"date":"2020-02-04 14:04:06.000000000","_account_id":29755,"name":"wesgerrit hayutin","email":"wesgerrithayutin@gmail.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29755,"name":"wesgerrit hayutin","email":"wesgerrithayutin@gmail.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 14:04:06.000000000","updated_by":{"_account_id":29755,"name":"wesgerrit hayutin","email":"wesgerrithayutin@gmail.com"},"reviewer":{"_account_id":29755,"name":"wesgerrit hayutin","email":"wesgerrithayutin@gmail.com"},"state":"REVIEWER"},{"updated":"2020-02-04 14:06:42.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-02-04 14:08:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-02-04 14:14:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-04 15:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-04 16:46:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3794493b739ab45638324e5908207db6a1d87d6f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-02-04 13:59:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e89cde06d74116e65c082afa06ebc5fe07e65215","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 14:08:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nMany oslo libraries will remove py27 support - like the rest of OpenStack.\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012330.html\n\nAnd see other repos like nova that have stopped py27 support already.\n\nI\u0027m not sure whether this is the right approach at all but leave that to requirements team to decide.\n\n-1 for inline comment, this will not work as is.","accounts_in_message":[],"_revision_number":1},{"id":"0a703a24296a7984d47ee99d08ca1e51df765260","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 14:14:03.000000000","message":"Patch Set 1: Code-Review-1\n\nyeah, I will not suggest to cap the requirement which can end up doing for many lib. its time to drop the py2 aggressivly. Same I mentioned in ML- http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012356.html","accounts_in_message":[],"_revision_number":1},{"id":"44f833721c12366f53db6e70ed7cf5a7e88ca55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:49:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2040d711289c43a1933d299958f8cbb6 : SUCCESS in 4m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0a040f97248342a5baa2f5f8653ce8af : FAILURE in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642397fb542447c5b574a4eb456cff4b : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b50734a038438f82ede33abd87d7ed : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfee6e6a69334ec791916e128810d99b : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3ebb609dd7a48c5b486d02661391eab : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9449b7c43a6a4744a3a650f3d0abac15 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81336d7ba7424459a8356b16153d2eb8 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77415ec904ac4cce8cbf5a1cf6122760 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb2c8ad6ee414ad8810ab15250ad73ac : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2779063697f40cca519a1237784a3b8 : SUCCESS in 7m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5ffc0d3c802642dcba872d1d6cdf9a9f : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bd4fcb4048d4771ae57cc67ddf99566 : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e119dbfd6d164d42aa19a71040d71b48 : SUCCESS in 17m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb08fac6513e4e5c805186b319fbdca0 : SUCCESS in 30m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cba71f1a2bb64bf8bd7301f6558963c0 : SUCCESS in 13m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e781fffdb79d491e9c063af754f6112c : SUCCESS in 15m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f20d811c90d4499dbc1be34459c5d7e7 : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/13a17dc627ab497890a79b90c338d237 : SUCCESS in 5m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2425f5cf3ded4bce91dde17fc368e1c2 : SUCCESS in 8m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9612c7ad801f4975be5cd3a796872e73 : SUCCESS in 7m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6bbc5901321d47f7b0e6dcc7fb622bf9 : SUCCESS in 14m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9aa1d84b5c8f4399bf01de6947164763 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bc983d455fb49269c460fb069bdc900 : SUCCESS in 1h 12m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/467580115da34700bf91ddbecc89406c : FAILURE in 1h 02m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/671d86a216084d0b8688a2d1e2d5e911 : SUCCESS in 1h 44m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/18e57265e7cc404e84bd047ede3ba309 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"313d0da96dfaa54b61ab214a2beb4b605b6aba24","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-02-04 16:13:27.000000000","message":"Patch Set 1:\n\n\u003e yeah, I will not suggest to cap the requirement which can end up\n \u003e doing for many lib. its time to drop the py2 aggressivly. Same I\n \u003e mentioned in ML- http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012356.html\n\nThanks @gmann @andreas for the comments, I am not sure how to move forward but by looking at uc we have some many dependencies are capped with py27 conditions","accounts_in_message":[],"_revision_number":1},{"id":"eb67d8ed4c09350b2abb3a6a91b68262950a7ca1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 16:19:08.000000000","message":"Patch Set 1:\n\n\u003e \u003e yeah, I will not suggest to cap the requirement which can end up\n \u003e \u003e doing for many lib. its time to drop the py2 aggressivly. Same I\n \u003e \u003e mentioned in ML- http://lists.openstack.org/pipermail/openstack-discuss/2020-February/012356.html\n \u003e \n \u003e Thanks @gmann @andreas for the comments, I am not sure how to move\n \u003e forward but by looking at uc we have some many dependencies are\n \u003e capped with py27 conditions\n\nYeah we did for few dependencies but now oslo and QA tooling droppping py2 so many/almost all projects still running py2 jobs will start failing. \n\nNext week is the deadline to complete the py2 drop work so projects failing on py2 jobs has to drop the py2 on priority. Tempest gate is also blocked due to that and we are on fixing it by dropping py2 jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf6a0a6273d16f2646a74634809353e99b9f5df8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-04 16:46:52.000000000","message":"Patch Set 1: Code-Review-2\n\nThis would expand testing needed by projects still supporting python2.  As the py2 and py3 versions diverge over time I don\u0027t see this being maintainable, even over the next month or two.\n\nIf this were to be allowed it\u0027d require a change in global-requirements to keep the cap on py27","accounts_in_message":[],"_revision_number":1},{"id":"22afe559285a25438ce151305e77bf73c27a7108","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-02-05 06:30:46.000000000","message":"Patch Set 1:\n\nThanks everyone for the comments, addressed here https://review.opendev.org/#/q/topic:1861803+(status:open+OR+status:merged) abandoning it","accounts_in_message":[],"_revision_number":1},{"id":"176e832fd1f3f05caa750dd697d60affe5d39a5d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-02-05 06:30:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8d6b22e30e759bce5720ed9e71f3079fa87b39be","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-06 09:08:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fa8131a826369956b8bf2007cee34901733f1fa","revisions":{"4fa8131a826369956b8bf2007cee34901733f1fa":{"kind":"REWORK","_number":1,"created":"2020-02-04 13:59:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/85/705685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/705685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/705685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/705685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/705685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/705685/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-02-04 13:57:17.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-02-04 13:59:28.000000000","tz":330},"subject":"Keep oslo.concurrency to 3.31.0 for py2.7","message":"Keep oslo.concurrency to 3.31.0 for py2.7\n\nTripleO jobs still runs on CentOS7 but new oslo.concurrency\n4.0.0 breaks the py27 jobs. Pinning to a working version for\nthe same fixes the issue.\n\nCloses-Bug: #1861803\n\nChange-Id: I3f2ead92cf8ff27b7c65aca952ec49eda3070173\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fa8131a826369956b8bf2007cee34901733f1fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fa8131a826369956b8bf2007cee34901733f1fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
