)]}'
{"id":"openstack%2Frequirements~705770","triplet_id":"openstack%2Frequirements~master~Id818eac1a3e1783c1d3622a705dfd01e50e230d8","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Id818eac1a3e1783c1d3622a705dfd01e50e230d8","subject":"Update SQLAlchemy to 1.3.13","status":"MERGED","created":"2020-02-04 17:05:01.000000000","updated":"2020-02-05 23:15:20.000000000","submitted":"2020-02-05 23:13:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705770-1580944386869-e685ee51","meta_rev_id":"be310d5a484aea19f80f4bb2566a0f8ceba49fc8","_number":705770,"virtual_id_number":705770,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-02-04 22:08:06.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 23:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-05 20:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-05 20:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 22:08:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-05 23:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62e06fab9d3ef35af0d6c8fa311171fa023d925e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-04 17:05:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f181cb7f416fd821976ee654d4c691728ca6fc8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 19:46:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b2e7a777fb4df3a388936fab1ccbb5 : SUCCESS in 4m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6c6940e84d5441b5b3c58e9e286c6549 : FAILURE in 7m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6177279abd4872b40bd4668fa263a1 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddcf7a7ee5eb476c8f881b0a2aeec923 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26f4606fdb074413af30237824943bb0 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b82064703094809a1cab0ed20167400 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7720edf0461b47de8d7a07e310f3db9b : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e08ffbc7532e4efdb21a294a66793619 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37a831a31e284156abfcaa0347c04331 : SUCCESS in 7m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/34a556d76b3c496db6d0723dce5b535d : FAILURE in 8m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e75701cd98074e04b5607c6e78082f60 : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2eebcaf0ad064c03b6ac3c62f8e82595 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6db8fb2670f492d800bb9c73e7148b9 : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/86ddebcbd9f843cc8a27b61c1c703a14 : SUCCESS in 15m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b27740a61d5b489cb4bccd24c18f425a : SUCCESS in 29m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0eb76ded4d24a0a9217610da2dea3a9 : SUCCESS in 13m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3903b389d0474beaa11fc5b2edb18da5 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8e6a9c492824a3e9a1bf63e85b683e3 : POST_FAILURE in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/221eb94a6b344ec680355e6637c6b566 : SUCCESS in 6m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b97419246b9f4b1e98d6a6c74ec37347 : SUCCESS in 7m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72edf8fdd06042d4a3a80ae5eda1f515 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/01e4935762c745648a94baf88892b571 : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/694fd9037e924b779b63dd78ca3a5603 : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e7936583c5b434da67bb9761f40ada6 : SUCCESS in 1h 01m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ea4a01bb54c4c8d8a86e84bbc367c8a : SUCCESS in 1h 09m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b9cf0bce887844a8a88adee9dc84f502 : SUCCESS in 1h 28m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8fa022a65353402e85596fcc22d81e5b : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c3fa8d1019be5bebfa6554927b1bec5a4cef4f62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-04 20:00:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6cdde1cc8c5ee63c3a57eee2b61d91af0a96341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 22:02:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc3edd9e51a403ca5650a9ece103e0c : SUCCESS in 4m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f13d8f36912e423596542359647792dc : FAILURE in 8m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce30470353d945e18ba64897e0e43bf2 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/382845612f684db99f1b0c490bea4aa7 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91e724005235492cb3f64f54fb16f228 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2446a8f272ad43408b4bed915206b9c7 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f196caef94c045f9be8348c6b4532b12 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db863ced9d1b431699ffaf9124e23dcc : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/07a9bc2bdf3d4af88246c47c9ab3f8cb : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4449ce714fdf417299cc4d53b0a38009 : FAILURE in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/595ab285041f47089169130f7cae7d58 : SUCCESS in 7m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aaa1c0ac37514456913211a13c112f3b : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3afd4408feab4469a02b65a36cdcf5d9 : SUCCESS in 11m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/81afdda93f304e2e8292e4bb55b31e73 : SUCCESS in 24m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b0b7c3b161604f83ac4fb81d936a78f2 : SUCCESS in 24m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bfc8fb490894d48a8e232c0c8d9571d : SUCCESS in 12m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4cad9977eabb4bab939ff81e52796204 : SUCCESS in 15m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4954697d8514a709be21dd9a45adeb7 : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/361d707f8b574cf9a57d4e5765f81e1f : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/75abd904b22b4657bc4977ad4eac891b : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76f9808f64344d14890aa64f71df7ba6 : SUCCESS in 7m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/834cfdac6bbc4a04ab81f2560b1d439e : SUCCESS in 13m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/beaa6079aa7e4e8b934acc8d49e59d75 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a05036a42b204c088a34a59642da847f : SUCCESS in 55m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b34b2635194048abbd3bb1159c833983 : SUCCESS in 1h 22m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/99f13d79e83f4dc0bf9ce520f737009f : SUCCESS in 1h 48m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c99bf85c23ec405aa27396eb6a272076 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bdaa088ed5dc6ac132eb3c014cdb12044b95ba94","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-02-04 22:08:06.000000000","message":"Patch Set 1:\n\nLooks like it\u0027s not picking up the dependency.  cross-cinder-py36/job-output.txt says:\nHEAD is now at 819b4a0fc Merge \"Support multiple stores of Glance\"\nAlready on \u0027master\u0027\nopendev.org/openstack/cinder checked out to:\n819b4a0fc020730c3914d440f8b83ec0a30b73ef Merge \"Support multiple stores of Glance\"\n(that\u0027s the parent commit of https://review.opendev.org/c/704425/)","accounts_in_message":[],"_revision_number":1},{"id":"deb2daa2de410022080c1499f3eb7fb0ec823ff2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-05 17:24:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af0752eddcbbc9a15681d1a4ef8a35191914747b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8f8e41c54941f5b25dcff9a73012d0 : SUCCESS in 4m 35s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fbf4c33625104c849ad4d6a20e7421be : FAILURE in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba044c6a56184aae95471f4acd6879d4 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e64fc253654bc7adfcd8a4b5272792 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6dd1ff614854f508322b578a2f21e1e : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b04dd42d74e41a3b54c28dfedc8a743 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21940a85b75b402698f84c6836572a02 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e02828d4fee4ce7b6d6a51bfa0399d4 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f7254217f1e4a9383b41a34e55f71ef : SUCCESS in 7m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8bdd67329b3e49d3838086493277952d : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/16a7fccf12b441008f50025f0fc16b13 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe646c762c7a4dac840cafe10bbc8d47 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af9978847a8643d28d3f7f869556a100 : SUCCESS in 9m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f147c36f1b9b42d3bf5a330ca2b5464c : SUCCESS in 21m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d0e9a753db64403abf2513cc6bd851e : SUCCESS in 26m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f6b1c90697e41b0813006cf7a2fc479 : SUCCESS in 11m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b03c345eddb746449c00d9648d7d3f42 : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/541fbec8d97e4c85a26f7723e23865c9 : SUCCESS in 15m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a8827a6be60b4561b8a5c26e2e6d4b3c : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f8b170e103214d2899df9d6a437d6444 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2160859b7087448192a65ddf6e935e2a : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8b34f0425fe4f958f6cffac599bedf2 : SUCCESS in 12m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c51f5a09b2144fe9cbfaeedf352b502 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a27b9cc20d34783b50457073a991c34 : SUCCESS in 1h 16m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87c90e24adc34a478fd53e7281f27fff : SUCCESS in 1h 08m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2db9e89755894acba17ac8466f0d16c5 : SUCCESS in 1h 33m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b85bcab81b094854a6c11e44a7802aee : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2d8b722ca7c565f43331d37937af53e884e6cbcc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-05 20:23:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"520dadbacdd7b413716754c93492e3794c9b6ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 20:24:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71056e77f734109731529ed2af29ed97429c45bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:13:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f31621f518a12fa62791a0c731548798d9276f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:13:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88aa9160b44141e6af84aff0ae450f77 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e0968f678d4cec8a3ad050d8fede10 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6d516beee5f4ce3a50d427cca4911f6 : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb22cc62913742699144fcd4435cfe24 : SUCCESS in 8m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b95ad950df9f4b0ea78332e37c24bc04 : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb61ad5ad69e453e90ff76844845e910 : SUCCESS in 7m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc2ce648cea24b5891814aa932c4c9a1 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e27e13ef225d45ef9c8d7e58a8199aa6 : SUCCESS in 11m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/87d1896edd5d4911b57e614dbfde28c4 : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bbe84737c4724a13a0a0242bf09876c8 : SUCCESS in 13m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f26f18e719f34d5ab21d12b605678427 : SUCCESS in 10m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f818a00132f04d91b394f05a3611b09e : SUCCESS in 15m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a340b860aed5447fb111d24413fbd5dd : SUCCESS in 30m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d30aa97220674d88b674a02613e47cd7 : SUCCESS in 18m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/baf2dce9b455479fa21e5ac6c42bf735 : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88fd4f1522de4c5e84b1452188acde69 : SUCCESS in 20m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/022b2e29d1ae4c01b950f7e163220645 : SUCCESS in 6m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/29f11fbc75d34ef2a777bfa85c8073d7 : SUCCESS in 8m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95e6d88ad6d3489db2f48d897d32e3f0 : SUCCESS in 7m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3985f79c4061481092f4b1a8a41c3cd7 : SUCCESS in 14m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cd6b074bef44ee58dd2da52f54e9d79 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dd775347b2746aba7eebdbbdbfa56ba : SUCCESS in 1h 02m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/669a21ab6ca54642b17153e32ed30b8d : SUCCESS in 1h 18m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/98cc744c292b4e0a808753b082c889e3 : SUCCESS in 1h 38m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b7b7ab1b0394a39afd2a293f9eb2eaf : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":1},{"id":"be310d5a484aea19f80f4bb2566a0f8ceba49fc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:15:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994b079b541c46bf8fb2f01dc482f2b8 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78bbba90a3f8b558194cf5642db52ce87c5a1105","revisions":{"78bbba90a3f8b558194cf5642db52ce87c5a1105":{"kind":"REWORK","_number":1,"created":"2020-02-04 17:05:01.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/70/705770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/705770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/705770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/705770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/705770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/705770/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-04 17:00:12.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-04 17:00:12.000000000","tz":-360},"subject":"Update SQLAlchemy to 1.3.13","message":"Update SQLAlchemy to 1.3.13\n\nDepends-On: https://review.opendev.org/c/704425/\nChange-Id: Id818eac1a3e1783c1d3622a705dfd01e50e230d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78bbba90a3f8b558194cf5642db52ce87c5a1105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78bbba90a3f8b558194cf5642db52ce87c5a1105"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
