)]}'
{"id":"openstack%2Frequirements~705806","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I6901c00aa7a59848b6f4417128f9ad7000383aac","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"I6901c00aa7a59848b6f4417128f9ad7000383aac","subject":"Bump diskimage-builder to 2.30.0 in UC","status":"MERGED","created":"2020-02-04 19:14:29.000000000","updated":"2020-02-06 01:12:31.000000000","submitted":"2020-02-06 01:12:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705806-1580951551377-bbef5f2b","meta_rev_id":"8f5f05791064272fa7391f76f399003180c8d511","_number":705806,"virtual_id_number":705806,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 01:12:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-04 23:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-05 21:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-05 21:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 23:46:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-05 21:54:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-06 01:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc817f8093ff6c5b1923c7f38c737d4cca648955","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-04 19:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e4dfe924c5fecc4c1b35a2027b9e22e45756b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 21:44:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1a5f4800c46e463c8a90e4624181330d : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7e8f7deda54b54b46ffeac7f9f6d3a : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b81a5eb072f4ee0a491fbf3ca0ba91c : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14a8276ba0244d4aa4e7d263983632b8 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6b5db5904434c0785fd5d5fd090b2e6 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/212b1df61e414f8cb48916805984c511 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c64c3be593e1421d872e29e5a93d36d1 : SUCCESS in 4m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2004917099bb4b36b2dcd5ee13018924 : SUCCESS in 9m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fcc7e82ab2442eb90d329e4c00ecbac : SUCCESS in 8m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/34c68aa952244f8fb2ec13f692b91c36 : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7c6e28cef6b748fdb735e835e9f4d057 : SUCCESS in 10m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7fa97f63384e4bd6a6280ecf48569965 : SUCCESS in 6m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c9cc5fac2e594ca2877f9bc5eb7c5064 : SUCCESS in 5m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/18b83cf4f6df4717b17400187d4cc612 : SUCCESS in 14m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4883884b75344969ee6f85bfe2cec71 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79a17dd9c6da4aa9a78629326566958b : SUCCESS in 7m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/118a6a280d8240a38bbee69c3df37f71 : SUCCESS in 16m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/94c1447f55bd4c26aba67211fd2d085f : SUCCESS in 14m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c69a1bdcaec44f729c40648b539af56e : SUCCESS in 26m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b8535abf10c4ec79c1f3b738a577809 : SUCCESS in 28m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/eda1784f26404914a7dff3c4a6817542 : SUCCESS in 13m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2c241bfd3dc0497cb7d5f0feb75b782e : SUCCESS in 14m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0060f51bb39742d49fe99da241f280d6 : SUCCESS in 17m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/297bc0ddceb64ea49a80006e6fac0fcb : SUCCESS in 22m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7de4e6ce79324566b39d3b457de6a1a6 : SUCCESS in 6m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90143abe5500408bb4e71597b6f3b265 : SUCCESS in 10m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe980808308d4cb8965f82df347bb7a4 : SUCCESS in 9m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31105ab762d6497f9bb90b4f856dcbd8 : SUCCESS in 1h 11m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba2c4630f84a4638a7cbb07494c5fa21 : SUCCESS in 1h 42m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/207144e69b164d81bf2688ca6a31b9d4 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a20f6c1608d8c137101950686da4525f6a3b9050","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-04 23:46:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea72ed487c2e778e2e438207c8c3ed41d47e6d76","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-05 21:54:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2ec393915722b192d984c08a2155a0994a525a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:54:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"405a0c6c857c332d413cf392d553b49544caba55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:12:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f5f05791064272fa7391f76f399003180c8d511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:12:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d43fe74c49422d8b27acbae2f838f3 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c5e21cba824dcbbabf631b82496574 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37d747d44a9b472f9646fd35d81b6b81 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b699c51b465448d69ab3df1e6b92f165 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa4fd86a729044ba85fbc4799e30ca6b : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/976c0f8848d34fc383abd29ef051d95c : SUCCESS in 5m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4e11477555034724b626d0b067aba3a9 : SUCCESS in 8m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a24ba383af20498c952a6b4b9d5d651b : SUCCESS in 7m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f7d81b05b4e94a9c9aa1b9bf59ee1b23 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e20a8e0cdcf14806bb8d68bf4a923d8e : SUCCESS in 10m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/73094895bdd6476ea4a9dcfeac4f7aed : SUCCESS in 5m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/45c1538b52114cd490657708f77c6e73 : SUCCESS in 5m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/06808fdcdfd24b1ba6ef91b6ffb3f452 : SUCCESS in 11m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5951f3eb8862419fa26436ff83d0f5d9 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a35bad73c134745a6da4cd4be111556 : SUCCESS in 7m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a436d9edbda44bacb0f64dc52b84f273 : SUCCESS in 16m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/684f6808af48416197e9c7824ffc953e : SUCCESS in 14m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/60963b3821564c0ab1d8741f271e6553 : SUCCESS in 25m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c70f5dcc08224f16a34fecb71b397a95 : SUCCESS in 27m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/15671780ac624700b8b114f88f29f248 : SUCCESS in 14m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f9d435d195c470ca5ba9f129bf48605 : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c3e609ee9934482a8e0cd832a639d007 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a22047584ba4e499af011aed8398b4b : SUCCESS in 17m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4c58d4ffac046e0b31b37a75cc85937 : SUCCESS in 8m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/66841e1acbad4e2086bf60e6daa0658e : SUCCESS in 10m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3385e71b6d7a41439ef1f5513329ea4e : SUCCESS in 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f7cd684db4d4fa6912b01c7d94ffe15 : SUCCESS in 1h 08m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/502225d79ca846669097160d2421aaf4 : SUCCESS in 1h 53m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ba6b84ada694171b7967499908603e6 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18df89f228bb5c9bd53668c3cc1f9740ced629a6","revisions":{"18df89f228bb5c9bd53668c3cc1f9740ced629a6":{"kind":"REWORK","_number":1,"created":"2020-02-04 19:14:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/705806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/705806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/705806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/705806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/705806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/705806/1"}}},"commit":{"parents":[{"commit":"7d964de99c6bb0481f48b98ce32fe19b52697032","subject":"Remove trusty wheel build jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d964de99c6bb0481f48b98ce32fe19b52697032"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-04 19:09:55.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-04 19:14:28.000000000","tz":-480},"subject":"Bump diskimage-builder to 2.30.0 in UC","message":"Bump diskimage-builder to 2.30.0 in UC\n\nDiskimage-builder (DIB) is a branch-less project. Unfortunately pip\nmade a change that required a modification to DIB[1]. This patch was\nfirst released in DIB 2.30.0.\nTo fix image building on the OpenStack stable branches, we need to\nbump diskimage-builder to 2.30.0 in the upper-constraints.txt of\nthe stable OpenStack releases.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I6901c00aa7a59848b6f4417128f9ad7000383aac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18df89f228bb5c9bd53668c3cc1f9740ced629a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18df89f228bb5c9bd53668c3cc1f9740ced629a6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
