)]}'
{"id":"openstack%2Frequirements~705994","triplet_id":"openstack%2Frequirements~master~Ie075d2a9962f0fc837894586f18d5add3ddfb0d0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie075d2a9962f0fc837894586f18d5add3ddfb0d0","subject":"update constraint for reno to new release 3.0.0","status":"MERGED","created":"2020-02-05 15:58:16.000000000","updated":"2020-02-06 20:43:09.000000000","submitted":"2020-02-06 20:41:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705994-1581021671551-4edda447","meta_rev_id":"91f0064630ed96e4e0fff78ff355e72f9cb8de9c","_number":705994,"virtual_id_number":705994,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 20:41:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 01:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 16:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 16:38:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-06 20:41:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6329464dab52e7ef782fa1f1244459aa0b128469","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 15:58:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9acb95273a625c93c9ed7fd230ee97624fe3aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:09:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a24d9531a87347f696559bb03b463b50 : SUCCESS in 7m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d86eb8d21e8a4d34a4935486328133d2 : FAILURE in 5m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81c0d5586c3041b088f609ab987373a2 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fdcd41aacb94fe8b17d95b8e9ec1458 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ebe2be4073f4876b02985a595327c2d : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36bf1260bb0a4ea7983dd3ed1ce2e160 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c42b46577d74f90954c94206e89c020 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a657305ba19444c9b998acf93b1c1201 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cdf2bab397d74d75b49a015c90aa0712 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/31aa4089d9e54ed29a407e75df25eb55 : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d17152aee5834aa8824fd4df6b2b70de : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0f12297d79744a8f8c8f26f25e6d0143 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c9ef9547c3eb4229a7358644a69318ed : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/16186dd9792046fc877740edebddd698 : SUCCESS in 24m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/594cafedea8845e096f62cf78deb74c1 : SUCCESS in 33m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9dfbd526aaff4df2bf5bb21edd3b330f : SUCCESS in 19m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/15bdd0df32cb4675b6f46197de712a84 : SUCCESS in 18m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/359feb7e5953419592d8c08e6ce9c00d : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/331425c5ffb34d8ba3b5813fd274e991 : SUCCESS in 4m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1b29916a9bde4aea98431cb490dd49e9 : SUCCESS in 8m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1df895ad4fe548c6b3b7e9340ab4a860 : SUCCESS in 7m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e249e5f16b254aeb9367dafa40e5cf93 : FAILURE in 4m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d19e9252ae84a7887e2b163265d8ce9 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72e5345bf24947c8916805dd4741fa76 : SUCCESS in 1h 11m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8fa5406af3b4ca489e98e61b471f490 : SUCCESS in 1h 04m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4472f4c791df4f8db04c265576cbdb21 : SUCCESS in 1h 34m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cb07b6a89daf4b63962e5d371d5f8b9f : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"54977e349feddd584e75dadff49cc43af152debb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-05 22:22:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"44827aeece43106bec6125790bdf4ef2ffdf14e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 00:13:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c3fee889514fb497fe4c0964d64e47 : SUCCESS in 6m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3fd3ab8fa0424b5eaed2c4e5b873d68b : FAILURE in 5m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdb555835be44448dafca3abebc43c5 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2000ac814bc749ed9870941d7f92a36c : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6f41439dbc844f19ec7ef1793240f1f : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bd76a60c1b4cd38b22070194144829 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f00c089acff944f8ace418fb82587501 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb14afaa84624f5bb01dde359c9965ee : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a2031ca50d3e4e319a7e252181e38a85 : SUCCESS in 10m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e66d1d6b817e4886b624b6a256877265 : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/be50f13b50b54e3890ae5464babad90d : SUCCESS in 6m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/72b6a517654e4e9bb790a919129755ed : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c34aa628f624aa2a959e0adb4b4af80 : SUCCESS in 8m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53ac5beb1ceb4c48932d704526e7e0c5 : SUCCESS in 13m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78f9d05e8d294d69b6278c69df7d9e19 : SUCCESS in 24m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0100a49dcfef46eda778a641c09bec6e : SUCCESS in 16m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4864c21153814a069275180183b78f63 : SUCCESS in 20m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/504386f312e2498bbddffb29b691593b : SUCCESS in 16m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c07323d931604debbf36b470b4268116 : SUCCESS in 8m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/22413a000d03424288a2dd1b61daab9f : SUCCESS in 8m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b82f74c9f0a745b884c6f417110608ce : SUCCESS in 9m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fbedc6e87f2b44849a7e4abf5de32196 : SUCCESS in 9m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/528d6b27fb3047ea88199e34a781a966 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33647563a2244d329894adbbaac25b04 : SUCCESS in 56m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a62bbc3b4de415d84523aa327af539c : SUCCESS in 1h 08m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/62013e191a6c42778c1a959f55759452 : SUCCESS in 1h 31m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e33f449c1544d4d9d788d2989289a5d : SUCCESS in 1h 06m 57s","accounts_in_message":[],"_revision_number":2},{"id":"85fc3c4b3faf8f4af021277cb606840904a092f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:08:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8fe6fd6ed71212ba5b65beca9ae3c1366c4c6210","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 16:38:52.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"93a3bc12496357aafb0b7ec9e74ea526263055ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 16:38:59.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"10e36247ddae13521c777249714a5f378830bc8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:54:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"687ee0bcf4d393cb5ea9d5d8cc7f69a9989efe2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 20:41:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c2f8cf6d064c7e9190d10c79481dd4 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b35dddf4230f4491992f47fd0b40313b : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3d72f55a6194404b93710e97a9ccf7e : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1118532e3fc14b299e578e1966a9c4a3 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fa761e91031472ab293021efa583041 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/542e13fcb88b4f0a8d9c0fc6c1388549 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5cd692d313814f1382eff2987b0cbc53 : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dfdc463d74654d78baf514a90b619798 : SUCCESS in 14m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8baa2190514949ccb3aa3a28536a4c9c : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d9d0370c6b7f4f50a5fd86fc415eb85a : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e43e71b8943846268fecbe3b5f1f004f : SUCCESS in 14m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8404f4c0f77740f9960083e6f0ccbe88 : SUCCESS in 14m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d08c6d479b243dd8fee4a5388c456d1 : SUCCESS in 27m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/da1e06eda3bc4086a92d2ab5b4e4f00a : SUCCESS in 17m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d6cc2b09deb42b2aed86c9968ac1066 : SUCCESS in 15m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc9483f6cc8f416b8ccc5aac48201b93 : SUCCESS in 16m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8fedec3927564558b423953fbe5dceae : SUCCESS in 7m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8f8f7bbd213a4f85b75ac252c4e1f24e : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e53ce39a124343388de31b22be66fdf9 : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a0373810c2a24b1a9b2206bffbd9242e : SUCCESS in 13m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/112fe30def6f408cba00728764540b0d : SUCCESS in 9m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2249314eae064255a1090b33983583c5 : SUCCESS in 54m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f24bc421ce14c78bb99335944b85960 : SUCCESS in 1h 20m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/35598964a0c843cb8d6604e1eb2c3616 : SUCCESS in 1h 40m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9d799760ec474c559737be013038a768 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4d677a1ed78fd5ef1b9c5743562b3d342842eb4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 20:41:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2dbac5d059f029f491b0a2c54ca89c9904b1a8a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 20:43:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1229629ff03e46bda5439ad64fbba474 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"50703abfacad0266cc7ccc2d319d1f24787459b1","revisions":{"c3a7b8e0363bc5a140be3f12e0c5176ca4d827dd":{"kind":"REWORK","_number":1,"created":"2020-02-05 15:58:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/705994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/705994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/705994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/705994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/705994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/705994/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 15:58:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 15:58:14.000000000","tz":0},"subject":"update constraint for reno to new release 3.0.0","message":"update constraint for reno to new release 3.0.0\n\nChange-Id: Ie075d2a9962f0fc837894586f18d5add3ddfb0d0\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: Ia8f07af5313c64b22bc29e919bc3b491f3d1dd87\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3a7b8e0363bc5a140be3f12e0c5176ca4d827dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3a7b8e0363bc5a140be3f12e0c5176ca4d827dd"}]},"branch":"refs/heads/master"},"50703abfacad0266cc7ccc2d319d1f24787459b1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-05 22:22:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/94/705994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/705994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/705994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/705994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/705994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/705994/2"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 15:58:14.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-05 22:22:51.000000000","tz":0},"subject":"update constraint for reno to new release 3.0.0","message":"update constraint for reno to new release 3.0.0\n\nDepends-on: https://review.opendev.org/706137\n\nChange-Id: Ie075d2a9962f0fc837894586f18d5add3ddfb0d0\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: Ia8f07af5313c64b22bc29e919bc3b491f3d1dd87\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50703abfacad0266cc7ccc2d319d1f24787459b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50703abfacad0266cc7ccc2d319d1f24787459b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
