)]}'
{"id":"openstack%2Frequirements~706111","triplet_id":"openstack%2Frequirements~master~I20e7f73f22f78c7ec9a6eff9282755c641212ffe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I20e7f73f22f78c7ec9a6eff9282755c641212ffe","subject":"update constraint for oslo.cache to new release 2.0.0","status":"MERGED","created":"2020-02-05 20:52:35.000000000","updated":"2020-02-08 01:36:00.000000000","submitted":"2020-02-08 01:36:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706111-1581125760034-ed07d331","meta_rev_id":"7a4d713e634caf5e284ad001e5fefbe457df688e","_number":706111,"virtual_id_number":706111,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-08 01:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-07 23:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-07 23:13:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 23:13:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-08 01:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df3cc534421e4a9d451cdb86a08b48d3c6f7be99","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 20:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c06cea57ecf1667bbfaabddcbd7a686b3424ba64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 22:30:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971da4de63874d12bcb1d716b551555e : SUCCESS in 7m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6c83490ca04b482db50d241c4270ebff : FAILURE in 4m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25ce4c8ab6a4a4787333b77bfaa8390 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/390d52209ff24cbaaa8cc41908098d94 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2768d428510740ac95e0eed3f25b53ef : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d04ee03d69542ecb624a5e790f66344 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c865d2af5c9f44958079ceb12200c261 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be4961e5f1d34de7b0dc85f8f433d295 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7cb040a08ff347afaaa358fbb7e790d1 : SUCCESS in 28m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/154fa2662b5c4d348b85399071524e49 : SUCCESS in 10m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d66d5d8a5394312bbf712d70cc4789a : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26e4fea152fc4116bd4d908034530302 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b1abc422a3e42cd906066e8c22a64a7 : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8ba8ede30ef5437fb60a30e60362332b : SUCCESS in 15m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f491d553f34b45aaaae3d5ed89aaef98 : SUCCESS in 27m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/137cd909ec1e491a8dfcedcad19631ac : FAILURE in 5m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5e76d06544fb49bea298d089a4f76e8b : SUCCESS in 12m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b18777d403ea4a90a2b05cb3aaa5de53 : SUCCESS in 15m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aacfd6ab19fe49bfa21c45b5f1a284d6 : SUCCESS in 4m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/296557bf810d4bcbbb8b1a0b1cb5a21f : FAILURE in 3m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6bc27ed88cc45c1b052c4e18d0d1353 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3033d0e239214f5a9db44388fe43a632 : FAILURE in 3m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4acad71658a8416daf72c5068b3c92d6 : SUCCESS in 9m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1221134e079a400e9e008923c2df223d : SUCCESS in 57m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb6fb1d3b2dd49e9b022b62017c8f4a7 : SUCCESS in 1h 08m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f968bb7236041cf8ab86824c48f6f40 : SUCCESS in 1h 32m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1f35ccb3fa484c22ad9d7f7d12b9a218 : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":1},{"id":"97569979ada01d7b19d2c54063ddc2cd95f450ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:08:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9fe0ee33bde8f95ccee0561045bfe6a91d0d849f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:06:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804b080cc3894e5f9acc34c04f4659c5 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1a73ff8e0547edad4ef2014e55f185 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39d35e3664fe4e70a7598081ecedbef3 : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/506fb5619ac04a499e80897dbb2c8959 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fe466a29dc941d8a4d0519475342b41 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e350cee5a02640f4ab3e543af0967016 : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1be2dcc1109e48c49d51f6e450623cb8 : SUCCESS in 14m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db902e50c756414d945d689cbb8a440c : SUCCESS in 12m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fec65aae840f4e958e5c35077314e0cc : SUCCESS in 10m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9fd072aadecf4c6fb889a3d0d3743fc2 : SUCCESS in 11m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/38e0f930986c403bb1ee413b12fd66e3 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb88aaabb3ea4cbaa2cba86b8d833e3d : SUCCESS in 11m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c313d99b94d4884a05032bcf7ed5c69 : SUCCESS in 20m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9e12fd49543f418fa6bd029e694f1b5f : SUCCESS in 30m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e6493d635c3486caf4eddc388a4f1af : SUCCESS in 14m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/38c9fb161f304715802bb1645bdd7f2c : SUCCESS in 17m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8aa6629d9a7f43c9b0e254409af6b0e7 : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/217b713866cd47548b99bcc15a5732d5 : SUCCESS in 6m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dbb140aa45244d95897eff2325683b97 : SUCCESS in 10m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74dd8da569944de5ab80e1037d0bfb9b : SUCCESS in 13m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/061951525b5e45d2bb6bc43091335bae : FAILURE in 5m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/135ceb1f91904d10b14333899fc65d29 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c15a99ea74b04c149d45d7e050d5decf : SUCCESS in 1h 06m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eeb4bdc5b1ca49fe86eb7492139a664e : SUCCESS in 1h 15m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1c6f1d05d680483dab68115bcbdcc9b4 : SUCCESS in 1h 42m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a529de97aff407aa118ca065244397d : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":1},{"id":"09883a80e2780c974487add38044862ba623dce3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 04:05:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9614c3778a16ac84234348804002bce9b9206155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 05:37:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dab614829124294b66752c71122931a : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5801961abf4144f784b97c7c08d0cbb7 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83532c7a747e4a26a8c014e46631aab3 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ada2db5c314408689bf37765ad8e81e : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26a1c4f18a404baa98c60dcba4ce1ca2 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2793813eeebb4d08b83a063d0c271b80 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d74454dbfee4925878d38bd6d67ffc8 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12afa88b2f8c4978a4a5464bf191f340 : SUCCESS in 7m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6c68bda352aa4bdbb5eda9618668cda3 : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2f3ac54c63534b66b49bd11b51274261 : SUCCESS in 4m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3371ccb190b8465aaed4f6a4451bca93 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1394c56c96ce4ad7bebacad500796a10 : SUCCESS in 6m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a0cffd5e9d60431c9dd46ec78d9ff6b3 : SUCCESS in 13m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1fdb3210f01459ab93264bc3870e5a7 : SUCCESS in 25m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4b71eb65976b4c9fb1db20b4c4b35a72 : SUCCESS in 13m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9b13c080c402419b9efa4fd3f90b180a : SUCCESS in 12m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0f07e4c76674ceabf710db83bffa7ea : SUCCESS in 15m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6491dc5416714d58aa73f7384a28a240 : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9517cde6f2354f668bffb741d2e0ad2c : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb0fce025f374e8492f781244d1a2446 : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f48d0300182e4ccc8732837767bfcb39 : FAILURE in 3m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/526ba6260746408da5d39ba3ed63cc8e : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e1f4bfc2dc0442dace7db7c7789f8d4 : SUCCESS in 55m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8503bb8367a64df0b51e9feb543fddda : SUCCESS in 1h 10m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/392326c347084d9e9413686ec4af59d0 : SUCCESS in 1h 26m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/06e66e7280004b5d94fec58ba2794d2f : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fa0d7234874e701353290b05e5a45855fb60e071","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 16:38:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b5c1d1a4e654eecba05bb92a3ef73095285a838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:32:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9d4462f9c942a0bd1020179867b02d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83817311daf348ff981f06f35bd483b1 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbb6fbd34768425f8414e7ee8bfa2634 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cbba0e22ce54846aab611482a0dcd36 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e00427ea0184b538cdea018dcdab7b6 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57524bccadbe4c3abcd129453aba6bd5 : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/802d296e69e74eb990466fd8e31d2448 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f636c547d3564b2f93293f27a5414443 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/63b829967ae949ae8e23c5bbac76e274 : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c5064c2c8ff34909bfae4848855ada43 : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/283ad4d5197b44c59856fc992cdbb854 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e3a062587a143aba66cba98a81981ef : SUCCESS in 9m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b518261013dd4ce4a6165d04bfd0e2f7 : SUCCESS in 16m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/67b26657c6ad4cfbb06b5c6894ed884d : SUCCESS in 34m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd23c58475a64ab382dc2ec19fc4bd66 : SUCCESS in 18m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9fd1a3be821247c9aee6edcab2294169 : SUCCESS in 17m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8550688f70924eb7962bb74ffe26e6ea : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9c7587283fb4d32a0d50efd1837481b : SUCCESS in 8m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3bc7b0ea359d40499b6afad0d8751165 : SUCCESS in 9m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cbcce2147cc44ba936475b48858a2a7 : SUCCESS in 8m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/55faa77ea45b4f29aeccd22da6d1d394 : FAILURE in 7m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d12ebbe0a77c429b8e82f59422d64941 : SUCCESS in 15m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/605d62ba93c64e74b45f26e156d70ad5 : SUCCESS in 1h 00m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/942aba9911e8476daba0aee5927215d1 : FAILURE in 47m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dd9bacdd33e74dd38818da523fc13a17 : SUCCESS in 1h 50m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9be22256ef414b8592cc17a84b3e1858 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e649ca32c18beb78c66cf04e7240153ba23155bc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-07 15:27:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06e13e195e5e82dfbd165285d408e304e0fbe289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 18:39:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a44965fdb94c4c830773ec09c4222f : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdaa49742494c39b9085d51952c2695 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c449792907142d5a9b30b276d9d5491 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c67fa852e074e3696bee2216bc4e19c : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e22b22947f0847ceb8ba75cd876e0a14 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9af2c26613842a4a6dbe3292143a366 : SUCCESS in 7m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1dac3c0855b34700a4cd2d33568c7633 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93b41be80702412abf58d1b40a650a4c : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/486c5f774c104eea918a20140be3547a : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef75e34e24544f5faa42ccd02aed8e31 : SUCCESS in 6m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/685232fa83e2407fa7eec40ceb231036 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1ba031abcfc486c921aa7e950d59061 : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db4d51192a5c45569fdea7c4f1b56410 : SUCCESS in 16m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/402c56fd3c74468a99d63a133ebc42df : SUCCESS in 26m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/94ab59574ed446ce8eed10afb128f84f : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e2dd49aa5fbb4e4dbe6e13982571b9b5 : SUCCESS in 21m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2992a16a0dc4e1db93bb322d88cbec9 : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce505102902f4d2fa04772c30490e996 : SUCCESS in 5m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c48f4c30062f42b69709abd80131eddd : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1a77f8c57644b41bee0509be1ba4b13 : SUCCESS in 8m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7725d35e6e3d453a9ede224fc0733222 : FAILURE in 4m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/347341ce6bea4a7580a65055dc3c1247 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5d8c2c596fe4affb579b3b59f8dd1e6 : SUCCESS in 1h 01m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7815ae829974e778382ac169469d376 : SUCCESS in 1h 07m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c41a6446a4744ddcb02bae384eea2a20 : SUCCESS in 1h 33m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c108963a303d4458adfa85e2ed1a8508 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7167cced173e4dadbe54725a834b58a590ade0a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-07 19:37:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7c6ba4924fd2c061f21efeb4ef66a0a1ad9261c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 21:40:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4df7c192564e3cac7b2386595f2909 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53872e42aac4f1f90b358f382c41661 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a57c5ad3ba89437f981c3055ead8923a : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/216fbc9b2eca4cc4b429dd8b61d3d5a5 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bb0607bb7a4769ad5a4773592ef7ac : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f1796b533e04f9b97891ca1b5dc49a1 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1d27f945ed749daaf543d34fd6c8c87 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a73bedb464845b3ae4ebe49a6ba7ad1 : SUCCESS in 7m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aad15153f7594f918510e901904674cf : SUCCESS in 12m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eec8ca6e95894e2d9973e1d05bc1d132 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/988924a552c94f6e8f5d4c810e74ff61 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ab2f967e34e404f98522b9ed042c231 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c558e49e239440d29addba0afdca6986 : SUCCESS in 18m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/89dd6d10ee614c33b194b6b8e41398fe : SUCCESS in 30m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7bce3ebef98b4ffa8eb446a6d93a172c : SUCCESS in 14m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e83fb4658f784c35ac0f971ec2a1d1e0 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efd3a94e0d824cd995e78e370ad69200 : SUCCESS in 16m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a4e88dadef1b4f28ae1f2bbc5a1d677c : SUCCESS in 4m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/55e79a0916ae417aaa0a315fa08f3773 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc96768f0aeb450daa524137a486c3ac : SUCCESS in 8m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/88a5e067e03c42b384193a73afbad7e5 : SUCCESS in 13m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5204d7ca526249eaaf30f3a0d0a413e9 : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cd2fd2efd6e401ab5ba96722e395a4c : SUCCESS in 54m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/436f0d0bba3c401bb5644743c8d30110 : SUCCESS in 1h 12m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6081d3a7d9d54e94a79ec72a18049424 : SUCCESS in 1h 31m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e8dffd52feb644ff99157ae543460e90 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":2},{"id":"046c774ca5cce7584a987d7d6c40011f5fbcbe17","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-07 23:13:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba8ce3723a63fa8bbc8396ce7c7538e102ddf895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 23:14:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"50472b4b7161fbd67fd27276ad6b83069040e35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 01:35:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df2616cf0794a46a7df5c81e1b4d5d8 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b230c9bdf24fbba829fb48bf0404ce : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8776f68caebb434cbc5809a93c12cf51 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d410e7b3cdf04bc2a7c6d935e14c0d8d : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7daa3854964a44e387619bce11144764 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41426cb86d4d4763bb4e3bfb14d8a0ac : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a2b56af878f040e6bf4e4857d853af67 : SUCCESS in 9m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/35696b008dd74a5e80d5f2df92c578f2 : SUCCESS in 10m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2945bbe0ecd5438f99abf1583a7483d9 : SUCCESS in 7m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/719b93139e934b4da29542df71ea0759 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2675c55e881c4ea1bb9ad61efbac679b : SUCCESS in 9m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90ef9015dd55490cb2e713caa0da6e00 : SUCCESS in 17m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dd7dfbc25bea4fa282d1f1a6f80aa4cc : SUCCESS in 31m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed2d8611f8b34bd2b9dc74055074e09e : SUCCESS in 12m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7e4ec37f4a664469be83702cd368ec62 : SUCCESS in 14m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2d53c99a66348ad98570ef351c957f6 : SUCCESS in 18m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9d48d72ae5b24657b6687554da100614 : SUCCESS in 6m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f8f8a8e92ad743c289f83b9f345902b1 : SUCCESS in 7m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e790f26d72f749a984db9793f6369474 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c33d131f9b44d8f96752beddcf89f9a : SUCCESS in 12m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/507bc54bc1564fec858bdb87affa6ba6 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f20963a1beb4848b5ed854085722a46 : SUCCESS in 1h 00m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8faab84e924414992a1baec3a7784b2 : SUCCESS in 1h 16m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/07237d476cd248c698a49b8ee291ccd5 : SUCCESS in 1h 26m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e46cc5b340dd42d5a0a6abd729f40350 : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":2},{"id":"66b1b6dee0644b5ce1f59d9f02dc180fd9e12dad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 01:36:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c7066ae6d53932ca14a973ffb38f0222697470a","revisions":{"2a148bca3bf734eb876cf80c49bc556e77d2acd5":{"kind":"REWORK","_number":1,"created":"2020-02-05 20:52:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/706111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/706111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/706111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/706111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/706111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/706111/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 20:52:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 20:52:32.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.0.0","message":"update constraint for oslo.cache to new release 2.0.0\n\nChange-Id: I20e7f73f22f78c7ec9a6eff9282755c641212ffe\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ie05bf00fc0b3721eb5c80b1d9bd8c2689786f17f\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a148bca3bf734eb876cf80c49bc556e77d2acd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a148bca3bf734eb876cf80c49bc556e77d2acd5"}]},"branch":"refs/heads/master"},"1c7066ae6d53932ca14a973ffb38f0222697470a":{"kind":"REWORK","_number":2,"created":"2020-02-07 19:37:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/11/706111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/706111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/706111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/706111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/706111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/706111/2"}}},"commit":{"parents":[{"commit":"8a73e436b723fe5a80bb12eba2a7b763435b36e4","subject":"Merge \"update constraint for oslo.db to new release 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a73e436b723fe5a80bb12eba2a7b763435b36e4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 20:52:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-07 19:37:42.000000000","tz":-360},"subject":"update constraint for oslo.cache to new release 2.0.0","message":"update constraint for oslo.cache to new release 2.0.0\n\nChange-Id: I20e7f73f22f78c7ec9a6eff9282755c641212ffe\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ie05bf00fc0b3721eb5c80b1d9bd8c2689786f17f\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c7066ae6d53932ca14a973ffb38f0222697470a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c7066ae6d53932ca14a973ffb38f0222697470a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
