)]}'
{"id":"openstack%2Frequirements~706122","triplet_id":"openstack%2Frequirements~master~I33208d9ba41277311a4741c72bd27f2787ed3f1a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I33208d9ba41277311a4741c72bd27f2787ed3f1a","subject":"update constraint for oslo.middleware to new release 4.0.0","status":"MERGED","created":"2020-02-05 21:46:51.000000000","updated":"2020-02-06 02:56:07.000000000","submitted":"2020-02-06 02:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706122-1580957767592-798fc2db","meta_rev_id":"405afa7434d6daaaf26e4c4cdfd4825bb1847602","_number":706122,"virtual_id_number":706122,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 02:56:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 01:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 01:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 01:09:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-06 02:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc9da1d8b21596b24c958305cd4dd7a93fe72009","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 21:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"903cc990ae6ae39d8029482c73f17414f1827dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:31:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f91ae908ca7434eb959166682966fe0 : SUCCESS in 7m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9d984c41a0ca452cb4cf77cf778efe80 : FAILURE in 7m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b991c8a212842d4abe525612fc4d33f : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ce8096c2f7b4a9f96a150bded36fa15 : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ceb6d94b8814445b3648034cf771309 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4238951142f49ea8efc8236ad4588ff : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18993ed6e29f4eac87163b26a4863796 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97e0076b5ce749d7959a142b80268a6a : SUCCESS in 7m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e3c5a1b588b749bf87c50f4f40947eb7 : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/72068784d647481e9bc2306b7e2f5ea4 : SUCCESS in 11m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/601d78a40ffb45c6a109e82e0b4f1b4b : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a2e8ab9248474a029e9efddcb2d4b885 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8944d66203994cf99f6c45fc1c7d9cc0 : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05f85487ea1c40a6982baeb0a22b2427 : SUCCESS in 18m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/65416453b0424e4095183de2b6671efe : SUCCESS in 28m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/50a60a66151e4a0db44faf2acf3fa236 : SUCCESS in 15m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fd5927b1940d42099fe818828ead6d4f : SUCCESS in 14m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ada03a5f58d4f0ba18df7a54254d0dc : SUCCESS in 22m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5bb9e7fd1d214c7d8614ae605a6b9c85 : SUCCESS in 8m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8b16df63e3a145159134173c9f245c3a : SUCCESS in 8m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd4bbeca5d6c44c8b9f64e8389b9b3eb : SUCCESS in 8m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/73651a0b535e452182c9d90e701ae32c : SUCCESS in 11m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/142a1d8f26f6452f9879e111815d3d2e : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31e0a2e595e04dd3876f933859bbf763 : SUCCESS in 56m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e493468d1d74004998eba01f91e4483 : SUCCESS in 1h 10m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db34f60609bc434fa8b235ac9974d9ee : SUCCESS in 1h 40m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5885b5da9e4f4976ac465017c4828536 : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":1},{"id":"06fb390eae66a234cfac3006e20bd8d49804cabf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:09:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77ffbf1b40022f36c2c74d3842a34b6c80ed756e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:09:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"460f4e7dcea12eb2376d381ecc82c177aaa11f60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 02:56:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b23ecb12eb42f10c2041105502f374724e0cf700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 02:56:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e045991bb958493296fdc5e2a695787d : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fea0f3b29f4a65909fbef1c8ea58e7 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bbe81e8548d4895866150df7c302f21 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8570b90764a40d9865d89285fe8cdeb : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9eb82a0600f401bab90bb2119fe5928 : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/333e0f45bff84541b9e6d55b6663744d : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c8085d58f1d4dd791e9fc0fa5849473 : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8fe406fabc424d49931b067d9405bdad : SUCCESS in 8m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa7b77b07b9a4f32a7266d2df9f2e784 : SUCCESS in 7m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5fd84561ff3f432f8c55cc7344361e82 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6efb372eecf24fdc82c3fffc3510f00f : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9ea22331a3d547468dfe9d1428cd3feb : SUCCESS in 15m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/39541dcbc302420ea5e5acc659b5309d : SUCCESS in 24m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2cfa5b2567364b67ae0acb38b389e7fd : SUCCESS in 13m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/702fa36edd2144bfab2fe472aa6a1036 : SUCCESS in 14m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f153ec6f4c7482ba0c83b14ffd35d83 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f93f0ac37d24989b52a88ad3b0bac3b : SUCCESS in 6m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/16958dfb00fe412b985d9eb73cfcb934 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c620e69a829047b8808b1ab7143c9e5a : SUCCESS in 9m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5843bdf9a075483d8b7b90bda0ee14aa : SUCCESS in 12m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/980e7fa7e99f4bce8b75eed7ce030caa : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca5bc782a5074b7a92f587289b125408 : SUCCESS in 54m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b82ca71cf2fb4a73b03e22391284910f : SUCCESS in 1h 13m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e9666535231548f7b4cb7f1a90e2bff1 : SUCCESS in 1h 33m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7d1fb8ddd8d4af6aeebca8aa9729182 : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fd3ad254153aa4241240ff6192f4a506c0f60c2","revisions":{"8fd3ad254153aa4241240ff6192f4a506c0f60c2":{"kind":"REWORK","_number":1,"created":"2020-02-05 21:46:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/706122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/706122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/706122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/706122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/706122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/706122/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 21:46:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 21:46:50.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 4.0.0","message":"update constraint for oslo.middleware to new release 4.0.0\n\nChange-Id: I33208d9ba41277311a4741c72bd27f2787ed3f1a\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Id14f22f10dd0b200640ac74e47ad37bebde744de\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fd3ad254153aa4241240ff6192f4a506c0f60c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fd3ad254153aa4241240ff6192f4a506c0f60c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
