)]}'
{"id":"openstack%2Frequirements~706128","triplet_id":"openstack%2Frequirements~master~I8c10e87c2ca4a1196a86855bc620dc24878e6f47","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8c10e87c2ca4a1196a86855bc620dc24878e6f47","subject":"update constraint for tooz to new release 2.0.0","status":"MERGED","created":"2020-02-05 21:57:32.000000000","updated":"2020-02-06 03:01:46.000000000","submitted":"2020-02-06 03:00:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706128-1580958020041-5b512e01","meta_rev_id":"5b4120849edf8101baa2667389227ab482a2eef3","_number":706128,"virtual_id_number":706128,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 03:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 01:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 01:09:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 01:09:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-06 03:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4a725ba2727402052620f7e31505854431fea77","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 21:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0891296072cdf4a88f4dacea5b281745046c432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 00:12:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af741c08eb34067ab019a43556f93ae : SUCCESS in 6m 16s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ebbb17516d6743ca8807835d45b7480b : FAILURE in 7m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809411bf6e034e3c8b54ba24fe9851da : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81860c4da6214d6d9d616d4015b2686e : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac5bdda3aad043d0bf2a1c28fba8902e : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fca26582064c4f218a723384fd0a3e34 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0758865a90c94b12b6885d398b93e837 : SUCCESS in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7602c8b70814b16a46ee7f8f57f3320 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4262dbfeba8408e86e04c997cfab62b : SUCCESS in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20829df7a34b49da97fb93f5d904bd95 : SUCCESS in 12m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc9645cea545471b9758817adcab3bcb : SUCCESS in 7m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/37fdc518d7534295b2df229fdd6552db : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3dab1f07486644d4a2215049e60f1d84 : SUCCESS in 9m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3f0bdf12bf14f428626001667449e76 : SUCCESS in 24m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37e57e2439504384a3e679406ddda791 : SUCCESS in 29m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3413c24656e24777aa027e22f7c0d2f7 : SUCCESS in 26m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2bd13444c9584e93a79a9a8a17b05064 : SUCCESS in 27m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73fd2487f89a40f3a8712888efc59620 : SUCCESS in 24m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3fe8280cdb034412a29fe8d16997f56e : SUCCESS in 9m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c742e0bba1794080ab7fa2e7087c7fb4 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bccfef19dc3b4fd3a3ed3c62ba22a5fc : SUCCESS in 8m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7a9f56376d0e4e7c844c59b7db41b77a : SUCCESS in 13m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21d037c0567145ce9aa77b41f6173d2a : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b26f3a6865054fbdbfb3ab629922e7a2 : SUCCESS in 58m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b397d72147c493687860ad81a513598 : SUCCESS in 1h 21m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b466b01d00b0414f93dbe5d96afe9d25 : SUCCESS in 2h 03m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/659d143c356e49dfb4685658a6eb28fa : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1e678683c597afc913cae4805d0c111acf8c5ca5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:09:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9873df8891138eb8506af7be63ab786966aecb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:09:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b44b46430277471462818969d5e8875e77eb348","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1ef9d9a0c14a0a9b335a1c3074b1a9 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc894cbd3bf41b09e98cb5e8f55e00e : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad0e8f22468e417d903e019c374f4c26 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1177637024204e5786b892a31fb096fe : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c0427008783429eb4734a5188c28d98 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d13ff42802b74101b6953729ddccc388 : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/562cff604a994fadb73dbcef3f550dbf : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bc0d5fbf0fc94dd8a33533714ad7fd9f : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8020925651fd4658ae4a077df0dcd699 : SUCCESS in 7m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a9c8d11ca655475197bbb2f3312e9728 : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3845a66ef3d440eea11605e14ab846a8 : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e0776819bb824d2d8fc6f8d0157cff74 : SUCCESS in 15m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2e0aa6a38cde440e8fabfe5846b4a452 : SUCCESS in 25m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/905d8c76d16f423c926fb8b4abac58b3 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/57750057779b4f59bb8916084ab8291c : SUCCESS in 14m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3497c9ea76944169b5144a06613810bf : SUCCESS in 18m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/531408ed059b4fe6a682615a527a7ab4 : SUCCESS in 6m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2745fdb8720c45beacc5edc10a138ed1 : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8fdf9f383064de29e8bd9bf4842caea : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d8a7022f762345649494bb6d21d9c6bd : SUCCESS in 11m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/366b0141db6d4430b74028258f149768 : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e003bd1e6ea54bdebc325a207d93c7d8 : SUCCESS in 1h 04m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8b6569672444b55945a99eed4e82952 : SUCCESS in 1h 12m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b0680ba223664c9b896f0d4183ddb7ec : SUCCESS in 1h 42m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/98fd34f8cd9d4d6ea26c92d8f06b1384 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e7f7733b58a7ddaa8b63fa78c9162e39697b8d6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"95e9219e11a9f336b846db9bea39927845c0b527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04181f36aaff4daba0504734a67f088a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fefe0b281a0e437fada65e62b20550912987dfe3","revisions":{"fefe0b281a0e437fada65e62b20550912987dfe3":{"kind":"REWORK","_number":1,"created":"2020-02-05 21:57:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/706128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/706128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/706128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/706128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/706128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/706128/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 21:57:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 21:57:31.000000000","tz":0},"subject":"update constraint for tooz to new release 2.0.0","message":"update constraint for tooz to new release 2.0.0\n\nChange-Id: I8c10e87c2ca4a1196a86855bc620dc24878e6f47\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ie6adf3596941ff9efd02ba23cc4f59d8fb0f3040\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fefe0b281a0e437fada65e62b20550912987dfe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fefe0b281a0e437fada65e62b20550912987dfe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
