)]}'
{"id":"openstack%2Frequirements~706129","triplet_id":"openstack%2Frequirements~master~I51d39b63df9be561df57d244ad3c1af1006a6f63","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I51d39b63df9be561df57d244ad3c1af1006a6f63","subject":"update constraint for taskflow to new release 4.0.0","status":"MERGED","created":"2020-02-05 21:59:33.000000000","updated":"2020-02-06 03:00:21.000000000","submitted":"2020-02-06 03:00:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706129-1580958021236-71032318","meta_rev_id":"5ae7b91620e68d1886478ad147a14a9a3281246f","_number":706129,"virtual_id_number":706129,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 03:00:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 01:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 01:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 01:09:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-06 03:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3f1989fa9a19285b3bbd31b88888d11975fb2b5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 21:59:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26243b758ef61153f4ca16dc5587f61b2eef5866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:43:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968c1e29f3d34a6cb3adf5c0a6e0d561 : SUCCESS in 7m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9c7a59a50c374cad8a5f2a47010f6772 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5568ef27aab431d874dc5d55e476baf : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5277bd4f55047de93cdafa48d7bf3ec : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b580a5d267d41ffa0caa2df68654492 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6d968683e514c358777ee5e2472e896 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35e7631752364e13901661321bf6e806 : SUCCESS in 6m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30bc5ccc25074dfe90ab9145e9289c82 : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/142ab75473664bc89f6e2db595b2c653 : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a39de78ee7d4afda619b29e21b4c8c2 : SUCCESS in 11m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3e935c4e69544659c8cd7d434ecbff3 : SUCCESS in 8m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76472a6b6a604343be8f51ff7148da05 : SUCCESS in 13m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/394f77e5704c47119a9b710ad425aedd : SUCCESS in 10m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f5fe038a7904ae28db09e0286893c7c : SUCCESS in 16m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/96e4a41c45b04f899f884218763638cd : SUCCESS in 29m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3fd57d327f8f47b7a9739e3fdb939a33 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e87d6d5c5ec84c56b081725f933923cf : SUCCESS in 15m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8922440d06347a085bbefcc16dd72fb : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/11fa998b2cd24be5989b230723c4d389 : SUCCESS in 6m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dde71800ba644a21875e88f409fdfa84 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53d6d488ab004d0da1b1d09b5970e165 : SUCCESS in 9m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/84e0e991b9dd4d358db4ee37d8dc9780 : SUCCESS in 14m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/64793382d3ec4c0e83c3e0a07b3ea4e9 : SUCCESS in 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b68506b776c64a72a1fd42ebe7922d4e : SUCCESS in 1h 13m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/456c98e64f4b41bdbb16cd6191bb51de : SUCCESS in 1h 06m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7cb280be3468415e9d374c6397d216d9 : SUCCESS in 1h 31m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e52e9c1969694acfa44b52e8aaa3911f : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e830d9d8cc91c2e72460a9f51dfd4e8309fae02f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:09:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fab7dbc9d31acdfa919998c5c79939014ca5bfdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:09:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d0d2a353b827fcc5f90c89e264df8dce42740cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4d55da1f740b50965e9960adc96182e93b161740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8103722807434b7bafd87dc3876f787b : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99fd3016acac45208829de4829c6e1fb : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f3af05238c643bab09c2c2620fe55dd : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46a0882f202b41d2aa9889bfd38ea988 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d3785d75d8b4a9796c542162602c9fb : SUCCESS in 7m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f296e7d237704707a496eb8560d4f51b : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a765f9b060154b2e84e8a77764c63d2c : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f23210da60e84ad190cd1d4d39feddf2 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1cbaba2a27dd40998eed564e9f79b37b : SUCCESS in 8m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/03bafa6ea1054965b04309e37a5e9410 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66b842a00ecb4fae95dc71ae86e08dfb : SUCCESS in 12m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f3af5aa06e3c4b7b9adf98651420b084 : SUCCESS in 23m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/28ab1ad1e11b4fa8b05fda6f882499d0 : SUCCESS in 27m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d4320fa369e4bf49a69af5c38a30c94 : SUCCESS in 16m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a87fb689036e43619674b5ac5a312f70 : SUCCESS in 16m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9aee255531040ffa08121274ba2d822 : SUCCESS in 21m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/91f9f60bc20048a294c94f78fe0fdf70 : SUCCESS in 7m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/78d036ca93884be5be4e320b813cad9f : SUCCESS in 8m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c78b42087f0341c381810062635d9437 : SUCCESS in 10m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed5c8d83f87449e1a5028cfad1ba4aeb : SUCCESS in 10m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bdeb486913e74ae7a021c3769e2982f3 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a09d41b0f0c4fb6813630946dd07666 : SUCCESS in 57m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f20217b8f8e4870a94d9a0a99c5bae1 : SUCCESS in 1h 12m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d810780b6cf54c2994e6c4f56edc3ed5 : SUCCESS in 1h 24m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a1602b2192f341309f2ce14023b0b729 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"534616484e66eec3c789a41945bbd165b4950b45","revisions":{"534616484e66eec3c789a41945bbd165b4950b45":{"kind":"REWORK","_number":1,"created":"2020-02-05 21:59:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/706129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/706129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/706129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/706129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/706129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/706129/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 21:59:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 21:59:32.000000000","tz":0},"subject":"update constraint for taskflow to new release 4.0.0","message":"update constraint for taskflow to new release 4.0.0\n\nChange-Id: I51d39b63df9be561df57d244ad3c1af1006a6f63\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ie8171b6f0108f3826d1c3462c0fc7c204973bcdf\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/534616484e66eec3c789a41945bbd165b4950b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/534616484e66eec3c789a41945bbd165b4950b45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
