)]}'
{"id":"openstack%2Frequirements~706130","triplet_id":"openstack%2Frequirements~master~I79d0105537898fb82a4cfae38b101797164d53d6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I79d0105537898fb82a4cfae38b101797164d53d6","subject":"update constraint for oslo.service to new release 2.0.0","status":"MERGED","created":"2020-02-05 22:00:22.000000000","updated":"2020-02-06 06:57:15.000000000","submitted":"2020-02-06 06:51:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706130-1580971871735-ffdc711f","meta_rev_id":"24f804e6cb4f86e589d3eeba7dcc151dc9a1f344","_number":706130,"virtual_id_number":706130,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 06:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 03:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 03:03:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 03:03:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-06 06:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"313edd7e02e33f44cb6e115d1816f6eec26d8569","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 22:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ee3050e9fb0d66f186cc0bb7f7b9426c8cfb6e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 00:01:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e274cb94a1442949d6449878b809d05 : SUCCESS in 5m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b60d6dc29bde43d283c4ffce76fb526f : FAILURE in 7m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5998bbb048884176bf59b54fa356627f : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b4c113b9ed47daa6e4209efbffbe35 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/859ae365765d49b5b7111de86e2393ca : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9657aff35534a2d8c1dae2e8d706ddb : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6cf6c47ed15b4b4c82fc8a129ffd7b21 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3076df82b5b74265a0a4b364a9f65af4 : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea33e795ed80495d890e8196b98ff159 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d57da9ed3ba44349bb0b675f30995336 : SUCCESS in 12m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2362fc73bb4b4c699bd5ea2e01ddb6f0 : SUCCESS in 8m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/071db04244914c79a783a5afcb43b906 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d92d39fdb01d4592aae34f8de30fd1bd : SUCCESS in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/556bc50c91af4f87843d2c42fda1cc6a : SUCCESS in 13m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/521e066607f44eed9d230ce4d49ec109 : SUCCESS in 40m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bb40d564b2448cbbcc6b1328fe68336 : SUCCESS in 14m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/954d5d532b844064a9f3ae07604955f1 : SUCCESS in 16m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c55f9b96015246e58da5bdb0bfbaebda : SUCCESS in 25m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e73f45f8e0bf46f38d6788bb68ba5e51 : SUCCESS in 8m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f09ab55b17944d099a23ca07e9f2f356 : SUCCESS in 9m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68d75a66fcae4e5d9cec60781a1055e1 : SUCCESS in 8m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e34b5583ced1401e97ca4e30fe54b138 : SUCCESS in 11m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/56b6d1011d13405f983bb1b504490682 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/627cd29effa24d629608eb489d459fe2 : SUCCESS in 1h 00m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff070cd90db546579a7e30d62b416c2c : SUCCESS in 1h 03m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/338308ccc3534147811b44dcfb076ca6 : SUCCESS in 1h 44m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5d71eb881c4c4e499033fa9192a5ebf4 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"037eadc1e3b28f1217a5a866115c320132b39f5d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:09:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ffe64bd95ec40e2ae135a37446bee1204435345a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:22.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fc24569e4febae4c5c277ecc6a537a963f8c0f47","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 03:03:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d98f452cb2271e051de8b133552fd7995e3f0180","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 03:03:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0de4164865f7bb4263964fa64cefe83fb42c6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 04:33:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa63ff089b74920a635b64c622dd4e0 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b22b339a1124524910b1c413f133eb0 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/226e07f947c249af82dc315eb51fff6f : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef273dda0ae7464ea20e0fbfb6a65e43 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/643638f681ba4ee390d641c0be50f2fa : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff61d6efda4b4e4f83f34dc10f7667c4 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdc59a2ddeda459798e64a0a2a104f4a : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71da54ee5bf24b23aa33ecd657d31ca0 : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/470925afe2474e9ab91496b2415ea1c1 : SUCCESS in 8m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/52b66dc3533d4fbba65076176cbda59f : SUCCESS in 4m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/200f8505c450461c9e15350d092939f9 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e9c87ee8f1d450fbd83bfe76a51879f : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b3688913e7943abb2f67b86681f971e : SUCCESS in 25m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1802ad95686b46999c2fa55dbac66fb4 : SUCCESS in 26m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/58ed869c01d046509e89af9ff0d03561 : SUCCESS in 21m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59b686411ad04ae4b8ff994317f739ac : SUCCESS in 12m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e5c9d420fb9948799e58fbf6290dd8e5 : SUCCESS in 14m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/457a97a537c349579e32db9a0665e6a9 : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/24ae171b00a2493e8e55996c79d7050a : SUCCESS in 7m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76d96e449ee34175b4fed1d827bca26c : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7816032fc28f45c4818381fd3e53076b : SUCCESS in 13m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dbf48878bf5a4e3ca9f8fac7683c0c5f : SUCCESS in 8m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74c372ec167f4063acd82a0cc2b21bf2 : SUCCESS in 57m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/acada366bbb04f69b7125ae9fcba856d : SUCCESS in 1h 23m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a899ca65e14a4d0cab37cc9c824cca34 : SUCCESS in 1h 25m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/978e363d15d14600b953e736deab3091 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":2},{"id":"ba2d29842125e425cac49ae2a1b2ac58f3cfc507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 04:34:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5af1f340661cb01d50205f38366398eb72ddafc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:51:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06321ff72dc2470b9fc1a43fbb7b8b4e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb77a024e5b4933884e911da82d0928 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00ca399ac5b4860a2c469b6d0bd4634 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cec5c41c43e543b0b5934300181bdb65 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d407f653b44b4f2a94963792f5c8c89c : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/438f4bad6a18460eaf13eb427d579429 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2ef8111629a844f9a47cb231c7ac1ceb : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f56f03e4817949cba4f584bd24482107 : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c338b6d1ee374ee68caf97ad924deed5 : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6048257a5fd645ccb855ba06aad879a1 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac881c2a2f71422a9d261f27245e6577 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/115fefc2bb1a485ebd76210060e5ec51 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1ccfea04a1bd4c44ad93964e459cff75 : SUCCESS in 26m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3fed45cc76a34ebc8f2a5a9f2714781b : SUCCESS in 13m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1d9a53aaafb64899b15014fec0e7a041 : SUCCESS in 14m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0e49630f37f486eaffc46a1a87528ca : SUCCESS in 20m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/96f7a873d38a4fb8a7958c012477b5f1 : SUCCESS in 4m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/93dd5fd3a2744024a5c36e16143c57fd : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9033a8f96b6f4876b29aed1955014845 : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/21c674c1c6614a9f98e8b357ae9ef4ce : SUCCESS in 10m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c32f6b594da342b2882916195c6ba14c : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23f43bc9b73e4b24844ce434aed78917 : SUCCESS in 55m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d7e7aaca3d84965a315b3c8448ee158 : SUCCESS in 1h 01m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/06461e74b3214b6080bac9943bfbec01 : SUCCESS in 1h 30m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fba42de1f4924a49b9134f61ae24f0b0 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":2},{"id":"8822f546f1ad68845b923e0d7d694a9b089a2ff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:51:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e8a34633fecc09e6a8fd6125fcd362ff720c3069","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:57:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757dfcb2df304d5ebb73cd0ac11e6b88 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e87845e6bc29945e1a4e50f5795af7a90f1ab19c","revisions":{"85f9b174c82e55ed7987513592adcaa9735cc7e0":{"kind":"REWORK","_number":1,"created":"2020-02-05 22:00:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/706130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/706130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/706130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/706130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/706130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/706130/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 22:00:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 22:00:20.000000000","tz":0},"subject":"update constraint for oslo.service to new release 2.0.0","message":"update constraint for oslo.service to new release 2.0.0\n\nChange-Id: I79d0105537898fb82a4cfae38b101797164d53d6\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Icbb73481d01f6841d3e5fa9d4b83ddd3cb372331\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85f9b174c82e55ed7987513592adcaa9735cc7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85f9b174c82e55ed7987513592adcaa9735cc7e0"}]},"branch":"refs/heads/master"},"e87845e6bc29945e1a4e50f5795af7a90f1ab19c":{"kind":"REWORK","_number":2,"created":"2020-02-06 03:03:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/30/706130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/706130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/706130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/706130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/706130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/706130/2"}}},"commit":{"parents":[{"commit":"8a73e436b723fe5a80bb12eba2a7b763435b36e4","subject":"Merge \"update constraint for oslo.db to new release 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a73e436b723fe5a80bb12eba2a7b763435b36e4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 22:00:20.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-06 03:03:16.000000000","tz":-360},"subject":"update constraint for oslo.service to new release 2.0.0","message":"update constraint for oslo.service to new release 2.0.0\n\nChange-Id: I79d0105537898fb82a4cfae38b101797164d53d6\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Icbb73481d01f6841d3e5fa9d4b83ddd3cb372331\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e87845e6bc29945e1a4e50f5795af7a90f1ab19c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e87845e6bc29945e1a4e50f5795af7a90f1ab19c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
