)]}'
{"id":"openstack%2Frequirements~706132","triplet_id":"openstack%2Frequirements~master~Ief6d844c81ae88452aa7e93d8f992c327c991701","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ief6d844c81ae88452aa7e93d8f992c327c991701","subject":"update constraint for oslo.rootwrap to new release 6.0.0","status":"MERGED","created":"2020-02-05 22:02:42.000000000","updated":"2020-02-06 03:00:22.000000000","submitted":"2020-02-06 03:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706132-1580958022963-ed07d331","meta_rev_id":"17a132ab61a22f379e7317222f28b9e6e52c6416","_number":706132,"virtual_id_number":706132,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 03:00:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 01:09:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 01:09:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 01:09:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-06 03:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78e99914d523e8c7c0b6552824ff8e984c8c1ec9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-05 22:02:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85059d0ab934b5be37b02876124495e52a1f2659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 00:00:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f344810d2efb4a27bef1fb00b4b1dae0 : SUCCESS in 7m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6c4b9ed3767642fb986c0ca3b7926534 : FAILURE in 8m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2e2c20ea994fa8aa49d1a696e6906f : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc676b1ad1dc43c2b90483c7b68a50f5 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40c3d4bd05fa49b3958a36c6ba56c3b4 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a4c87494a0140928ae895c3019c8724 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e7d4a1777c34e4ab73cb2831ab9da10 : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a762bdb8236648b5849a812a577237ea : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/827aea511f934df3b5ee569031c4376e : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/776b82e29f2e401ca1e17b7a9f9f9fad : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4d347a3b0ae34f92ad00bb2c78a04595 : SUCCESS in 7m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba92e9f1f8de4d9ebe832620a3588231 : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e14591f8fccc43f7994c8afeeeb64024 : SUCCESS in 8m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b70d0ad875d848228cac3262944a76c3 : SUCCESS in 16m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8a48249a9186444d93c0ba956876c8f2 : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7975c85639a435aa0612147f1acc283 : SUCCESS in 15m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/24ce848c052041b697271387e305fd2f : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17645eff3b644173ab51426288a4597d : SUCCESS in 17m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f4eb7c7b2bac4a01b9f32cb6689c9381 : SUCCESS in 6m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/eff6a70cd2b545f482481e2f01145873 : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e4e0c5349004cb9ae742e184ee9c222 : SUCCESS in 9m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/242fa78b9e0d47bfa2aea1ef17c90921 : SUCCESS in 12m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2261cb79f4d3492b8d58f25283b044fd : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e919f910c834540a4361722f04c96ae : SUCCESS in 1h 04m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/adfd8a1c3baf4449ba8bacea488d7e37 : SUCCESS in 1h 06m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0ebd68c1aa3243b0ae63bb94cc5e19b3 : SUCCESS in 1h 41m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1243acb542f447d5bf4f1cfd37baf4a8 : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2c2225dc76beab45715bac457db869c343bd5bf0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 01:09:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a3064b699a3c1ae8cd0dac58fe3544a0b8b5742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:09:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f22c90dd17103f86b04326cb44ce079152fc4f3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d22322005fb41ec517272277d34cff4017de211e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 03:00:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5df629b22734cceb6957b374d92d2e1 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5128769192f6444eb0b468cd7582dd35 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3edc13a3e7894d34b6b0b57d7764af8b : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bd20fe5b612442bb9eaf01db9261de8 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43d3acba2f174286bdcc4e35ccba2a8a : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f04704fd3ac54a9084e42b05dfeb60f4 : SUCCESS in 8m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/545cd6f171ec4d48a7127efd68846e28 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85e78f6c0cd14d3ab6a402600ef6475c : SUCCESS in 12m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e6929f5bd2b5463b89d902bcb01e8125 : SUCCESS in 7m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/832cca71817a4b14a7213db389cd1f08 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17fa8eeddda543c6a8eaa7fc46c31e34 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/079270caa22d4f4ca8c7f7e591e1f4d7 : SUCCESS in 17m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/060b5fb116f44d3fa93babb6fef2d82b : SUCCESS in 28m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e64103f2003c4a379dbe655746863480 : SUCCESS in 13m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/68d8847c36e944d8b10730d0c1a79fe9 : SUCCESS in 16m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d57930647fed46fc80d05eef4353dbcf : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2da30010377e44c78f99d995dd0823ab : SUCCESS in 8m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5400f774223449aaad8d152abe3e7393 : SUCCESS in 9m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6cf668a3f8a440a0aded56b4109b5dc8 : SUCCESS in 10m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8832ed0188a748f99aa20105d3060b5e : SUCCESS in 14m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/41a16442e0ad43b784eb76782b4d4193 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a167749659204c4383403281e7ccdc0b : SUCCESS in 56m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/623f0495ecb94ad68948f63d6516711a : SUCCESS in 1h 05m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2458da488abd4010921231304b215d4c : SUCCESS in 1h 32m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a6b92eda90a74bed911ec1c320a19bf3 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fa05c387717c6a037bae722d8077d9b5f57ce69","revisions":{"8fa05c387717c6a037bae722d8077d9b5f57ce69":{"kind":"REWORK","_number":1,"created":"2020-02-05 22:02:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/706132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/706132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/706132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/706132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/706132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/706132/1"}}},"commit":{"parents":[{"commit":"0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5","subject":"update constraint for oslo.concurrency to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c5ace0ec8ac670b6cbeb64fb8577d11adc90bc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 22:02:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-05 22:02:40.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 6.0.0","message":"update constraint for oslo.rootwrap to new release 6.0.0\n\nChange-Id: Ief6d844c81ae88452aa7e93d8f992c327c991701\nmeta:version: 6.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Id879b66025049e7fa6a7e503e0007aabefd27b13\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fa05c387717c6a037bae722d8077d9b5f57ce69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fa05c387717c6a037bae722d8077d9b5f57ce69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
