)]}'
{"id":"openstack%2Frequirements~706167","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iad2615a49eef52722207e7ba9b7a1f199c40cfbd","project":"openstack/requirements","branch":"stable/train","topic":"py3-libvirt-python-train","hashtags":[],"change_id":"Iad2615a49eef52722207e7ba9b7a1f199c40cfbd","subject":"cap libvirt-python for py27","status":"MERGED","created":"2020-02-06 02:17:25.000000000","updated":"2020-02-06 06:51:13.000000000","submitted":"2020-02-06 06:51:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706167-1580971873056-b4850ae3","meta_rev_id":"e0098f89abe626c62ef2d00b827f6632cf68e8f2","_number":706167,"virtual_id_number":706167,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 06:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-06 05:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-06 05:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 06:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cec519e6cd398e0fa8b25143817763eeaf7fc32","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 02:17:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b58b0261a981d28c37a6f54976c51f31f245696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 04:00:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5670a4f52ab84f8a8cfef9e1ddcdacb6 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f65c1e17fa4de1a6ef32ab5e78a3ce : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/215bd701e8cb4d0bb1aa32bc8d5e3b54 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c97206485c0a47efbf144cfe8d7f1bc2 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa6e304d006646859ab8a95dda0a6943 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19af700fc85e4731a173eda2a13d71a6 : SUCCESS in 4m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b534d0564f924f59815046c44f019d71 : SUCCESS in 1h 07m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/498c90b0488948359887773ce3d9863c : SUCCESS in 1h 41m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af0d5a10d9f8468f81df7f142ff0cb33 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"00a0c787506c4134bda9a7fa6626c01e7d58c633","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-06 05:09:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6ab5df04fdd2bf03b4af006a19d31d6048b0beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 05:10:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e3769ba051b88f2d246f1b153a294318c5d7f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:51:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631204ee445142ada8e4c4a8d29443c8 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e0ad948a45e48169114a4c55ad720f6 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c0b69b246e04ebfa4f95663fcb28a01 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0064fd795894463a92199ff3eac46cd3 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/736286579829467488c9d57d00c46dec : SUCCESS in 3m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b527afa6f4e14558b72d362b0769976a : SUCCESS in 1h 13m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/36151e0ca8cd499f8b6c9c2fc3b65148 : SUCCESS in 1h 31m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e17ca66d2de14227b881b8084b60bdc8 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"65253c0b58e8b125ed72f798c01865bedf065b24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:51:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f91300104bc0802fa04cee5a62f5c04ec77ae41c","revisions":{"f91300104bc0802fa04cee5a62f5c04ec77ae41c":{"kind":"REWORK","_number":1,"created":"2020-02-06 02:17:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/67/706167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/706167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/706167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/706167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/706167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/706167/1"}}},"commit":{"parents":[{"commit":"3384012bd4b906e8ef8e2365b52768f7b12a636e","subject":"Merge \"Bump diskimage-builder to 2.30.0 in UC\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3384012bd4b906e8ef8e2365b52768f7b12a636e"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-06 02:10:26.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-06 02:17:17.000000000","tz":-360},"subject":"cap libvirt-python for py27","message":"cap libvirt-python for py27\n\nChange-Id: Iad2615a49eef52722207e7ba9b7a1f199c40cfbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91300104bc0802fa04cee5a62f5c04ec77ae41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91300104bc0802fa04cee5a62f5c04ec77ae41c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
