)]}'
{"id":"openstack%2Frequirements~706497","triplet_id":"openstack%2Frequirements~master~I86e27629007cd08a8e21f79815519174b6630888","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I86e27629007cd08a8e21f79815519174b6630888","subject":"update constraint for castellan to new release 2.0.0","status":"MERGED","created":"2020-02-07 13:21:51.000000000","updated":"2020-02-07 20:15:19.000000000","submitted":"2020-02-07 20:13:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706497-1581106384831-5f1f21a0","meta_rev_id":"aabd556039f49de34c59589f69ecb602587a4c85","_number":706497,"virtual_id_number":706497,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 20:13:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-07 15:05:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-07 18:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-07 18:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 15:05:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-07 18:43:09.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2020-02-07 20:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd20ccea82f289e467981f1d6d199aad00631745","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-07 13:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6968a33f2c064bd4dd24d95cb5dcc184ecbfb821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 14:59:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1830196583544b37abff8c57c02d9d73 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a416f8d84c945fab895d16f2acc9dc4 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23f74c93c62a4b619efed41053216c5a : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac4b5346210640c4a2c540bd3851b700 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfc6a4e0caa145589232292a5928fd1a : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a580315b24f421dba8bc72278db34aa : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4116eb68c1cc4cafbb947f11c7902d07 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4fc625e39b0b48aa9beb8167816bbf8d : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/daa22c069a524e0292485fa01930cb05 : SUCCESS in 11m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57af5b521a354b49a637aef2cbcd1ea6 : SUCCESS in 4m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/59f36363d085452a8b3b84211d8010ef : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11179aefb56445348ed4ecc1f5b8371a : SUCCESS in 6m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d13248948b746f58adaf17c93199465 : SUCCESS in 12m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ace27d2c90f46528f795315247d2233 : SUCCESS in 27m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df7e27658b084eb09eab63f7369ad171 : SUCCESS in 11m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bf550e7aa7dc45fc8fa7a3c990ed8b05 : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80302a5059484976bd725483ce03be2a : SUCCESS in 17m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/73861b924e504ba494448747fb14e6a7 : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a73de7c1adfe4285b62cbb64c68f641f : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd60afcd84ad460d8e4d769447c16597 : SUCCESS in 7m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cfbd57f926ec4798ba5cec37b598507d : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c39976c915c44680ab2ba9c1c07e62a4 : SUCCESS in 9m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f317c4b7a08349a9a01f59e315299227 : SUCCESS in 59m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99aa5072e39142e9abf3f1d7c6506e65 : SUCCESS in 1h 14m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7c4047e555094eb9a8150517d87096f9 : SUCCESS in 1h 31m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c3f9ef89bf8b4852a2f4e927906d2a29 : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1bc1d88177e4edfd5493590d2a57d31f3bd04424","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-07 15:05:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b7da5de25d0b07808a7ded2ed0b03232b1926d2","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-02-07 18:43:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a5597e3a8680548efd1a7aa1769ef9fb0c2c4203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 18:43:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65f7e9062274f8941673c6dd445fc8f80d5e236b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:13:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b3b8b64604e31e9abf3485666c798f1ef4064e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:13:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd9a8eaabee4ea7ac8a2c2c8b901d48 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926454d27cc64e5b893f2d75fc1787ac : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74515eca3398420baadaf19916bd333f : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a185bca06394be299bf24abe13fdddf : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a398b8f12b1b4bacb7a8f192980bd3c4 : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee7ce12ebada4259956081e842b5160f : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8221841a00f14e80a9d8965177cdea0d : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13af945a590645ec9e75f018bbe54e00 : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/253168ae71c249b1a273636db622d649 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76621a75a98f4461b6fecfc486e19a13 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0fedf8b8a1d459db10839dd26bf1a1b : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4da38abec0f94ca989aebfb196a219b3 : SUCCESS in 13m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/23d8e676e27d4b4888ea47080b714dfb : SUCCESS in 28m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8afd7c732426438b8f915e4304019df3 : SUCCESS in 18m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7a653e70c855419d88c545e3b4aa06ec : SUCCESS in 16m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbb4370b2467474f9e6b54554ee1abab : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b3ead082f8364d6ab82830f69c18a630 : SUCCESS in 6m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b4f5ad9b20a741ae92a6c8cfdb07ff41 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b144085038094ff0a071e6968710bee6 : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/25bac9e1dd2242bc8b1c3bd31bf961ec : SUCCESS in 11m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bde663138a204c99863b1a53d486f23e : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c73eddb90af4df2a6249f75a8738f54 : SUCCESS in 53m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14695a2c9c2e4e7b8f19c65061ad7676 : SUCCESS in 1h 07m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7135f07dd1884a7d8c6284e93dd0713f : SUCCESS in 1h 20m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f1ac97524ee4194a31df684193106c1 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"53134d6ab22ee69111d16f1c9aaba748e031341f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:15:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7713885805f4d7a88a38a9e88d521b5 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f8c63aec0e03917756ee3bc9a9028ad2c812c4","revisions":{"58f8c63aec0e03917756ee3bc9a9028ad2c812c4":{"kind":"REWORK","_number":1,"created":"2020-02-07 13:21:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/706497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/706497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/706497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/706497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/706497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/706497/1"}}},"commit":{"parents":[{"commit":"6851e4620aafddc5e2917d48be039db92aa79ba9","subject":"Merge \"update constraint for reno to new release 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6851e4620aafddc5e2917d48be039db92aa79ba9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 13:21:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 13:21:47.000000000","tz":0},"subject":"update constraint for castellan to new release 2.0.0","message":"update constraint for castellan to new release 2.0.0\n\nChange-Id: I86e27629007cd08a8e21f79815519174b6630888\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ic8cc6d6523089280a0652982161d535184c4e84f\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58f8c63aec0e03917756ee3bc9a9028ad2c812c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58f8c63aec0e03917756ee3bc9a9028ad2c812c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
