)]}'
{"id":"openstack%2Frequirements~706569","triplet_id":"openstack%2Frequirements~stable%2Frocky~I751572304d0ca73c147852738880ff6ee5b3b9ef","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I751572304d0ca73c147852738880ff6ee5b3b9ef","subject":"update constraint for ironic-lib to new release 2.14.3","status":"MERGED","created":"2020-02-07 16:51:57.000000000","updated":"2020-02-11 06:31:31.000000000","submitted":"2020-02-11 06:31:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706569-1581402691982-2176cf01","meta_rev_id":"7a2eb8f01bfd035c7dac7b6946093330b5de6fb9","_number":706569,"virtual_id_number":706569,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-11 06:31:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-11 00:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-11 01:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-11 01:12:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 00:39:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-11 01:12:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-11 06:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98d7789cc05353ec380a9822c0ced476ffe4c36f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-07 16:51:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d3d9400f718bcc214f93124bed686d4a9fb0da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 18:10:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c97384493e094b0b8cbebb3b2ad83ea4 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269d27805f1b4b11bd80a1d316f43b00 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2d79209cad54c4a8281c592d21f517a : SUCCESS in 6m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9feec1d6918543aaa59dd5e06d86ac33 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01e2d437582b401c82eb97a9e82988f1 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c7328299fd54726bc442536c1047750 : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0e390dc94b06447da76eead3af16b4e0 : SUCCESS in 9m 48s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a9fabcbaee4340e8b73e1734e5df21b7 : SUCCESS in 13m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1df35e35007d4537aa0d00bc8458df77 : FAILURE in 7m 53s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/82de4ad6e1cc462a9f00b547be4a7abb : SUCCESS in 10m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/dd4fd182a74148a698b01fb7d8e67011 : SUCCESS in 15m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/75f81c82853f419aaec38728415d98b0 : SUCCESS in 6m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c892dd65f77b4c2bbfa333d18e0d1335 : SUCCESS in 9m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d90ce0b5614c4932b8252b4c765704aa : SUCCESS in 11m 33s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/25a65b9cbbca49bfb74bbe2621873d44 : SUCCESS in 10m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fc91f6512a9b4b8e91f82135a0caa776 : SUCCESS in 33m 52s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/867523dabe334ab389f8c74ffcc9f0c6 : SUCCESS in 18m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5389430f7fa747aa87f0d3c5be0748ce : SUCCESS in 23m 27s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/be724a4a025d40b2a9828376e56e3434 : SUCCESS in 28m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ccbc9e62d5414cf0ad351fe8831b293c : SUCCESS in 26m 06s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d097bc675c8d42489bb3d40e50f5e6a6 : SUCCESS in 18m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/470ef407c832416d85a8056da3fcafb4 : SUCCESS in 25m 36s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/fe47ddf2445b47468a7ae8a84463b6b9 : SUCCESS in 8m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f58e5c7de13b43e08c4d622459b62990 : SUCCESS in 8m 35s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2e70808f27624169a254bb3abcb6b678 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7ddfa4697754262b43559ba40a2a9ea : SUCCESS in 4m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e747537f5764fe3859858616d7dbfd1 : FAILURE in 23m 56s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ffef9875e0484cf2aa8db18b03b2092e : FAILURE in 20m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54c3522bc9764db792a80f820d2776f1 : FAILURE in 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"533dac61493fa9ef9954d6feef1f4fcbac3369bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-08 01:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"000d0adeafcde9e930f651cf473abce93c30b3b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 02:38:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ce966317d83e40eb9e2be2be9d00075a : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28e829b14caf46a4ab1e7ddb28c1ffe5 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e8134e5b4964b68a5d06b60233e4f2e : SUCCESS in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c03ffb885471449eb28be8339c2b696b : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26fb7d36d28d4364a7c59c4be8cca436 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3bf06405f7c486090146a14e5170847 : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/70714414da9e4825ae586c81f5e7642b : SUCCESS in 8m 04s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f81479eab34b4aaf8176c9e42b203039 : SUCCESS in 8m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fc26ca4e5f5425488b94403cf8b7c49 : FAILURE in 7m 33s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/561ae017f7b84d51a92ae1e7dca35637 : SUCCESS in 9m 19s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/415296c9cd5a46dea262a15032942ae5 : SUCCESS in 10m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/34cddfc501df48439398ee6fefd1bee3 : SUCCESS in 6m 00s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3ab865a52d8a46b6b812d88c01efae81 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1f85ed2b051e41c48edbb9cc3b58e43f : SUCCESS in 10m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/4af43b3a90f84816946aec06d192538b : SUCCESS in 10m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e0331b3813644426b8adae270cbcd0be : SUCCESS in 13m 13s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9923fa40fcc94e7f8d1aff4772412b4c : SUCCESS in 12m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b5ac0a3cf7724a48b22b90a21c70183d : SUCCESS in 29m 34s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/398de6cbb17d4537bd3f497e6b999545 : SUCCESS in 23m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/be045fe8c1f04bd396a19e3823027b93 : SUCCESS in 12m 11s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7496838784d74df0bde9d724e4d8fdeb : SUCCESS in 16m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77ca2a8fe81f4c1990945b6bc3ebb7ab : SUCCESS in 16m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/342d7ff622b64ebabe24a9f5771de23e : SUCCESS in 5m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/92a18bf25a97447a8e8f889b112d8ac7 : SUCCESS in 8m 21s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/800611bf35f9484ba7f220dc3d5844a9 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d26f560e5516471b846b8197cb38fbc8 : SUCCESS in 3m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd5db25576d042d3bf53554bb9c0e14b : FAILURE in 19m 31s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/5f4966f2c24840e78361b7ce4ae365da : FAILURE in 22m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8472a56112bc4a1daef12a21e9c3bcd8 : FAILURE in 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9b850364517d1dd13589a4e37f452bf17851a5ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 16:23:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b6e5114edc4b3571bc747885eced4ad784d5b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:54:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d8e044806a5e4d2a9b7cb69802d0086c : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2fc132ae784a7092bfb99a34130c74 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43c14acdb8d74c8d886fcec7b859d32a : SUCCESS in 6m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c28a282970e45169c88456183aab466 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8b1888b16d648adaa68c1bf89328577 : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4dc5f1985bc4ed4963edf125297141e : SUCCESS in 4m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/17d0cd6367b64e9490cae7ad0e2ba4da : SUCCESS in 12m 06s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d9955376bff445b8840e880bea5dba69 : SUCCESS in 11m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6fca909eb5c04efba703a82253e99fd1 : FAILURE in 7m 32s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/00d3eb0c9db64e4fa7195a29c693e2f7 : SUCCESS in 10m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0c4e2b8944b049e8aafc286d8fbd4173 : SUCCESS in 11m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5a4e45d98b7d48189f8a86a0f2577b24 : SUCCESS in 7m 20s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f3088155e8e24a4c8a7b81bf8571d04e : SUCCESS in 6m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6ba7a46e7cf74ec9b22a87d9e1cea1b0 : SUCCESS in 13m 05s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/188e3b6abb734410bbec79dece603442 : SUCCESS in 12m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c0319bc912a244b6a06f2d1683181d6e : SUCCESS in 23m 57s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/458413396aac4880b023d1c8ecc52d8b : SUCCESS in 23m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/42d9f27c2a25452b8cce1a3c66ae846f : SUCCESS in 25m 39s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ecbb956c771247f7be563b509218fec8 : SUCCESS in 30m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d26ce077f53846b8b74eb1017c648757 : SUCCESS in 18m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/49986dca8f6843b3a593cb5cb644b375 : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ba027991ad04719861de87ac37d5bed : SUCCESS in 20m 18s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/330d5ed806e04113928fbc9301f13537 : SUCCESS in 7m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ba0f29e8cf314010a73a2a761f9ac4e2 : SUCCESS in 9m 29s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/75815a13f09a4e19bb42c5e5c23f6cd4 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36d0cdffb73a4881926bb9ad12080ef8 : SUCCESS in 5m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50bf710b89e34622ab66919992a3955c : SUCCESS in 1h 09m 45s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1c7fe5ad027448a7803d05d860a11d8e : SUCCESS in 1h 23m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ddaddd7df2dc4b618998bbdfa3842791 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6e25d248920fa3694cbffcb2a1966297bfb98dcc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 00:39:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"396d07ddbfd82d7bb0ea2c0932a703ec08418530","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 01:12:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcc29d7ed6f6fb9fbd7353ea159bb33b7c019b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 01:12:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcd6697446b77464f39bf20267bc8a8282ee3fa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 06:31:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee824f8acd884954437531be34e3c2df5986d0f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 06:31:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3cae4af81545d3a51d154011749d83 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/354df4f7b855409c97670aeb75f2aac3 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c419a148c16d48b6b976a1e51867bf8d : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c21ffb58b0946cd8d646481130b2487 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1ae54645e1d4a0586885f0aca6b0a9f : SUCCESS in 5m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d228007ca87c4b2a851f64478342d2fd : SUCCESS in 7m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3d2c487e8b5749628a0dafba153f6def : SUCCESS in 7m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9c60aa5ffd25494382815a69943b1a05 : SUCCESS in 8m 44s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/90c33dbb681b4ed2bab99308c05e05e9 : SUCCESS in 13m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d5b4cffb0f8c4ba0a3fa7ed6989d319d : SUCCESS in 5m 11s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/01a808e00f2d4d8a9791959715b46976 : SUCCESS in 5m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2e75ece2336647ccb72184de501e0ef7 : SUCCESS in 11m 30s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a457e8147dfb4d218cb5081ef060a46d : SUCCESS in 9m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7cecabb7ad3a41588541777f7c707e3e : SUCCESS in 11m 51s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/0dec202e28aa40bc9f9ac051bbf9c389 : SUCCESS in 17m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bd5776ae4f4b4a7aa9ad835df54c9c7c : SUCCESS in 25m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/109ebe7b0fcd4c39afad440fc9dbc3ef : SUCCESS in 24m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/068a7887746d4bfe9b2471074c6d8f2b : SUCCESS in 13m 35s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1071334d296d4fef98ec0e763efbd9f3 : SUCCESS in 19m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e6aa0f1e3b74df386c0d81a4cdee8d0 : SUCCESS in 19m 13s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/258377ee274d4fd99ac3684ed21523e4 : SUCCESS in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bdc77a5413084e25b633b4c6dbf39007 : SUCCESS in 9m 21s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/93b31c239a02492d9259ccc3981779bb : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c488abe1b794df3835ba7a567772d37 : SUCCESS in 3m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b8ee2a1d95e439aa6e0988ec2bda81a : SUCCESS in 56m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/329395c4adae44918c42d650a110aeb1 : SUCCESS in 1h 23m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/17d81f7bdff941f9918b5425603d25ca : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b06bb9ed9b93df7a91c1603428a22cd163f92493","revisions":{"b06bb9ed9b93df7a91c1603428a22cd163f92493":{"kind":"REWORK","_number":1,"created":"2020-02-07 16:51:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/706569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/706569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/706569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/706569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/706569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/706569/1"}}},"commit":{"parents":[{"commit":"3898e19b8ecf96cb729977ab1b2820345ab532bb","subject":"Merge \"demarcate python-memcached requirements\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3898e19b8ecf96cb729977ab1b2820345ab532bb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 16:51:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 16:51:56.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.14.3","message":"update constraint for ironic-lib to new release 2.14.3\n\nChange-Id: I751572304d0ca73c147852738880ff6ee5b3b9ef\nmeta:version: 2.14.3\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: I6d9bf2c38519d69ac0798769af78e3178f10725f\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b06bb9ed9b93df7a91c1603428a22cd163f92493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b06bb9ed9b93df7a91c1603428a22cd163f92493"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
