)]}'
{"id":"openstack%2Frequirements~706575","triplet_id":"openstack%2Frequirements~stable%2Frocky~Id278308c5bb3d83694bc8fedc05fae994219fc4f","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Id278308c5bb3d83694bc8fedc05fae994219fc4f","subject":"update constraint for sushy to new release 1.6.1","status":"MERGED","created":"2020-02-07 16:56:11.000000000","updated":"2020-02-14 01:29:04.000000000","submitted":"2020-02-14 01:29:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706575-1581643744726-25e6796c","meta_rev_id":"cd172758a3415a4dd5bce96890820eb6f8701ba4","_number":706575,"virtual_id_number":706575,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 01:29:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-11 05:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-12 19:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 19:09:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-14 01:29:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b44ba21625a737c16b2b0bc3b0ce68711b351283","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-07 16:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02c8ab11b4503be2f1233f6915a6cb1e19539f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 18:34:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2aae65113a464621bf0f339f0fa54cec : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b33f5bd93f744689c060d849f5fb074 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d6cf9e195e441d5a74b72ff16ec56db : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d6bd42f6de240dcb5bb4301803ee6cb : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8c58aa5935d41cfb2e40591a6533ce5 : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1899c2daaa594d07b57dc1b6c7edbf4e : SUCCESS in 6m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4f9030a8a9754d088b806887784174cc : SUCCESS in 9m 34s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f6921f9175e84bbd991d5bcb629aee96 : SUCCESS in 10m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e6ee3aa07e74385b53aa48ae08e5dbf : FAILURE in 9m 13s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b52b428ad0714671afc44e2e15312167 : SUCCESS in 12m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/123a620fb6ab4fd59cb68a0917af0983 : SUCCESS in 11m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7cbf36e25d0849edb6cbe25af73b8c06 : SUCCESS in 6m 51s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/46f2731097534880bb5adb78e085a492 : SUCCESS in 7m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fce155519a6c4be3b02522cde1a626b9 : SUCCESS in 13m 05s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/00852f75df21429582f67e0982746de9 : SUCCESS in 11m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ba58c7bef1974593a8b2470f0573002f : SUCCESS in 18m 42s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/929cb790bc424bb8905f5d4406ba1d31 : SUCCESS in 14m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8871254425f742538c3402a8d0c0180c : SUCCESS in 25m 54s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/648a771ddb624b0ba5dba4d8326b8903 : SUCCESS in 33m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0cc1356480d44000ab5ffa643fd8c860 : SUCCESS in 21m 25s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9cd59712eccd4fce8f044b75b5abd7fe : SUCCESS in 22m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1fa690552734d5f8df42267669340ec : FAILURE in 21m 07s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/7891cd6b742d4a27807ae2a3ce4f510c : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/396f0a62029741459b016b01374235dd : SUCCESS in 8m 14s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4ea0f86eeb84445d9fc6d95e8e761ea3 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eee583c0d9e2431cb6995bf27f5024c5 : SUCCESS in 7m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/967b2a7dfe0242589b8d468ddbeaf74e : FAILURE in 23m 53s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8010368c265e4734b469c6a95e2f3539 : FAILURE in 18m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1f7068c3b90b497c9368eafb550f5b4b : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"64aa9d6690939f92032c686a381555fe2d3cd6ee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-08 01:39:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77528ea573ceff19ba5f4b7f06622ed6ed0b0e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 02:13:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/af01d5c5ebd54782988f8293a6a45033 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7b2f06369448f48e2160574ddcbe05 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ace370aebeab415fb61b1411566e00de : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b8e5010c709241adb3ad893d1409834b : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a4faeb0541c47cf8ee04851b5ab1f5c : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/540599ac0f0f4696b690c7735b7b5227 : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e75d0c55ab7447cd9bfd18349923011a : SUCCESS in 7m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3449497f2a7d4f56af3bb622b8454325 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0df6141f88b46f5987576a8e7d5f9ca : FAILURE in 6m 28s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8bb27ad5d2e14adb9b8a63b314ef2692 : SUCCESS in 10m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c20e9f9d913b4c7393bfcf2568ee93d4 : SUCCESS in 9m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa25df2a2d9f4e9ca8040a9ae67604cd : SUCCESS in 5m 36s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9128ac96b5214680b2b036c5c4a0ca32 : SUCCESS in 6m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/29f89edb3cf44a00b200be59f1770501 : SUCCESS in 11m 53s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7d8009e0b7d14075920007ac544e1777 : SUCCESS in 10m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/498854331e634332bc3feeb006f7bf4b : SUCCESS in 13m 16s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a844b81ba9484ac493ab412e6591c69d : SUCCESS in 16m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/59389154c2a5420ab5b59e1eb2f8a263 : SUCCESS in 23m 32s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/fe1bbb93af3b488a9aaeccc1457ef75b : SUCCESS in 23m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/761ee228854944b58de4164340aa11af : SUCCESS in 12m 44s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0dd30f94bcd14a6b9621b8b5f1a24cf4 : SUCCESS in 16m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5925ac69d2ac432f8989cf9c287c2bfa : SUCCESS in 20m 10s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b0c6c0a02e6147a1ac527724ca03fbd2 : SUCCESS in 4m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/815b97d4d65b45b7aab150eb958be5ad : SUCCESS in 8m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ba0f159aed484e09874c619174121187 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7ffd2fdf6cc48609596e07b7cdd0df8 : SUCCESS in 3m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/49f05ddddb554271b4cd348fbbc779fc : FAILURE in 23m 51s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/cd830db02df142a18a911ea932d95ff4 : FAILURE in 20m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c1f9fc55e44466493ed823c7c35d4d8 : FAILURE in 21m 28s","accounts_in_message":[],"_revision_number":1},{"id":"14d2c02de7c92bbc29ae4fad96bdbab7b9b164c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 00:40:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"409a1ab87fbbcb83b1de6f5bf88a9e78a421725d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 03:45:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a4ab882373d943708256887bbbbf370a : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f2c5a0d58948bba44f4bcb2d83778a : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e51d584e671468e9b698b1fa66125a3 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c7edbd3a5da4171829bb858e4873db7 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18a384be9f844bbda652e54a1115467f : SUCCESS in 7m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86fa0b99c06c436db73adab361a7a464 : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e09a0e48a2424da497a8dfeb58f5c58d : SUCCESS in 8m 54s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e118f4cd181d4b539a521e98f7a9be26 : SUCCESS in 8m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/664cb4cf7eba4348a094853371e9bee0 : FAILURE in 8m 19s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c7d1d427ffb344d99531c2b345fd6758 : SUCCESS in 8m 59s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/53dfc819ab9d411dab84c1669202494e : SUCCESS in 11m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/45726ee8ba474329a00e61e414b40974 : SUCCESS in 5m 53s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2db8bd1cd111445286e1b28de77577b3 : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/10169374c7474f94b92ec59966766632 : SUCCESS in 11m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a88fa180d36b4e0798f0474dbe1e3c09 : SUCCESS in 12m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e46466676ff443748f2d9c3a438cdb95 : SUCCESS in 17m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5c00e474ac4f4e4890a0f093529a9ef4 : SUCCESS in 17m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b2a2f12d1c64461cbda27f1f1eacf215 : SUCCESS in 26m 46s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5cf45faf82f84b6aad744f397f620fd8 : SUCCESS in 24m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/114a2454c4e24754a88e1e18e32f305f : SUCCESS in 15m 47s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/66de93fa71d94c58aeefda3b976d95d9 : SUCCESS in 18m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b4118442ed34c8fbc595c18e2ad2e43 : SUCCESS in 18m 41s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f3e7659701bb42a1a499cb0155683011 : SUCCESS in 7m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3d248c21e0584e62a378fb3c0c4d015c : SUCCESS in 9m 27s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/5b9e07140afd4531933b79703270e295 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7456e2ebfcb547adae3174a0950bc4d3 : SUCCESS in 5m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/707739dd742f4a8eba01d42295461abc : SUCCESS in 1h 13m 52s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/80a4f015ee564332ac11fe7b2fda034a : SUCCESS in 1h 30m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/29f468917dce414d8e44183642b97b53 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6392d954fa8b9b70ce2ee22ee8c007c9c30be1f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 05:25:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ec09b2322ed1e39351ac026e88067c687cca68b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 19:09:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f11aafcbe4e92f50560cc02badb2d97f505ad387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:10:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ef362ca44766029da054720ba1502f41667aee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 21:29:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1cfb37707742d49ae54de465936668 : SUCCESS in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d033fca442c4d15af8da6a5d3429a7a : SUCCESS in 7m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0832f7e8b50b4154b6ddff1c5917126f : SUCCESS in 8m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a9be6a9ed434d80acd6ca842db70667 : SUCCESS in 8m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/767e412706ba4b88a7d8c346a17d96ff : SUCCESS in 6m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/94720f3f5f4840d88397c42ed56800ff : SUCCESS in 13m 10s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/2a2e90d42a14434588e6807b7a8af49d : SUCCESS in 11m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/91982d04c98146339a61582ff3a3678d : SUCCESS in 11m 59s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c24f22eb024c430f845043f483493f68 : SUCCESS in 13m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c54b17815901471d8582af4944341698 : SUCCESS in 9m 48s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4d44dba0246c440fa6f517badcaeb011 : SUCCESS in 8m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d407b0bb79c9483c8899465890bc6751 : SUCCESS in 14m 07s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/4badb6a36b154b50857525f892e68ae5 : SUCCESS in 13m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ea189934e5b74cc8b540d08e5a679583 : SUCCESS in 15m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a45653f5c79c4f9394e234f92f2b66ed : SUCCESS in 15m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4ee7922cb33e4ecb8472aea44efbc164 : SUCCESS in 26m 05s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/332321b680624c80849493570184029e : SUCCESS in 26m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e47302219404413f8c8cc287f9a5f30f : SUCCESS in 19m 23s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2d9717f2aa39474ba239c09ba2f2f28e : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b3f894fc17e458590a833be10fb644f : SUCCESS in 19m 04s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/564e2fd5f81d41b88b8b69dcaf494280 : SUCCESS in 8m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ee7ff4287e2645178b1920bfd9b2939c : SUCCESS in 11m 14s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8a451be5030b4b4a84a610f974398ef8 : SUCCESS in 9m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecb635ca32cb48e5b06a748a5c64ccd2 : SUCCESS in 6m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3d86db37a5f4c258f4d5ad8d81eaa69 : FAILURE in 12m 57s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f3b08dde8320404ea7976db97968d37c : FAILURE in 8m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/484a96c22265444cbf50409b2fb21dfd : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a144310eb6deb30f664874e63d38852f688057ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 21:33:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"759b310f6e4352dbc094004623af6d98019e32f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 22:16:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f97d88203aec4178bab4bb152302564b : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79fafa5ba1942c09cb7f7e61948bb0c : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ea604d34732409a8aeaccf73455c4be : SUCCESS in 8m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3acb8fa87f1e4a2fbd75915d6a34f8ec : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d04d0e9997174f579a312291ddf686f1 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/829e3267df69418c8e6ddcfaba0755ee : SUCCESS in 7m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5755cb706251414dbeb75c431f2bbdf3 : SUCCESS in 11m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8162a322bd054c239237856c13bc650d : SUCCESS in 11m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/486ee86e481e4b0497069efddafc25e2 : FAILURE in 9m 42s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/df63637673f448329e68ac836a5cbc03 : SUCCESS in 12m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8487dc8b81df4ef7b3dcd0c5d5c2cefe : SUCCESS in 12m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f354ed6547e8422e83f785c8e8be4bf6 : SUCCESS in 7m 10s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/50373b97977b4cc1a62a135c6d78e113 : SUCCESS in 7m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cba8ececee5a443fb76527c82eb928db : SUCCESS in 12m 52s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e759db7bb7b442d29017040969254a55 : SUCCESS in 11m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5f300de26a744f1aaecffee4fabb4dec : SUCCESS in 20m 44s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d7ba5bbe1cd644a7bb77731c11a9c4ad : SUCCESS in 23m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e4aa9e9ef7b04989af380abb20436ea6 : SUCCESS in 25m 19s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b5d852c1fb894c7b89d7bbd67c36e278 : SUCCESS in 30m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/49ba502c4d8b4980a77e454ba43eb094 : SUCCESS in 16m 37s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/a0252b2af0f140069afccfd1c64d2b24 : SUCCESS in 17m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59c159b4d1c74b02a28a631b93ab1fbd : SUCCESS in 22m 56s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/4cc2f4f52e054a9fab803e4b6f7946ec : SUCCESS in 9m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d58828ba658a461d8e279db527bca7b1 : SUCCESS in 11m 02s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b3c9041c60224a8c9368ceae571d658b : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36851f8f052b42148989404fc486edbf : SUCCESS in 6m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/63716adb6b884491adba1b5bd68dff63 : FAILURE in 10m 13s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7b88d1024fa24ff8afa3290ab91e0c70 : FAILURE in 8m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b050d75ae3df403698ce77e10ae60c37 : FAILURE in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d88a9655660ddbd51ec35cac1b74970fb9be4907","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 16:02:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85a6a96ea31de70127b9a5967da1a1941466d57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:45:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ed5659c5dec6491b841bba0c456624b9 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022db831b55c4f8a8c8e5c71d9afd3c3 : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d19affaee6c1464dbd1be16b9f1569cf : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb3e4f045ca349cc9b53c34f2e86aa08 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67d22089d396484eb37d89bff3345c00 : SUCCESS in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53b7d0f17c204ec2be49a285d5977ca9 : SUCCESS in 5m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/abca88c68b994c499834ff7aece8601b : SUCCESS in 10m 58s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/76b9f638d2a74baa9059d3835e8b15e0 : SUCCESS in 9m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c275c17e69064cf493baa9959e3157ce : FAILURE in 8m 50s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1f153d2285a54abcb18f57fc4e53247b : SUCCESS in 12m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/5ee74c8116e64ee6a3a3434772616359 : SUCCESS in 12m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/946fad72a2114e2ca79cb0d606e5e767 : SUCCESS in 7m 01s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/638eb0eda1a347c897e017977c7b086e : SUCCESS in 8m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3ea5a15e93d04f2187f02c95fa0c24bf : SUCCESS in 12m 11s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f88078eb024b45c3a208287113eb812a : SUCCESS in 12m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b3bf9e54ebd443e5a282209521a4f5c5 : SUCCESS in 15m 21s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c309a291b9b746068e106938d2568d27 : SUCCESS in 15m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/33b8c3d917104f32872686819f542d3f : SUCCESS in 29m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e76f1856eebb4f5ebeafd45bd57378dd : SUCCESS in 30m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ab492eb3bc974e3f8988378324733166 : SUCCESS in 16m 15s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4366cff527e643f6a89fdcdcb368cdbf : SUCCESS in 15m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8558bf636a564420a66c7372ce564649 : SUCCESS in 21m 05s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b15e0932227447358dae16ad97ab62ef : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d8622d20c5fc4e46945cff6b78fa1c2f : SUCCESS in 12m 51s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ccd1774e420b48a994517995d0ef213b : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61b38c7f6674097b3f4f0a63f0050ee : SUCCESS in 5m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94c84c6ba4df419cb245cc5a851341a8 : SUCCESS in 1h 12m 01s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4a734c9fa9c74a17bb9d825914f7bb94 : SUCCESS in 1h 24m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/38bc9aebc61145b5a32a074745b683da : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"98ed80774646ab13bc2187dbeba306d09500f7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:45:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9b0f4878449696edc9d1e443b200d3192d7cbee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 01:29:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ebeecc8a3b0562e91373255490f0d7afdf6f9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 01:29:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f51c8fcd9247ddab55804569dc6194 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88922bb05114416f919697374bc35b79 : SUCCESS in 7m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0615ab8307234dc7be921089f8d88819 : SUCCESS in 8m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46d14a2a0ec14b0aa800ec9a91555709 : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41343a4d3e8d4c6c8e40e42469c9ac8b : SUCCESS in 5m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6e7d6a5401fd4cae995ea48b51d74e0a : SUCCESS in 10m 17s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9f747d871df04238a83b412717d3eb94 : SUCCESS in 10m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/03349df46c60488992236d6dab84fa3b : SUCCESS in 11m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1c843485ca874d17a719f83612a23ccd : SUCCESS in 12m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/114741321b3943da91b6b8d6a355db10 : SUCCESS in 8m 22s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/607937e33ac046d3b30236c634349ba2 : SUCCESS in 7m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3d5b05bde02948b5bc8ec48a8d015aa4 : SUCCESS in 14m 41s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/389ac8297b8246829433f5491732d005 : SUCCESS in 11m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0db09f8e1aad41a5a098db99df627a85 : SUCCESS in 21m 36s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/6b59e9ee3cbd466bbbfe8818d80d1496 : SUCCESS in 20m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c815f20f9c394e2a97a52bba6e5a7b58 : SUCCESS in 29m 57s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/afe1f2c62dff42b29bc77c30e9cb4602 : SUCCESS in 23m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/00c43bc63b3a4da992e87be9f6ab085d : SUCCESS in 17m 10s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/77713a0195ce419fad698c7c32a06e70 : SUCCESS in 15m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c46d3e1080ca42cf86180b2d0f518feb : SUCCESS in 19m 18s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a68db2d2f7294dc49e98bc1698bc350f : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/df9f38396c0442619155e81dd47ad3fc : SUCCESS in 9m 40s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/329fd9bad13442fa93d81609cc0f20f8 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bace1426e5e4e408b5228447abd51a4 : SUCCESS in 7m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8599660ad024480d94732cc197e1c9f4 : SUCCESS in 1h 11m 35s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4a5965c2c25544d193470d5bfbfcd783 : SUCCESS in 1h 24m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c96dabdc8dd74b36ab80691b1a7f510b : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b248da299c85a17230b62236e4092178843cbdd6","revisions":{"b248da299c85a17230b62236e4092178843cbdd6":{"kind":"REWORK","_number":1,"created":"2020-02-07 16:56:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/706575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/706575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/706575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/706575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/706575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/706575/1"}}},"commit":{"parents":[{"commit":"3898e19b8ecf96cb729977ab1b2820345ab532bb","subject":"Merge \"demarcate python-memcached requirements\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3898e19b8ecf96cb729977ab1b2820345ab532bb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 16:56:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 16:56:10.000000000","tz":0},"subject":"update constraint for sushy to new release 1.6.1","message":"update constraint for sushy to new release 1.6.1\n\nChange-Id: Id278308c5bb3d83694bc8fedc05fae994219fc4f\nmeta:version: 1.6.1\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: I03fe8816413e00c3f0bcf06729b5c601bacc752a\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b248da299c85a17230b62236e4092178843cbdd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b248da299c85a17230b62236e4092178843cbdd6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
