)]}'
{"id":"openstack%2Frequirements~706576","triplet_id":"openstack%2Frequirements~stable%2Fstein~I706cdae8bfa5bd2eee2d16c2009b8d1888794c0a","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I706cdae8bfa5bd2eee2d16c2009b8d1888794c0a","subject":"update constraint for ironic-lib to new release 2.16.4","status":"MERGED","created":"2020-02-07 17:00:27.000000000","updated":"2020-02-10 22:30:46.000000000","submitted":"2020-02-10 22:30:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706576-1581373846486-7c055bf3","meta_rev_id":"26d6520206c4e555bb0c37b22751bec3d7951cee","_number":706576,"virtual_id_number":706576,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 22:30:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-07 19:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-10 16:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-10 16:26:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 19:50:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-10 16:26:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-10 22:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f234f2b39d63a4a6271ffc0034769b6b56d0798","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-07 17:00:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"758ca5a3811a459b2a3898742e7ab71d0b77e74b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fd16cade8d2a46d996253dd8ab0749b5 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db26c75d1ba1499db5ca7bee8e699c35 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/629bfa4bbdfb49a9ae27d4e934a04611 : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fd282cbb120b423184d016b91bcf56bb : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a04a621b0f864717b152fcaf514f978b : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/699d969fb76946e2afa78db59623e0eb : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8061f3bef484839ab3aba637bc8e202 : SUCCESS in 4m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cff310afdba349379d66a194a77d8894 : SUCCESS in 11m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4bc88951edd0493bb916b17fe25c8bdc : SUCCESS in 9m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec56a0a0d0a94f988dbde38785e5c227 : SUCCESS in 9m 22s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a540e20cb8ef4d7e9eb0f9a230175c19 : SUCCESS in 9m 48s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/42219bcadea8447c8883c56b7082f9b8 : SUCCESS in 10m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/76e4f861018e4b35b153234e2a066120 : SUCCESS in 6m 01s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4cd080650e5d413ca288f7acfa816ea0 : SUCCESS in 5m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7dc8cd80118740768c9e538fc8dc12bc : SUCCESS in 14m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/253c266486414d9fb94b00e542c6fa45 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/683a87341c8e47b2b50c350adf758d0e : SUCCESS in 6m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7c8a047c55334fa383d15aad573f0076 : SUCCESS in 24m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/548187f14ed6496fadd53bce933f5969 : SUCCESS in 29m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2b9d838b41aa4397a30bdce32834ba73 : SUCCESS in 30m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4cd5bbeee89c451da14b39583f211b22 : SUCCESS in 25m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/34544dcc27ec4a6481196c24a6930c99 : SUCCESS in 23m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/70cbd2e803cb4ef0bc0c27c5596b5108 : SUCCESS in 15m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/612a375ce08147c5886f79d0da3f5602 : SUCCESS in 20m 36s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/4231886544884a53ae133ea1e7cd939b : SUCCESS in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c3e7e65c1b55433faaf4b1cb84118004 : SUCCESS in 10m 43s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d6c7efb64a584b6787514e04a15bf454 : SUCCESS in 8m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df6e62d2d06945558f0a7d2b2f005d19 : SUCCESS in 1h 13m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/98e1216a15da487cbe1234def86a38e3 : SUCCESS in 1h 28m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb69bfc495d246a9846b93cbcf82b2eb : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f6f6fd09a2c368cea2a26b161b12e14efbeb1847","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-07 19:50:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9da63f94a85d09281149af94d23d68dc5256052","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 16:26:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af7118ffce0e3c1054ff04f33d0889fe0a50cf4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:26:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5567535007f4761281339c0b45ab4b37fc80ee9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:30:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"500cc96507d69eabf41527fc775112adab7926bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:30:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aeff6f17cbc4a93bebadc32f6f3cb83 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7222b10d9f7485b8d8482f93957fdb9 : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/608b74bce56a4a17a418c526aaed6cc9 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb3e6a926d74080a1e50b965d12f4c7 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab124ff69409448bbcb576237f07323b : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ff922ba6c4b4dccaa2e62017b5352ad : SUCCESS in 6m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/94d775100efb4e9cb9f1ae520830f241 : SUCCESS in 11m 19s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/89dad635bb0d411983e97e4c5402e7d8 : SUCCESS in 11m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7c7bb42e0d114158b49dd9a2d91ee86a : SUCCESS in 9m 04s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/602fe8553c8e4b9088232fc818e24b90 : SUCCESS in 9m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/70de569f19cf4ea2951b0cb546534155 : SUCCESS in 9m 54s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/1404114ba7a0462f89fec0d1d6a62ed6 : SUCCESS in 5m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/41313cf10b784e81a1ebd15e5f972be2 : SUCCESS in 11m 36s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/967712ea96354883a0773cbbdc377a48 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2a3092458324eb3877959bf020efd82 : SUCCESS in 8m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a682babf64de407fb92c63b6dd309121 : SUCCESS in 20m 22s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/abcb826042644471afaf2e0732a3d144 : SUCCESS in 38m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f3065fa46ba74676bbc7ed11af555c7d : SUCCESS in 25m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/669d5d23ea6c4c90a979e0e752d393f0 : SUCCESS in 25m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ffc9eb680f8f4f15bc651fb7cf7557e3 : SUCCESS in 17m 26s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/98c7ba4ff37a40e392d51cafc0fbb5e7 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3111745c34c64cca8596f61c1eb0909d : SUCCESS in 19m 50s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/604ba045f2b548cfb98e1e0755e985c7 : SUCCESS in 7m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f99d76ed131a4bd9ae293a6c5f151240 : SUCCESS in 9m 16s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/87711ed3831f4d118f8f8c31c2c64e2b : SUCCESS in 8m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b74b1185460248888452bb232b4c234b : SUCCESS in 1h 05m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d94b6af4cb074b82b3426d0cb9432163 : SUCCESS in 1h 32m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0187ed28b7a343e7816e4652ab684f2a : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98717dd442441539ac7a9df7ee6b71cfa4e0e633","revisions":{"98717dd442441539ac7a9df7ee6b71cfa4e0e633":{"kind":"REWORK","_number":1,"created":"2020-02-07 17:00:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/706576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/706576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/706576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/706576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/706576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/706576/1"}}},"commit":{"parents":[{"commit":"ebacd42fb7738bace4d084f70695fa46227db70e","subject":"Merge \"Bump diskimage-builder to 2.30.0 in UC\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebacd42fb7738bace4d084f70695fa46227db70e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 17:00:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 17:00:25.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.16.4","message":"update constraint for ironic-lib to new release 2.16.4\n\nChange-Id: I706cdae8bfa5bd2eee2d16c2009b8d1888794c0a\nmeta:version: 2.16.4\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: I6d9bf2c38519d69ac0798769af78e3178f10725f\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98717dd442441539ac7a9df7ee6b71cfa4e0e633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98717dd442441539ac7a9df7ee6b71cfa4e0e633"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
