)]}'
{"id":"openstack%2Frequirements~706579","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ifb1caddb6158be1ce595b15cba092b377d4d8331","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ifb1caddb6158be1ce595b15cba092b377d4d8331","subject":"update constraint for sushy to new release 1.8.2","status":"MERGED","created":"2020-02-07 17:03:05.000000000","updated":"2020-02-10 22:21:22.000000000","submitted":"2020-02-10 22:21:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706579-1581373282978-4f5f8f67","meta_rev_id":"496b0bb838ae01a2d6a8a96a0283ac29ed39ed73","_number":706579,"virtual_id_number":706579,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 22:21:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-07 19:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-10 16:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-10 16:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 19:53:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-10 16:26:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-10 22:21:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"262ab05b58e51a4d0a55276872f0691765fc8dd8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-07 17:03:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13fff2d741728a64fa927a74269fba40bc84d008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:50:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3dddc583677743d68d2b3d04574c4bde : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1c4a5e2ef44d77b53da8311da2c2c3 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/889256def2994f318e0bc021d80d946b : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c7555f49fd0441e8a57c375b9a60746f : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e86da95b50364dbc93fbac58c85ef918 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bea984711e3427aab42a062af525572 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bcb4664a481b4a79a243163b19ff6aa5 : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d98ac0b5c6c749e29c566eec47394ebb : SUCCESS in 13m 59s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/03ee41c29e15449b9db6938d477bb23a : SUCCESS in 7m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/668044ae61ee4092b50dbb330d2b46ec : SUCCESS in 7m 22s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1d4a9d605f3340f9a719b32e6918a7e1 : SUCCESS in 9m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b6221b93653a4670856d8361ae925d66 : SUCCESS in 11m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4bcabfb249a9452fa5e9d0356f345aa6 : SUCCESS in 5m 42s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/49f7b40b0b2a427895708090c1fbbf5e : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7eefeedcf0be48d1bf843f8cf212c4fe : SUCCESS in 11m 40s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cdf2564184584fa89021b27384d4423d : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f93b7a7deefc4266b3c77c7b2de026fd : SUCCESS in 6m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/75925d51ac154bc898349c9a021b21e8 : SUCCESS in 25m 07s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7c37489bcab94cb2b7cc01d7ade64520 : SUCCESS in 27m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6d43bc10205840278a58a7918cb4dbdf : SUCCESS in 25m 56s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/3689a1f96f584c8783a8ad28911939a1 : SUCCESS in 22m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e852ac328e774d33ae11ea938a485b29 : SUCCESS in 12m 35s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/ec60ad2030c44829886ee2f7a9b1f327 : SUCCESS in 13m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1912b8c924ae42f2a5713e5e57c98821 : SUCCESS in 20m 06s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f8d99d26a8374af693742c46effb8f7a : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bb205a87a645431cbf5940f2f2eeec78 : SUCCESS in 11m 25s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2a538e24cb5d4f748832fbd0615954d0 : SUCCESS in 11m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fd5643ab27c4f4ea3569d3ec451f683 : SUCCESS in 1h 04m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/22fa3143e62443e99946a690094d0b1f : SUCCESS in 1h 19m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e71fbad4e774817a3a8d63d9bb335e8 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5ec089b50eb8329ec5c63edefd5dfb4002898c87","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-07 19:53:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d86061a95e0fab96206ef441f057c7b95caba1b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 16:26:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb1ecc29fa4672b1e7f328dddb01d03558d1bdbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:26:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d681a6739ac923e81b8e1a56717ec9662dffb022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:21:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"93ef3390cee23efdde9f5fc9ae5a4a311fe081d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:21:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6381310331e4e42aca4c563a743f9eb : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f87db178c604749b9c2f087d6460804 : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6aa7fb61c7714ba6ae518b6497cb9876 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c75aef69af54bc58c8ca328115a01f8 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5257d24fbbb48e18b68686cdc3cbcc2 : SUCCESS in 8m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d2dce93bdf549e39e0a857de812fd54 : SUCCESS in 8m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d37bcd3e595646d0bdd9f16d3c3f33ba : SUCCESS in 12m 02s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6ac38eb6ef9643adbef042126c2c0dc7 : SUCCESS in 11m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2ab0efa7732247889acfcb62dc53b34f : SUCCESS in 11m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3855b350b4e54e79b5f9af629d2fe4c2 : SUCCESS in 11m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6cd3f0a8c1f34ebfa01749d77709af7b : SUCCESS in 7m 09s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bdbf715eaecc4176992f01721ed2a14f : SUCCESS in 4m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fbc2c36b4ca244ea96f7d7ef24ff40e6 : SUCCESS in 10m 28s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f93a94b9360745db9001cb61620221d4 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78b38bf5563e4e46a7d75fed55ed8539 : SUCCESS in 9m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2adcb14d744344b1b53534e2045fd8fd : SUCCESS in 25m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/0ff1cdcf1a844d39a23b3bfbb6787603 : SUCCESS in 31m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2a4cb7efe4e44cf58d4418905d68335b : SUCCESS in 24m 17s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b986efa2b24341b69f8e8806bcc76805 : SUCCESS in 35m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/05c2dd08bfb64916b58b5d7d11a07099 : SUCCESS in 17m 18s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b0b5c4ebbcf14c858b7174ae5267fe76 : SUCCESS in 19m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e21e2cf09a34a18b15c55c6bab9f0f0 : SUCCESS in 21m 01s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/363beee4a32946a2a859b8e84c2b5ee9 : SUCCESS in 8m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/124fc15517f0488e9a9c7caa3fe7b4dd : SUCCESS in 18m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6f080fdc16054acaa12f9046594ed7e2 : SUCCESS in 10m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de7cd0506c154f1aaf1d85c7f5219143 : SUCCESS in 1h 09m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a4d0d054e00c4ca5880a0344d738940a : SUCCESS in 1h 23m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/67e081816a8e4a18a56f826177eb7371 : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7a01eddd551efdc97ca51e69c050f1ed08eb153","revisions":{"e7a01eddd551efdc97ca51e69c050f1ed08eb153":{"kind":"REWORK","_number":1,"created":"2020-02-07 17:03:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/706579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/706579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/706579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/706579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/706579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/706579/1"}}},"commit":{"parents":[{"commit":"ebacd42fb7738bace4d084f70695fa46227db70e","subject":"Merge \"Bump diskimage-builder to 2.30.0 in UC\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebacd42fb7738bace4d084f70695fa46227db70e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 17:03:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-07 17:03:04.000000000","tz":0},"subject":"update constraint for sushy to new release 1.8.2","message":"update constraint for sushy to new release 1.8.2\n\nChange-Id: Ifb1caddb6158be1ce595b15cba092b377d4d8331\nmeta:version: 1.8.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: I03fe8816413e00c3f0bcf06729b5c601bacc752a\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7a01eddd551efdc97ca51e69c050f1ed08eb153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7a01eddd551efdc97ca51e69c050f1ed08eb153"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
