)]}'
{"id":"openstack%2Frequirements~706692","triplet_id":"openstack%2Frequirements~master~Ife25b91b603ffa42f32a6c77ed4289f354d3b14c","project":"openstack/requirements","branch":"master","topic":"bug/1862439","hashtags":[],"change_id":"Ife25b91b603ffa42f32a6c77ed4289f354d3b14c","subject":"cap oslo.concurrency for python2.7","status":"MERGED","created":"2020-02-09 03:20:11.000000000","updated":"2020-02-09 19:45:45.000000000","submitted":"2020-02-09 19:44:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706692-1581277453910-e685ee51","meta_rev_id":"dc804abcba27d46905191f9d2cc760a4da154322","_number":706692,"virtual_id_number":706692,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-09 19:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-09 17:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-09 18:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-09 18:17:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-09 17:38:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-09 18:17:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-09 19:44:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3837a04882d0b0b36524135bb055ce17959e452","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-02-09 03:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b89618bb9847851ed23ff922c247ed428ebe4d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 04:16:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b36bfcd62ae43b6831394e1cb91d538 : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b8837e99a5f47e58f0e8707cbeb55c7 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75fed3592ddd43c794a7cb82ab892383 : FAILURE in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f87b91dacad44fc081e49ef62c1f9b18 : FAILURE in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ab4d2ae973d4335b67617f719a51c92 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9093985199924c7e8f73fc4fdb86bc9b : FAILURE in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e1e8d42ac1045f4b7da853024efce8a : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a231adf549d4485aa3ab7ee03ae37cd5 : FAILURE in 4m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a82b2dca27dc4ceaafff0abdcd890719 : FAILURE in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/564256a6ec36469b881a7fae08f6acce : FAILURE in 3m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1711f69368a4481cadbfeb09ad2ad0fd : FAILURE in 3m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ee6ca22b83c47c2b1b4070d6893391d : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/38b4194dcfa24596be1b53c6b820a4e5 : FAILURE in 3m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/519e768cd8084a59813c7922b6c91d3e : FAILURE in 3m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b29de9def0f14ee896851e114421ea83 : FAILURE in 6m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a02f6857deed4763830e5c0f2ec65d57 : FAILURE in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65bdd4ac832346feaf54508730a6dc68 : FAILURE in 3m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eeb4cecd3771477a9229c8e3e20b822f : FAILURE in 2m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c4c728ac24cd4c17baa9f3805f77e1d3 : FAILURE in 2m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2318e4359e9b40b791c71c6f218a2015 : FAILURE in 2m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/63537e2693c44478a40364a16bdbf6fe : FAILURE in 2m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7bbc648034d41188ce782b203d5594d : FAILURE in 3m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f58882c59fd5469881c9bebad501e12b : FAILURE in 8m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/20ecba87437446ad8ad0184dc76b6b3e : FAILURE in 50m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e1a513e14ff14e6e9ddad11612deda12 : FAILURE in 7m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6d3718bf92144be589cddc9a8224b5bc : FAILURE in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"02dd94284dbda285b5d09fb6f185474e5863cd8e","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-02-09 05:07:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1065d57ed4736e46e365d6b27e1598ab6e2f58f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 06:38:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd146311079b4023a1777da91c207f94 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb04e5dc24c487bb00f123dbcf005f1 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fae6d0ca8c247219b96c5cb4f908060 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cd8f1943e13447b9eb3d03c2a1d7bf4 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4ffa70ba02d4b549e335c47c84d213e : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06a903c947f24170a4d5f3ec5e5eafe3 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a713030b03d4268aefdd1a6b74c9562 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a54d27d1b79a45f4afd588d9df2cc250 : SUCCESS in 18m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ad371cd1b2f94051ae36ece9576fbdc4 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c7892413cee486e8a14a59e185db97a : SUCCESS in 4m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/576fece9dc984c60be797a1f0fe0772f : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0035c78119cf47b78d91bdd56f0c6e01 : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3966413e3ccd4bb8885d7da104c8512b : SUCCESS in 14m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/adb6ce9fa96b45f2a0a51e184576bb05 : SUCCESS in 23m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a3c4781582940748125aebb49b9a4e4 : SUCCESS in 15m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cd79b2b63ab24f00998df6cea40edff1 : SUCCESS in 12m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccff6553bdf545c78d2168fdb2296d6c : SUCCESS in 15m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/212a0648077d4379b7b5e7e675a50f1a : SUCCESS in 5m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2a9cf13fa55b4d85acd5a78f5ada4c56 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd273278655e4f188d948c47a595a1b2 : SUCCESS in 6m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/47d5e368f4d841a88d212ac02f66e037 : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1607c55cf22a4c3590e823c1e7f5afe7 : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30b8d12057764dbbaa8b652a5cc0be2a : SUCCESS in 52m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae328f085a97427c905de7996ed6aca4 : FAILURE in 43m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/de92f935821947bb84c9c31124034eb9 : SUCCESS in 1h 27m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b44ff1f95e4d48c187f87dffcdf0658b : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e17dd4141a22d83a159f7321c6eab85b215b8698","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-02-09 07:49:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c03fa0a6b45040f4a06c12e93e8a0e438038c442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 09:17:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1b6c01e7184c26adfae009be58afe1 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02371cdba12746dcbda98cea16aa291d : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4eb55d4a1614180b78e65f151ce60a0 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c21b17b2a81b4445b8d9b4861ee61b5d : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/586e122d3fda431697c9b7abd8c78f36 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c65096ee08d84091aad960fe371632dc : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6da20029bc44d3e95eeae6caa310521 : SUCCESS in 3m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0390195bbb2646fda3d699a795c759d1 : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/336a3c9c158e403da4bc6c476f71c9c8 : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff5cadf83c604116a0dfa7f483e5a5c8 : SUCCESS in 4m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0875e4a2153e4797965b76c39ca0c6c8 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f5d9caeb796642828e03d4fd8438509c : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ad863b1d5de448cbf4738d4469040ee : SUCCESS in 15m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c76c024c8b404917b02c6efc51bbb815 : SUCCESS in 25m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/889e85d73bd34c8e97ee80926ca6cc9c : SUCCESS in 14m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c5a7f1fd74274313a8bd7da7553a1503 : SUCCESS in 15m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1659335a214246039944582d5101c0d7 : SUCCESS in 15m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c028d1525f5b433888b472ba31f76359 : SUCCESS in 4m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5bb35faac42b417a81359398eba58fe1 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9d258d8fa8849f195be9476a19c7d1b : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/22d6804d7c1d472ead1e10c5ea2c4f99 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/066b39fa0a6949478c679e7f87450953 : SUCCESS in 9m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebe235804c7d46638cc4cee469fc083c : SUCCESS in 53m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/84a970893266428d8ae33ac410e16615 : SUCCESS in 1h 09m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2cb9cf9988d6405bb18119e919a7d22b : SUCCESS in 1h 22m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9eae2f64e50d405a8f91b043d86deae4 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fae48658812ec0f37c24a42e82b9cb4231c03833","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-09 15:39:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9da4ddb516ea9b7b44622de36448ae37b3cae44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 17:10:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4e189d308c4d1a91b0e07b6acb88f9 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94fe544889c848b88ea6a4195373c665 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb3bdfb655594a3194f80690ab828b3a : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7241d4bbbfe84d31a657517fe322b8d1 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b35144336edd4369b07ee4c44a39bb4d : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d961efcd8e4d42a89cfd9e72cb5055a3 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ade7e9a0db8d46b9a8754b6f376cbbbe : SUCCESS in 3m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/023a371755ff48af909c3f060ceb0e10 : SUCCESS in 6m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d6ef054cadf54a05a06858d0dd52205f : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/135d8f7c9f96491f9230bc7d97d9d28c : SUCCESS in 4m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c885e5c9e7184c13acc9faf437446f91 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce1ad07c623f457488d3c5ff78029c42 : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f596da7dacb046cb876917fdd1381682 : SUCCESS in 13m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1e96068eeafd4c169e2d39af492a3706 : SUCCESS in 24m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cabc7cd4c5004823a147d3bed06c4276 : SUCCESS in 12m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7567d356ca9848c4a925fb14b505d4bf : SUCCESS in 13m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8cfa024820a44196896e3d9fe81dd056 : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7b64ac6a0d3f49fcbb0ac96524be4cf4 : SUCCESS in 5m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/632e55e505fa4966b56d85077f36b547 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a74aacc6465642068ab9f91df5e44d0c : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1206a9e9e3194ff7afab09d6fd7dc1dd : SUCCESS in 8m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c98b0066ab446c7851abfdb8e013190 : SUCCESS in 9m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab6e37fcbea841a59c7807de38e6476a : SUCCESS in 52m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fda3c0ac90541e4a238e1b3328f19c7 : SUCCESS in 1h 06m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fcda62d4a9144f698ce886ff285b4e71 : SUCCESS in 1h 23m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3fc59b1200d846a4a0db7a4d3e81c5ca : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":3},{"id":"69988ad9cf51c315bf80740fa5be656ada56123b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-09 17:38:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"02c100d22836b9dd77d7b84e45612e478de7874d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-09 18:17:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7585237112dbbcd7286dbb215d72b02740520827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 18:18:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c4329e3a018bbcccc3def60cfacef283fc5a274","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 19:44:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac9c671aca54859ac4660f1abf7bbaa : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c02764577614d1f8d4a4f393181d3fc : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54cf967aefb84241825a28dd7b52efae : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be36656edc2d4dbdba089b6860ae0d6b : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1aaca6c6900d4c27b068428e4462d343 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8c6676e8a9d4af89ba1a92f7d325da0 : SUCCESS in 3m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/41e0c3b0fe6245d084caae3cf5cdc652 : SUCCESS in 7m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97dac95206dd42bd80e655a181762d2a : SUCCESS in 7m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f337cb084bb6455a92183948b642d519 : SUCCESS in 4m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff3d9e356dcd46eca45b62c8cac94d04 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82b43a80aa844080ac054f9fc1bfb814 : SUCCESS in 6m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d0abbd50b22477696e1f26dcb3df9bc : SUCCESS in 17m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3a80c072b3743f588fba8e3e9e234a2 : SUCCESS in 24m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e84e68eba1564dcea13c1d7091af773f : SUCCESS in 11m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f571f543ac5846f1b13739bbebb60bca : SUCCESS in 12m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80f0adfcaef44ab48f443a3b3e3da604 : SUCCESS in 15m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/18b6520fe62749e2bda2001550f4eeaa : SUCCESS in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6096d8ec48914d789b3e95b502e10396 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a0ef1d922f146e2b52a6222b2058fec : SUCCESS in 6m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f887e9dccde849499dbfb4f0167f4f1b : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dce3c7f118164587b37369ba2900cd68 : SUCCESS in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bc7ea06628044d98fbb053bda319ecd : SUCCESS in 50m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b04872138d35425fbd2ffc410e5159f1 : SUCCESS in 1h 01m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a15d286a0e447159359110abd41d41c : SUCCESS in 1h 21m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2bd5fc3397794c238acab9f0aa7609ec : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":3},{"id":"48510cec9cca71ab3cbb8187f64c47333d13f866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 19:44:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"83c5eb5861ad90cb220ccccbfd46f00f5b9c55e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 19:45:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3f90b6cbf94b7f9e6558819e8f3a00 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4e86eea7c78e56b4ddd0dda647fd1c29a977b3c2","revisions":{"0053618e910b2e101aafced6ddd6b9c6aaf48f10":{"kind":"REWORK","_number":1,"created":"2020-02-09 03:20:11.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/92/706692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/706692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/706692/1"}}},"commit":{"parents":[{"commit":"f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8","subject":"Merge \"update constraint for oslo.log to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8"}]}],"author":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-02-09 02:19:03.000000000","tz":480},"committer":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-02-09 03:19:32.000000000","tz":480},"subject":"cap oslo.concurrency for python2.7","message":"cap oslo.concurrency for python2.7\n\nlooks like from this commit https://githuh.com/openstack/oslo.concurrency/commit/cfc27ef4972a99c5787e537b0f85067829517c4c\npython2.7 support is dropped.\n\nChange-Id: Ife25b91b603ffa42f32a6c77ed4289f354d3b14c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0053618e910b2e101aafced6ddd6b9c6aaf48f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0053618e910b2e101aafced6ddd6b9c6aaf48f10"}]},"branch":"refs/heads/master"},"932862fc70f20b0cb3d73975980f3a56413ed043":{"kind":"REWORK","_number":2,"created":"2020-02-09 05:07:21.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/92/706692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/706692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/706692/2"}}},"commit":{"parents":[{"commit":"f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8","subject":"Merge \"update constraint for oslo.log to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8"}]}],"author":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-02-09 02:19:03.000000000","tz":480},"committer":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-02-09 05:06:21.000000000","tz":480},"subject":"cap oslo.concurrency for python2.7","message":"cap oslo.concurrency for python2.7\n\nlooks like from this commit https://githuh.com/openstack/oslo.concurrency/commit/cfc27ef4972a99c5787e537b0f85067829517c4c\npython2.7 support is dropped.\n\nChange-Id: Ife25b91b603ffa42f32a6c77ed4289f354d3b14c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/932862fc70f20b0cb3d73975980f3a56413ed043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/932862fc70f20b0cb3d73975980f3a56413ed043"}]},"branch":"refs/heads/master"},"4e86eea7c78e56b4ddd0dda647fd1c29a977b3c2":{"kind":"REWORK","_number":3,"created":"2020-02-09 15:39:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/706692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/706692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/706692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/706692/3"}}},"commit":{"parents":[{"commit":"f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8","subject":"Merge \"update constraint for oslo.log to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91e2c61e4fd6311f94d9a6eb1f0b63d7585c7b8"}]}],"author":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-02-09 02:19:03.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-09 15:39:14.000000000","tz":-360},"subject":"cap oslo.concurrency for python2.7","message":"cap oslo.concurrency for python2.7\n\nlooks like from this commit https://githuh.com/openstack/oslo.concurrency/commit/cfc27ef4972a99c5787e537b0f85067829517c4c\npython2.7 support is dropped.\n\nChange-Id: Ife25b91b603ffa42f32a6c77ed4289f354d3b14c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e86eea7c78e56b4ddd0dda647fd1c29a977b3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e86eea7c78e56b4ddd0dda647fd1c29a977b3c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
