)]}'
{"id":"openstack%2Frequirements~706877","triplet_id":"openstack%2Frequirements~master~I8946476c011b3931b1fa1a318a5c7853a6c91317","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8946476c011b3931b1fa1a318a5c7853a6c91317","subject":"update constraint for python-neutronclient to new release 7.0.0","status":"MERGED","created":"2020-02-10 15:15:35.000000000","updated":"2020-02-11 06:34:41.000000000","submitted":"2020-02-11 06:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706877-1581402688298-4edda447","meta_rev_id":"262be543094b7a469dd8a5fdefc8d5e35c8bd9f8","_number":706877,"virtual_id_number":706877,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-11 06:31:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-11 00:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-11 00:40:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 00:40:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-11 06:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e682ece38cfc0e8ec1cfe0ce6af1eb86bd2f635","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-10 15:15:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91d6c83078636cc24bdf80409d83fcc90c883e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 20:50:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1f138a2a654a0eb56d757beedf9b4d : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/005584b222e74e09881dd38fd7551273 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd238343c6124e0c87b4950ab8f6fea1 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88f98ed550e749c38c304caa57690dc4 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e23e10c0037a4569a55c8a7be575a1ba : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99e757efcf294a0cb57f7bd407626629 : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66fd9e7d50d24b299bb7ee9b6298739f : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5964adf13f0544e7910b9bad61e7b4a3 : SUCCESS in 10m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4df6dffe716b49ebb74ef6178fae15a9 : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/51a141fed95245258d28f29f1faace5c : SUCCESS in 8m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ccb73492c1eb46dd9b1333b958a98d4f : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc0f24148c7f4f2b8b82e76922930320 : SUCCESS in 10m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15b2ac76fa0145d89d145d36f56d4c37 : SUCCESS in 23m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8344eb0e26c54a65877a874eb69c9689 : SUCCESS in 28m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4515ba4e3baf481596045822dc3d4067 : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7c89d73d584e49f9a760d5d310857658 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1eb9b830c9b74015ad5434016e5bf3cb : SUCCESS in 16m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c02fcbcff7d846d496cc8068de4d80d4 : SUCCESS in 6m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5d1727a1db314edf84241c986c4fa023 : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d424be9bd064bb484ce1836146a6bd1 : SUCCESS in 8m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/525e7f08b5304818898da7a9673c606b : SUCCESS in 14m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5afd9d103d53474996a05a558983531d : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d824b1e13c84ade9a1328f83654f972 : SUCCESS in 1h 02m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d200207ff50749d3b13310edc4e826c1 : SUCCESS in 1h 11m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a2f641333314cedab88373cb7e3b3c3 : SUCCESS in 1h 40m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b5f0990de2bf477484efc18153058f1b : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"14289be545c0bd5635e87b5de3b7f96dc4fbfed9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 00:40:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed5accaaa00e627b16a808b241efa798e56a6454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 00:40:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"603a08438975938cccf8f4a97cd0bcfcfa2017de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 06:31:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d4bedb7b7046b6b044a022f66e9d5e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77451dc527cc462a8651bd079d2ae751 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a0ed27df2324b8eb99735a5e326aa64 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7016fba0957f4a7fb980821f8677cc15 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91955a8f55eb4a4c9617fae4e4744810 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99b43467530d468f9668c2ffd0e59f2b : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e9786c2a8a3444be9a6c2e70f0f6df3b : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/32c2ae085ae848bbaf5d8f5703f22ffa : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/718e71ceb68b4402a872a6d362dfcfe1 : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/18f041d8b04f4195b56a4c03bb6d8e91 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb1c3d30bff24aad8019e816101f1c8c : SUCCESS in 6m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/600ad8d2f4df40cc9080d79e5bc0ce30 : SUCCESS in 13m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c582611c305e40d7861fbb0486fbe69b : SUCCESS in 28m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55b7ff16ad3f4535bf5c3ec8715cfece : SUCCESS in 12m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3f9cf18c7a584a529e09445e0e3d26b4 : SUCCESS in 17m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c48ed1eb1c94c6594b20b252d2d1e10 : SUCCESS in 21m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a3285e729920458ea97346bc95b0d57a : SUCCESS in 4m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/15d231e4ed1044778ded6dee62f347d8 : SUCCESS in 7m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f07618a92af642f6b6707ffad23b05e0 : SUCCESS in 5m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/11abb95aa6b243d2b20b246a4f98b5d9 : SUCCESS in 10m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b56eec86b2ec404e970dff41612e7a72 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3019409dbf74410a8f8f6b86705c4f2 : SUCCESS in 55m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be6048568c2d47ec9da36dda4970d48d : SUCCESS in 1h 03m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eb61eec864da4b7184d786250ccbc28b : SUCCESS in 1h 45m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/40968062bd764ffcb702b1998968f23c : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f58d8400823d657c6b00e708e9e3cdfefb4e126b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 06:31:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9aa6d7840f822bc98c3be03caacae171fc810d46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 06:34:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685a872f13724267a27d9d7b4a1cd810 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f39907b95f1fe10887a0429cec11eb38d138f0a","revisions":{"4f39907b95f1fe10887a0429cec11eb38d138f0a":{"kind":"REWORK","_number":1,"created":"2020-02-10 15:15:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/706877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/706877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/706877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/706877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/706877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/706877/1"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 15:15:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-10 15:15:32.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 7.0.0","message":"update constraint for python-neutronclient to new release 7.0.0\n\nChange-Id: I8946476c011b3931b1fa1a318a5c7853a6c91317\nmeta:version: 7.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: Ide08177206c95a810c6a941b5bcaed6429f88dab\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f39907b95f1fe10887a0429cec11eb38d138f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f39907b95f1fe10887a0429cec11eb38d138f0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
