)]}'
{"id":"openstack%2Frequirements~707064","triplet_id":"openstack%2Frequirements~master~I31e6f5a934005de3a0d6f21345949ad3c1c14171","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I31e6f5a934005de3a0d6f21345949ad3c1c14171","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-02-11 06:14:55.000000000","updated":"2020-02-20 04:38:41.000000000","submitted":"2020-02-20 04:37:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707064-1582173445623-bbef5f2b","meta_rev_id":"f6cd496a788ed615e3c04448da1c755b472ea207","_number":707064,"virtual_id_number":707064,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-20 04:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-20 03:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-20 03:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 23:30:24.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-02-19 17:06:15.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-20 03:19:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-20 04:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b3d99b2e0c36a9cb3efce2331d5528d180693be","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 06:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"229c1bd5f36a882093b6be4ab423c7d046c363e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 08:08:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9671926e420d47f78dba609bedb6c35c : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e5fde65dc4c49ae9530eebc6e8d9ed6 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9faa86b1340b41e78e3d72f4b5fcec9d : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f38c1028c02f4260a48b2e6c14ac3b84 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dc5672d45544c84b72012d3470d6246 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be61e273e9e941ee812dd6b59005a1ef : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f3f018d38ff4151a93ec76218697b6d : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/02b97a9ab328498e8db7abbfa2c454b3 : FAILURE in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7b97e748699a47dc8f6bf7f4267e4757 : SUCCESS in 11m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd918f06075a474f9b1b7c862858026d : FAILURE in 7m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/edef9317c7944f61834a2040241e0c1d : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f68c19dfb52e41b19ead5ec1ee4e0753 : SUCCESS in 9m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/694da2c592e94752a8ae1781deca748b : FAILURE in 16m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fa7f22312e147cd831226e3c64bc6f6 : FAILURE in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0564f501ccac49b692e5028f6185acbb : FAILURE in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/84af3e8fbfc54b45976cf0a9222efae5 : FAILURE in 21m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/942580a5688b41ff887183da7d9c1002 : SUCCESS in 17m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5b229061da0345209d7fa2904f7ce192 : FAILURE in 6m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1c50c563873f4bb6aa35576ca8049eed : FAILURE in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79ffe18f5eb441fab9bf0a42043f23db : SUCCESS in 8m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/67eb94a4523f4f69b3b1a0f985210dde : SUCCESS in 12m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/daeb13e49f7b49a4968076df9f1a0d3f : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54229684924b4524b134829b11e43559 : SUCCESS in 53m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/469a902bf3814496937068d20a5b04f9 : SUCCESS in 1h 03m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bd4f73b0b1064da9acaf0ea835f04c07 : SUCCESS in 1h 16m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ad95ac1554dd41b4936deb1fc230ff4b : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7e07108f33cc7864b658075f2bfb3312351ffced","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-12 06:19:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"120ca405cb208d7681c454476f1d5142d79dc585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 08:39:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67af4a2f6a2448568d79a6c74413d99d : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b752e32096ac45f3b41b11e2ff0125dc : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/951795f8483e48d385e47510e92fbb4d : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf19c5b949cf4c50bad6f1c259d83dbf : SUCCESS in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b66a3385eb0e4570be23646ae4c3f1a5 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91de2f90b86449d58f4ae85fb1f469e9 : SUCCESS in 8m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a88f4fd242541b6a1df2fc7d4f58f7c : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b2554f5cf2e04672b81340484f5a7138 : FAILURE in 16m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/132296d1a26c4f99a4b44a2018364737 : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/01ab8131f8034dd4b679a8fbf150c985 : FAILURE in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/247afa61d47948208b16d7635ad7d847 : SUCCESS in 15m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e6efe109495400695d1d8ebd5432a64 : SUCCESS in 16m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d886cd0feed34be6bf5b9ee21ee3ae98 : FAILURE in 27m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/155eb872d12f459bad3877622d5c0fd9 : FAILURE in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/966acb0057634b119a400e6cfc2492c2 : FAILURE in 20m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4ca635aa2e94482e93a734dbda17ae0e : FAILURE in 15m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/483e9c12f8ad4ca780ca82715c3892b5 : SUCCESS in 25m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f9d88ad2e3b247cea0f7efd841c7e1db : FAILURE in 9m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bc95a87cc9644102a96db5e2e5ff3c10 : SUCCESS in 13m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/573ff8b15e324046b62cf50a95140995 : SUCCESS in 12m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f4a8414d21d648c38228d70ae22e4ab1 : SUCCESS in 14m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6fa5599d91674ea490a9b020723b982e : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c528a5666a54f9e98c1ffc5180986e6 : SUCCESS in 54m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3e67738d41d403d8f5a4c36d6157c8a : SUCCESS in 1h 04m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b472d5aef63e4c56be7f4ebad0951de7 : SUCCESS in 1h 32m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b0da3b3d32d4f67ad72dbeb3c4c2722 : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6a33aa68b5d63a64b6dd1efa20367f64e03263b0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-13 06:16:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45e64ee3b870c4be53ca2c3b3a5cc4c8b3ae2ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 08:44:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c1e355c110435187a9fa4f0d067e74 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab3ed26af4b473bb1f2baf3a80b284a : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74264da52e624ad3bc9248077ec5ee44 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32e27f04d34f47f5bc01dfff617381a0 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa36398e058a4feda60091c8d79228e4 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ec9256e0c104ac08e43b73e1b6e6ae1 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40363ac2dd1e4e53a491ccd6bb5b1c06 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e529f4e8af2946539da8d215f3db6aeb : FAILURE in 11m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52d7268a2f0144799a1690364b4a645d : SUCCESS in 12m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/802a4ae4980242499e3bb99969e9ca4e : FAILURE in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6c3c286dab784028a7a3b673c6f9497f : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a413495f79984f84b4b25a7c45e46b74 : SUCCESS in 9m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f6caa60128a3429c83c2a22c291e9bc9 : FAILURE in 19m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4be4cc26f7524ce19622200c0194f9ca : FAILURE in 24m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b27f1663571459d9ce6d6133a695c40 : FAILURE in 15m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bc88817263044ed4863e5e2cad5aad66 : FAILURE in 15m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8cd8ed13eea14dbc80e9c6fd22ade4c9 : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/643efe5ff097418b82a8b31a565d66ac : FAILURE in 9m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/da99c396561845f082ca917b25fa45a3 : FAILURE in 7m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6049109e581a420aaaac05d06c1fd3d8 : SUCCESS in 9m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18c26194456b46c1aea917d748271d32 : SUCCESS in 10m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e4e3fb95b5c24c1ab7bb1e2f80dad246 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c836be77df7c4b9d92deaa0998350082 : FAILURE in 39m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd44b9861fd24967bfa59caa7e2908bd : SUCCESS in 1h 03m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f85527485744ef19df5b571775b852a : SUCCESS in 1h 51m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a2e025764be46df9a585f39c099cf11 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f5699329bfd727b7b79feb1e946a9ed1590a0100","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-14 06:19:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4b39702209f93a031dd380c78e37e265ed9dd74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 08:41:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baed2f9086ec4725bd715c90da81c04b : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6c275e54e24d96827710634e1bfe21 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eacaba2a98ba42b79e2f141bc6b5f744 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1b39b4b3eb04fd7b8c7b62fdd952a7b : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87dbcdf9880a4919b1dfbdd5d8590d42 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edf7df57f3bb408fb268f087aa6a42ec : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/910be17d0b7a47dab6df0b4d6b1d1f67 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7bf7f92f791043c9a2ecd2be5f56fb43 : FAILURE in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3bbacb14184145108c49bd5eae7d1336 : SUCCESS in 9m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d8bbd2868644c73920464106ab7d92a : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/025674b00df64908911234dc670f99d9 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8194ca69307a4dd0bb29c2b5197ced0a : SUCCESS in 9m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/558d36f618e247258445bd998c64394a : FAILURE in 22m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/94d2dfc3a9094a9b965bda2435697806 : FAILURE in 29m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76dc2bbfaa3f4186af88c93cc27068c1 : FAILURE in 16m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3f9748d6900d4f3fb666c2b9b604fdd9 : FAILURE in 19m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f624491abf864527974188ec1b0ee639 : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b6b65627559a48b28d8a55948d9326bf : SUCCESS in 5m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/abb227f3d1624eee8c2cdf6a77eb9149 : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4224db7d497c4501b78eab30a140778d : SUCCESS in 9m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc1049d093f9466db489c7c3f3fc9203 : SUCCESS in 14m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1c937855cbdc4e8fac3f3773f6a1b8a6 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9f7eac3c4cb4b2ca115c08d7b23bdaa : SUCCESS in 1h 07m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c6ee375641146b08606cd707943854e : SUCCESS in 1h 04m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/440e74eadeb04858b6f56ef07f3e146a : SUCCESS in 1h 24m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1630c79b4aa44b38b5ae68e1df31a5c5 : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":4},{"id":"4a1ce355389c8f94726c63df9166900bda4bf1f9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-15 06:22:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e90654c8a7c53d214f124920e07e7298d049549e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 07:53:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6daea319c08e4df2883917877f4e509c : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547c837f9bf54bbaa9ac75bbe0d32216 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a25ff97d8e1452e92520b883217ff4f : SUCCESS in 14m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ba9b4c5732c41c697aba02662d43551 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41f125b82d3d4d6aa105700df11d1d70 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3148fab78aa4343bc4dbd200001508d : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68d52ebf832246bd9f3cba3574e32bf7 : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e91afc660a7f4a10b7b99ec4c14359b7 : FAILURE in 13m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/17c41846cf9642fdb6d5f85582d23c20 : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1f6cc7173a294528ab30d8ce10034e7b : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/83bb6425cb034f17a250197b54db213c : SUCCESS in 21m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1c7a70187ad4d94ac644d498a80fd16 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/00a68f27e0ea4ec6940a8853e97fc3d7 : FAILURE in 22m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3225eee7376d4b7c8df172efa785bfe4 : FAILURE in 30m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1331e18961e347c482fa361482f947cc : FAILURE in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e6da0472a2414f76804dc4447897d202 : FAILURE in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f71ad9cc23c24cda86b5c1136fb0328c : SUCCESS in 20m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d43d53534bb04346900a8250d6ac8858 : FAILURE in 6m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/541937c1d7614fa1a596f8a4e50a9d6a : SUCCESS in 8m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/092720bfc2164ba4bbbbd27761355d8b : SUCCESS in 7m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/29eefa9c1a694935bf592b62016b0f86 : SUCCESS in 10m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e58653772e946f0aab1e00c58a258de : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01bb215b360d4215924c5772212e1e25 : SUCCESS in 55m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/850d6ab4c7ec46c28924941f128753e6 : SUCCESS in 59m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e0b29d8bc6ba4d77aa686f777b5cbb43 : FAILURE in 16m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/045b9284ff644b37a5344b62a4017fed : FAILURE in 41m 26s","accounts_in_message":[],"_revision_number":5},{"id":"2452d8bed1b13a03409542a8e256bd81c0c32b4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 21:43:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"08a36b0094116aff09da9a5f9456e90327e5b53c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 21:44:04.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":6},{"id":"6f7e78e0ac65329d4fc98a84b10ee4949e3b0098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 23:09:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0e346a6f1f4478918d30109afa594f : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b2fcdc9b6047dea661f74fad11478c : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09654c2bb654432c9ef7cddccf156e97 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2525b692df4c413c9a5202d1748820c6 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdf766cbc52f48b1aff5015169a695de : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd40b1e5860c412093bdc66076037057 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3dfbd73afda4d3183235e1ff77fb73e : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e422e583b16c4f519a6f637cc4836a19 : FAILURE in 7m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/37c3cd64381847d39525c7aaec44b378 : SUCCESS in 8m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/acabb950e36f4c639c9ccc2e30aff7f7 : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7531a599d0694ac6a10debad144ef384 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f629e7ab2cbb4f029020dcf575e6c365 : SUCCESS in 7m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b60d9ef8da0d4082b5e6ce7b995f7157 : FAILURE in 22m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/da3b7a7b80bd4703b5d45bd018c53961 : FAILURE in 26m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb12328235fd40569bcb6cfdb799ce13 : FAILURE in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f00db649e1cb4943ab460c6a9c4c1292 : FAILURE in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67d658e6b930422195f5a80f10fee56f : SUCCESS in 15m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1c495e7e86e04acdb68151bb2071bf5f : SUCCESS in 5m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/df454323692246b2a25a4097c15df991 : FAILURE in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3483cd7afb9e44e1adc623b752353e31 : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fc9704b0fbab4fdbb5e4a0edf9b1a37d : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e3beb5bc899143688480a5c322178d2e : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fda130e9eee34038a2f3a43cc8bad303 : SUCCESS in 52m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d643b6ec09b64f78823c6a470c6ccbaf : SUCCESS in 1h 04m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c69c61c22e95488e8bcf153389cbb50a : SUCCESS in 1h 19m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/755ba04d3a294a0687e064daf255fd78 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":6},{"id":"5bd62b5e802c00a7582322d4c93e1e6695e76b93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 23:33:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5bbe7c35b323b1f95f2c27ecd50bd4c6c34f90d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 00:56:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d470421a65a549888c1370b68d780090 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffed35fd3da4c859e85dd0d319ec81e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8df3f2a221d4e02afa4cc5c736d508e : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1291b58a02943a99828074c169c067c : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a4060f67d2427ba94755ba6053b1e7 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfd66b40bb084e0fb1f1f5eb1ddb312c : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/407165d43e0448889847ce5d06999e5f : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77b5ce9ffe6f46c890a9b3fbbaeabaa6 : FAILURE in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fdc7d35af5524cb68cb308bc18bf8732 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/520146bab5234932ba8c9c568dd77003 : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e0e1a11632e4d7ea3fbdbc9d901c89b : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4570a5844b7a454c9d4fb475de31e56a : SUCCESS in 9m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b5ddeffb05fb41e8917b9d69be0768fd : FAILURE in 14m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8ecba052d0440f295f096892f3d8d8c : FAILURE in 28m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/de818880a493423b975a3d6b2b4b28d1 : FAILURE in 12m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/13aa680af4bc45b3bb95d6a0f5860466 : FAILURE in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/667f64220aeb411c8ab2963187e8d758 : SUCCESS in 19m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65309102d16a4198bfc3abf6f6cf177a : FAILURE in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/89cb13add8784670ae7d76d490661696 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59be23da68b24f7994f8a6d841c58abb : SUCCESS in 7m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dda8b13c9ae54df283db6eb2aad1e971 : SUCCESS in 13m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/76af9bb86d7b4000921e565ae8fd9d6e : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cf1773107d94ce888259a8ec711f5da : SUCCESS in 54m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90255da870e4417fb1063c7adb77a861 : SUCCESS in 1h 06m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32a7fdae599d4936b7ec37e07ece36e0 : SUCCESS in 1h 19m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/18d542dee0d44eb0914a416ef269fb2d : FAILURE in 56m 14s","accounts_in_message":[],"_revision_number":7},{"id":"4e5fbfa6476a51810a0380fcd857b317e34993ad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-17 01:15:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8278d17f8afdea57db31eea36dab50807c5b573a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 02:38:44.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4122aef8d3846b3a3f339412a170468 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19979894d9147d9a40412a4625c162f : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f1f2051ff44e81b1ebcb3e17e5960d : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b383616c409411d85cb577943c1d147 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a91529ddd6be4e539c1596e42acdd2b2 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d04cd67c7c642f8b1d13fbdd35271ab : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20dbf5ac6be14f3aadc830e64be7d291 : SUCCESS in 5m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc36834c59144c9b99fdf2f2e32256e5 : FAILURE in 10m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f7b68404848e47e9909ae0d8e736f37c : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a820a96cc17b4c54899e5a231baf97fd : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2ffc359068b341a7854c469a16c5c4eb : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a08a801c3deb4946ab79366704e75db9 : SUCCESS in 9m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8cbfc10b9c0a400dad82b765317bb1db : FAILURE in 18m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ff9c90f30b4346f9a9733f6259a856fe : FAILURE in 25m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4200896d4bb04842add51e6cd1752b65 : FAILURE in 15m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/df0510362208443d8776fea7ba12fb3f : FAILURE in 15m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb1b6b501046427b8a996049d63b5dc6 : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fb0245b69cf546af905fdc96dc68fa27 : SUCCESS in 4m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0cecd4124e0643f3a4ac26b1b3adcbf5 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/705838873cee4bb28d9c3d2b1d0804b1 : SUCCESS in 7m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/583a239b522949dab5c8d8aeca3f32c2 : SUCCESS in 12m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/503c9431ae984ec49eb66e796c4edae8 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e482da878296444cbd9ebfb88b403464 : SUCCESS in 52m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1aa98cc24214a659eab1bae722dd859 : SUCCESS in 59m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3af910f4d31479b809e703797f7a670 : SUCCESS in 1h 22m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2abcd6d9e39a4daa8cd74003d33dde53 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":8},{"id":"d1ef5b77ec57dbf1455ee3a190981c51cf086470","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-17 02:40:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dd2fca03ec1919b00263715017ee5db55222f878","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-17 03:02:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3a6d9796066ea908b5a91ce78dcf48841c3c437d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 04:40:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29b9cb1088d4ec3a9c8c0d3be457b1f : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a67368d1f2a44e8926d6430734fdd98 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cec4b8539f9b437b8dcdbb135ab432e3 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfc7401e7e2244cab5e0eca4f5b9334a : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b76ce16f971c47e5bd905c34cb194b41 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c9e16256f54456598bceb970812ab7f : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fb85983fe6e4aaea5a0c81c13eba6eb : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac8dcf40a1d34335b6507991249051a3 : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3b3e04c9fdfe418bbb86abefae573b34 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/27a2d0a67ad04c6d86cf7df98a21784a : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/339d0525b7c1493dadbb820bc9aed477 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf4e246baad541d887110ad0442129bf : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b30dbb062cd04a789a775f9d01b8f0a1 : FAILURE in 19m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7efb7ee1fae04548b79697e7e785707e : FAILURE in 22m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1c7fa67b80649e79242870d8ca5604e : FAILURE in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/13b2df40466f435fb946e95fa34f3b43 : FAILURE in 13m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a42a9ba8e274d59bc82dd05896d092c : SUCCESS in 15m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5fd6cadc9bf4b3b9ea81372ef047246 : FAILURE in 4m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/624755a3dcda40afbde0a8ae33077240 : FAILURE in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/884079581db94b82b2e0c5f484024999 : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7c3aac0f2f244369b1f9cf9fb6ec12be : SUCCESS in 9m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/811a16095bb8426ebec5e450e2a44ec9 : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b34487d263fc443bb376a2ff210851b8 : SUCCESS in 54m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e163aec55d747b98ccdccd7272b1b50 : SUCCESS in 1h 06m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db8c251971104821b5788c3aa8bcdb2d : SUCCESS in 1h 27m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8f9880e892484f0ab4b86ce6a572a2a6 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":10},{"id":"a6681ac7cbe7c8c65b1652b3edea509833e5517f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-17 21:54:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"59a4f6a0623a160f1ca88b46adea8b35fc926cbc","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-17 22:24:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"17dad15d2efdc862d07ae6bce3b931352e4b9f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 00:45:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971b5f672be94b39a8611aa827c46fec : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f4142bea254fbc9a2d5d484cb9837f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c69b502538b4b438994e37bdda3ceb2 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f74db9cd228947b7aff954a906097440 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d90c1955e744582b3ee185db9691138 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/312df3d912af4fa2b5240be1d5238645 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77866c7ac9b7460e86091735a2e31a83 : SUCCESS in 5m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cdc018aa7e7748ef8690392fde697792 : SUCCESS in 10m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ca519057581402cbe31d3310e920959 : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4cfdd3e71f514e50a52b6a189ff99074 : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c33364c9781d454f92a8f46b8e51d5c5 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1072a4ea031a4cf39ac4f9ffd64ab286 : SUCCESS in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d07178449f8c443ba68455b16151e884 : FAILURE in 23m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/05ed9f3036c441618ab4233e711a7186 : SUCCESS in 32m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f5cc138068fe4d909d6baf8e85d5405e : FAILURE in 14m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4cfbfa8a21a942539cae345eba49e536 : FAILURE in 14m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/355bec8cb3ab43cba4a0dbacf19c19e1 : SUCCESS in 16m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/02ebc2f5532a4c44a3fccf200b5082bc : SUCCESS in 6m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d3b5db4c7fe34dcea01780d59382ca01 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ffec4de33ad48efb50947cbb6257997 : SUCCESS in 9m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/13d9919f88ef4b7cb77e60fe176fa95f : SUCCESS in 12m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d50e16a9a487490680d8de976f37a65b : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5650d2caf65a43b1a9f36c26d433f919 : SUCCESS in 56m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3fec9fb329f4c80921a772d240adc01 : SUCCESS in 1h 07m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11c37d50ed8c4ec7894676c668e99cff : SUCCESS in 1h 27m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6f413c4f78274605907bfb64f6a24835 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":11},{"id":"173afe9bfe43c1bcf2a355252fee35bc4f1ce98c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-18 01:27:45.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"98675b717d145b663723cd74cf47f523c5245e11","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-02-18 23:30:24.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"b9225b2748c305c3c5e486b9fac9bfaf40e5839c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 02:42:43.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc59ad861ee4c2097303da19b952055 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e43d8ed336145aeba0199aaba719f68 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/763edc672bfc4e039b279d119cb1b092 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/685cd4535a254a8989ec4f3ec540b012 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9b51ba1f4f0453c85738b0827b40d73 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06969c4295194125ae1dcb272ac7c0d7 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e7c77a07c564564b9f3cc8e27a20168 : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d6976c0116445f0b83cce2ba1a1d751 : SUCCESS in 13m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea80b4880aa344f1b77719d63605d401 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0116e4a1a03c40e58b97b9832d1853c3 : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6902e8c043024c7fbc86b1846faf6efe : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44653fc38bd54a56ba0174edfc641e97 : SUCCESS in 11m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c7de86fa19b0493db67af1e7a40af5e6 : SUCCESS in 21m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c798d9e30f34d298da676b283dd4cd5 : SUCCESS in 29m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c8bfc137411943e8b72f4e59f6fe361b : FAILURE in 17m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0041dd301e6d4e85a01e5bf4dc546ab7 : FAILURE in 16m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/590d9989ac1c43bc989c69a6f31aab24 : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d280f412666949869faf6d8b59c22a63 : SUCCESS in 7m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dbfcbc76062a4011a1f64d158ebfc002 : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fcc83ad5f6f44e0186bd4bd8ce2d259c : SUCCESS in 9m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6e949f4117da48f7b8d2e841b0b70cd4 : SUCCESS in 10m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c61df6483864d499eed69b0ffc728b2 : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1cf9e7cafd54ec78a358c76a39dd4f4 : SUCCESS in 1h 10m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71efe58860ec47f49b5744df6cf71152 : SUCCESS in 1h 24m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50f99f8b3b984b26b88d01f39f1cf7fd : SUCCESS in 1h 35m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d3549076380f43d88bc6504337d4280a : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":11},{"id":"2993de587049bdf48e765c1a12f17cbb1dfb3fe3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-19 17:06:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ed318d41066d477b0d91e5fdd8faaf8e7c85a6d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-19 17:28:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"586102d15453dcb81eec7f7f1071a53d0da96262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:19:43.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c5f02687584b39927ea728eadb29b0 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810a9f6a82a54d04a3727d9dcfe2ce24 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0875446f3364fecb2157a85b85651e1 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/744178e2d32b442ea4233ff025f3a22f : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67e7de94c9a44f90949b2a3e6fda3382 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a980280f9c094ea78184b1438d6e961e : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1674cdc806a6459faa3454a5049ad7d9 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/121920db17e249639e62775b9fac0e7e : SUCCESS in 15m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5035fb3375b04e7eacb8f315de917c5f : SUCCESS in 12m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e1c5201b8814b97bac01c70f0d85b19 : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a74aefa8adb043f79d126ef89e2d9192 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc213bdf5b994ad8becb2bffe34f2919 : SUCCESS in 10m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/254e6fe86c894036a06cdb7d3c4253e3 : SUCCESS in 18m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2f22dd4fcb3401693d2d2c712472232 : SUCCESS in 29m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28b07520ee144422b4dc120b6dc2f5eb : SUCCESS in 16m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0a3794f7f57444849e387e09b784d526 : SUCCESS in 19m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05a6a2fe150941a89d6eaf682dc2da5b : SUCCESS in 24m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/91d1bc42e1c949d681768d42624d73e5 : SUCCESS in 7m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/51424c891d184cd8b7457b7eb1de2097 : SUCCESS in 13m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/998cd01f434940a4aff7613bcbe43476 : SUCCESS in 9m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0c68323af8ee48fda629f93f59c9ab9a : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e023847005ab44719abe3a9597afd4e5 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbee715a4cbc4af6ae585a78dc6e810e : SUCCESS in 1h 01m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/13f5be55b00b4be7bb1cd071f2335167 : SUCCESS in 1h 14m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6402dfa35094f79bfb20771c4e56cc2 : SUCCESS in 1h 27m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c5aadf049adb42eaa6f8e2749bda4642 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":12},{"id":"f8a3aaf26bd7afb7480985c0360031bb1195a14b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-19 21:23:31.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"9f837deacad0d79aef2f3fa574f9617342f82b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:24:48.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"9c11f9bb94105e70fc45e1232efb44746798ed00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:32:44.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b7f555b4314a1984c541bfe2d12398 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4fbb60b69a048d596385f17f5787b71 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99630b6c044c4c0cbb35769dff9ca893 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4fc4f8554a04646a17daab83547bea5 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2ee5461b53e4d03a6e86f05de13ec03 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68843c6329574a0b988fd2f5a9317bd6 : SUCCESS in 7m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5414610d51f3486db0359c9de1291b89 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/592fd04ca1da447d9e09336091cc2580 : SUCCESS in 10m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/05b2f60a21b44cf3ba036a8acc55ec2c : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/758a85125a504e55a5f863ffa31cdf0f : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/130bc213180b41e286bfd48cb1ad97d8 : SUCCESS in 9m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/55302b3e50d840f6b197da10b3115a9e : SUCCESS in 22m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20f29ad4496b48f590f8ac5693e5b77c : SUCCESS in 30m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ff5a9a36764a44c695671b9e656083f0 : SUCCESS in 17m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b8d77818e33246029408fcbefe034843 : SUCCESS in 16m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01caf1382a0c4f009291c6b31b4bdb1a : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ac5e7fa517a7482eaeee0681db2e9117 : SUCCESS in 6m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/deabb58e6d4441d1a71c21e0f5a2ece7 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b185a2be14ce4d53a6f4a590733df6b0 : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce9e49b0532b4faebeea652791c8ea6a : SUCCESS in 10m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4f9991d449348f785b67d8ce283d6d0 : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/925adaef3684443d8e6754cff0fdfdb0 : SUCCESS in 59m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/780f66f7821e41089e959686b24c2258 : SUCCESS in 1h 13m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f19f65e401014f0496ce1740eb65f68c : SUCCESS in 1h 38m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1bb118ab7da34aa38da9134488783b38 : FAILURE in 12m 32s","accounts_in_message":[],"_revision_number":12},{"id":"89791a374bc40bab1b9816206084fb3768a6230d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-19 23:37:21.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"e8f67b2d74bc8e8ba16c6000141ce70e4135d6d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 01:15:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd14451d48d94d21ba247de933805455 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b8534a85f5456598a2cc09e91bd1b2 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa32ca5b31f748f8ae3b77e0706c98b2 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bd20be99d0647ccbf5daba9f59591d9 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3b42454acc742329076c11822c2ce7a : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2f876d37ad248f99b0cfe3eea751730 : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26c6d47f72524541a9cbcee1fea2cc7a : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6eac4dcf3835473886a0e864f9d42ab0 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1eedbd6b81bd4200872b13594322a7d8 : SUCCESS in 11m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/be783e04cea14bd6b258b419350cdbe9 : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/caf0b8b5e88a457aab326ed919ab15b2 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cee61e6e5bb0436abe72ba04b275fbf8 : SUCCESS in 10m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/700a9fe774fd428084534d87f548dfe6 : SUCCESS in 21m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/443f3508720d4bc2bc7cd1df3be9e16d : SUCCESS in 25m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ca93aa45c08419693ba95a344c9232d : SUCCESS in 19m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cbc39316c6b44d239df7d0b201cda1b6 : SUCCESS in 19m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea18a00ed4de42c99c1c506463e37c8c : SUCCESS in 18m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2dad890317454c95aa63a725f698d8cc : SUCCESS in 5m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c069fbeb20064b99aaff72d52fcfab8d : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a339bf3405d44dda6436757c08b7e55 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ca87c683c154419ea6f05ec3b448d3d1 : SUCCESS in 12m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0798711d76754ee3927c4bbad65512ab : SUCCESS in 18m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01b2c2ab9c784f1b90afc6c4d95880ea : SUCCESS in 58m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2299970a67c6433cadec31caa56ac068 : SUCCESS in 1h 13m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8f9991439eb4377b5c3ced585304e64 : SUCCESS in 1h 36m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b75a2a8b6ad496baa98c35fd7af37ae : FAILURE in 1h 19m 03s","accounts_in_message":[],"_revision_number":12},{"id":"7491e9f5bf5d8149b59a9281b2854d338a6e1386","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-20 01:27:47.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"3a03ddf6ed048e78cdb0add0d11dab8fd2d95deb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 01:27:59.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"753fde20a2404b113557bf19352f4b6110596d5a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-20 01:34:00.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"252885d15fce1c3da32481d6fe000232b4332570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 03:11:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c13738b915a445188fd6476f4a07be3 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288a3e75f5cd446baa46aed48057c861 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7ed697adbcd49d8a5c1249b44d33755 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de3003fe30054f899947f1bc9db6da63 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad5324c7a6eb48eaae640c0f75a6202e : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76d6c67a9b184690ae3e7c58180d0f0b : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbc30acfc9114851b37ea9eb56f8ebff : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e9bf2089e894911bc2d8be47d4467c3 : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7547c7f7ffd4e4a88bc53025b5cfccb : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/909e6dc7e15443ff80bd6681b63b1e8d : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9210f92ae90349daa6b30c6c326af6c8 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69104eb37234458b95728ff337220da7 : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/364ba6b6aac34fe58a72509286158fa3 : SUCCESS in 13m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea1883a064c9434c98877c69aa48eb8b : SUCCESS in 21m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/70462fae7f0d4f8b93a9353b0b9c4f21 : SUCCESS in 16m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/81119eb256664ad5947c052f5e9d0479 : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e6949ae3ec34940a6f50a5b8b178ff1 : SUCCESS in 18m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7c1eedbde2774a7192a9091cf03b4057 : SUCCESS in 5m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f9693303919b4ae5a425c613253ebec8 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fe99fbcd7064f758b4335939b719c8c : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c5797cf696942b19a968c82062969c5 : SUCCESS in 10m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fab3a156e53045948776b5436275c929 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0eff5d58fdb429cae5f59653c1788b2 : SUCCESS in 1h 05m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5705afdddcc2450cadd37adc4059308f : SUCCESS in 1h 07m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f2d358a906348f1a9082a7a1259955b : SUCCESS in 1h 27m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2603efd56227449c98208e154c23170b : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":13},{"id":"b94e945c3d184df1dd750a72f635cf31f4cf083c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-20 03:19:09.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"55a2e40d4950f821f9dfbd21f41ccd16d80dfcb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 03:19:21.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"cbc16ed41a2f21abaca8e13b7284867fc192bed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 04:37:25.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835f6ebf4e734de6895f33064e69aebd : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a4b5d89156b4b2c8bf3989652375fea : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/594434adcbf54e708942be161da54644 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c50ab6bf0f664c099dce6123fce55eef : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7e226c56d054d5b81eef50c790f1a0e : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/039ea9e2d17a47f5a34700a16a362a3a : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e25694b9c919439ebd974496eb480263 : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/410ada37a8da47f58cc7f5b73b8b00ab : SUCCESS in 10m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41d15f966dbe4d2ba13d174b35b0ca71 : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5141dec882424a3293dbea99a963c61d : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc2446a8d67b4432a1bea87dc1228125 : SUCCESS in 8m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0867502a119441048a2e034b02376ded : SUCCESS in 18m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b31e48b56a8e46fba927497df2b03076 : SUCCESS in 22m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55ad5a59f6e4452ba475118573b95e08 : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/de869b3d6761408cbcd39ab2355180e8 : SUCCESS in 13m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8988912add8347ad92b061903cd4e67d : SUCCESS in 17m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce5da367222940efac5388e5dd37788c : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d75238cfd1c842059f7b3a6c2375ff1d : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbdfa361a9324bb39e7c6c778f8e0c68 : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/125ff51ff78840689b1bc26877695d6d : SUCCESS in 12m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/820bf15ed3eb41218fd0ed43ddf51df9 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d0ae9231817469e9354e617c2a6b97f : SUCCESS in 58m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed41e6efa69f4b4b801180632c6d14e5 : SUCCESS in 1h 04m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be9df0aca22746479f5aba2876a09ce3 : SUCCESS in 1h 17m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/59aa3962704247e28e52effa6d73b1d9 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":13},{"id":"614e3d887e0196083906e17b396f8dfb2b6410a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 04:37:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"f6cd496a788ed615e3c04448da1c755b472ea207","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 04:38:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107ce81aaf36490c85772ecd131f8152 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"63aadd9472769bcde87b187ad490671a97c226a8","revisions":{"b157bccf8ce9ed265616247be0ec163b70fed7ea":{"kind":"REWORK","_number":1,"created":"2020-02-11 06:14:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/707064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/1"}}},"commit":{"parents":[{"commit":"64cdb32d77113c8132ba8290705744c4fca5565e","subject":"Merge \"Add py2 cap for reno\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cdb32d77113c8132ba8290705744c4fca5565e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 06:14:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 06:14:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b157bccf8ce9ed265616247be0ec163b70fed7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b157bccf8ce9ed265616247be0ec163b70fed7ea"}]},"branch":"refs/heads/master"},"bfb02d87ac33a7468cc663d64cb506738a4be059":{"kind":"REWORK","_number":2,"created":"2020-02-12 06:19:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/707064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/2"}}},"commit":{"parents":[{"commit":"452fc953f64103b6f0b6dd013e4d3dd5d291e5a9","subject":"Merge \"update constraint for python-karborclient to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/452fc953f64103b6f0b6dd013e4d3dd5d291e5a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-12 06:19:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-12 06:19:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfb02d87ac33a7468cc663d64cb506738a4be059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfb02d87ac33a7468cc663d64cb506738a4be059"}]},"branch":"refs/heads/master"},"9b55372c4daaa23a167774a93e555b738b080897":{"kind":"REWORK","_number":3,"created":"2020-02-13 06:16:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/707064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/3"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 06:16:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 06:16:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b55372c4daaa23a167774a93e555b738b080897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b55372c4daaa23a167774a93e555b738b080897"}]},"branch":"refs/heads/master"},"1f7c810305668e1dda169c0def7c8445dc2af24a":{"kind":"REWORK","_number":4,"created":"2020-02-14 06:19:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/707064/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/4"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 06:19:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 06:19:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f7c810305668e1dda169c0def7c8445dc2af24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f7c810305668e1dda169c0def7c8445dc2af24a"}]},"branch":"refs/heads/master"},"df091884607b321df926db5c650266729dc00174":{"kind":"REWORK","_number":5,"created":"2020-02-15 06:22:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/707064/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/5"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df091884607b321df926db5c650266729dc00174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df091884607b321df926db5c650266729dc00174"}]},"branch":"refs/heads/master"},"c8b877a80d7e79f04e713eb59254fbdf1aed7f49":{"kind":"REWORK","_number":6,"created":"2020-02-16 21:43:42.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/6"}}},"commit":{"parents":[{"commit":"b015a102c0fb3310e40394753c9baaecc43ec71e","subject":"Merge \"update constraint for neutron-lib to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b015a102c0fb3310e40394753c9baaecc43ec71e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-16 21:43:04.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8b877a80d7e79f04e713eb59254fbdf1aed7f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8b877a80d7e79f04e713eb59254fbdf1aed7f49"}]},"branch":"refs/heads/master"},"2967e35412be48035aaba41816badbc26612ac68":{"kind":"REWORK","_number":7,"created":"2020-02-16 23:33:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/7"}}},"commit":{"parents":[{"commit":"b015a102c0fb3310e40394753c9baaecc43ec71e","subject":"Merge \"update constraint for neutron-lib to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b015a102c0fb3310e40394753c9baaecc43ec71e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-16 23:32:03.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2967e35412be48035aaba41816badbc26612ac68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2967e35412be48035aaba41816badbc26612ac68"}]},"branch":"refs/heads/master"},"930c0ea56b2c2d887c7467c623c3bee79c59c48b":{"kind":"REWORK","_number":8,"created":"2020-02-17 01:15:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/8"}}},"commit":{"parents":[{"commit":"0a8a32ce5d54a09a2143763b07380a810a809f98","subject":"update constraint for oslo.config to new release 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8a32ce5d54a09a2143763b07380a810a809f98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-17 01:14:52.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/930c0ea56b2c2d887c7467c623c3bee79c59c48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/930c0ea56b2c2d887c7467c623c3bee79c59c48b"}]},"branch":"refs/heads/master"},"f4670e6ac936f92b1dc68ffc3e1ec40d67beccae":{"kind":"REWORK","_number":9,"created":"2020-02-17 02:40:34.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/9"}}},"commit":{"parents":[{"commit":"0a8a32ce5d54a09a2143763b07380a810a809f98","subject":"update constraint for oslo.config to new release 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8a32ce5d54a09a2143763b07380a810a809f98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-17 02:40:23.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4670e6ac936f92b1dc68ffc3e1ec40d67beccae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4670e6ac936f92b1dc68ffc3e1ec40d67beccae"}]},"branch":"refs/heads/master"},"789e3cd38c83bf0dd6cbf68a38e1104e1002f84b":{"kind":"REWORK","_number":10,"created":"2020-02-17 03:02:51.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/10"}}},"commit":{"parents":[{"commit":"0a8a32ce5d54a09a2143763b07380a810a809f98","subject":"update constraint for oslo.config to new release 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8a32ce5d54a09a2143763b07380a810a809f98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-17 03:01:31.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/789e3cd38c83bf0dd6cbf68a38e1104e1002f84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/789e3cd38c83bf0dd6cbf68a38e1104e1002f84b"}]},"branch":"refs/heads/master"},"23af4ea7543cfbc12affba5c2ba78db2413d3751":{"kind":"REWORK","_number":11,"created":"2020-02-17 21:54:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/11"}}},"commit":{"parents":[{"commit":"0a8a32ce5d54a09a2143763b07380a810a809f98","subject":"update constraint for oslo.config to new release 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8a32ce5d54a09a2143763b07380a810a809f98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-17 21:53:45.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23af4ea7543cfbc12affba5c2ba78db2413d3751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23af4ea7543cfbc12affba5c2ba78db2413d3751"}]},"branch":"refs/heads/master"},"b74a72813cc5bca974b084e92a58ecd894987fc4":{"kind":"REWORK","_number":12,"created":"2020-02-19 17:28:44.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/12"}}},"commit":{"parents":[{"commit":"0a8a32ce5d54a09a2143763b07380a810a809f98","subject":"update constraint for oslo.config to new release 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8a32ce5d54a09a2143763b07380a810a809f98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-19 17:27:49.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back mock for many test failures (being looked at)\nheld back alembic for nova fails (benig looked at)\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b74a72813cc5bca974b084e92a58ecd894987fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b74a72813cc5bca974b084e92a58ecd894987fc4"}]},"branch":"refs/heads/master"},"63aadd9472769bcde87b187ad490671a97c226a8":{"kind":"REWORK","_number":13,"created":"2020-02-20 01:34:00.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/707064/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/707064/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/707064/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/707064/13"}}},"commit":{"parents":[{"commit":"b5a1366cc45fa66d32198e740e091c3cfbbe5cf9","subject":"Merge \"update constraint for python-heatclient to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5a1366cc45fa66d32198e740e091c3cfbbe5cf9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-15 06:22:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-20 01:33:53.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back mock for many test failures (being looked at)\nheld back alembic for nova fails (benig looked at)\n\nChange-Id: I31e6f5a934005de3a0d6f21345949ad3c1c14171\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63aadd9472769bcde87b187ad490671a97c226a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63aadd9472769bcde87b187ad490671a97c226a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
