)]}'
{"id":"openstack%2Frequirements~707107","triplet_id":"openstack%2Frequirements~master~I5b24ad812c09c228eed1a51135555cb9de1fb3c8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5b24ad812c09c228eed1a51135555cb9de1fb3c8","subject":"update constraint for oslo.upgradecheck to new release 0.4.0","status":"MERGED","created":"2020-02-11 11:05:30.000000000","updated":"2020-02-14 01:31:46.000000000","submitted":"2020-02-14 01:29:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707107-1581643747035-b9790512","meta_rev_id":"9b8ed1800b5983ac80d472258a0e8b0e828e2c67","_number":707107,"virtual_id_number":707107,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 01:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-02-13 17:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-02-13 17:40:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 18:38:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-13 17:40:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-14 01:29:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"860136fe9f9bd81a7c6334c1a7d4962d9d1b2d32","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 11:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15671c3def1b09828d2eef8eba7556162fe54f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 12:38:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5531a2e5fc0476486b4761e7fb6458a : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735b744aa424480d958f859a59079bb3 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4648e48b1b5a4bd5b9e27ae8f03b1ee6 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc7c936397e842bd8bbf3204dca78512 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/874a82d49d1542d69245fa672d15b326 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74bb364cf69d44b0958be65e39528e8a : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0eae14337f84bba9a7408cd3a7a10fc : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d54e0f8a1fad49bca4f790b0669d3e3b : SUCCESS in 7m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/05583144a730482c91014f04abada031 : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/64fe1e1f5b9c44e8aa2c0c7ed8038834 : SUCCESS in 4m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b802df12adce42ca9c8e1b1742abd00f : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c917574d5fe40e59c720fd34acbee40 : SUCCESS in 7m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4731b8d5120f4651b86075dc79501344 : SUCCESS in 22m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d04a77b073024a08822b80ddcff758fe : SUCCESS in 25m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee60a810327943a1ab264cacbb5c902e : SUCCESS in 13m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8969dfb0cf3e4d5ea4589de51de50cbe : SUCCESS in 13m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53fc815405704c2c852a6fffda107731 : FAILURE in 4m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bb001cfbbf0242f599dca5141427b567 : SUCCESS in 5m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/556da71c0c8440a7946c66607230203b : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ea8aabda37e4d65884be139f9163a1e : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2b59005659594d19b0eabacef9eb53c0 : SUCCESS in 10m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6c8d362345314741a6f43d28426fad66 : SUCCESS in 8m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c23be8ae0fb40ff91ea103a570341d3 : SUCCESS in 57m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a7d274c42fc43a2a8942b04656e996d : SUCCESS in 1h 11m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2255b5d78fb14f7a93e10f2f6ee66dcb : SUCCESS in 1h 26m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e22f2674ffc24649a3dc27771387a2ae : FAILURE in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c7b18cb4a74cab09ebac0a1c888505c5b3d0afc7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 16:36:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e2821163a4a19a1f6c17421ff246ce90230a9d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:36:41.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6653dcff83b8ca22518d596ba6e9a0ee697e4cf1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 16:38:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5610c64e7c44f6438789b0a11fef04be6ff3cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:36:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39dcc3c45f9140e4b84a93dac9ba8eef : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702f5f0ee42a4506833e554358c79dff : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a34a2c7128d746fd97d0cc0c58c34192 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95f44c7ec29a47338fda4791cab4b26c : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc1f7b7db9e34e79bd67f305901663a3 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/407e34211c434e9290054be1026969dc : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f3766d4e3d4443eb30cf44624cb8db3 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f319ae69f48d4350bf9452e132a0ddae : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4aeae7cea3bb4f4fb98ea37c24b4c7e2 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4feabcdfcd764b23a10d464c424d080b : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d21e1115c8884030afcce2cbc6c21a58 : SUCCESS in 15m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1aa1094070e64e058ffdee6fd065a2a7 : SUCCESS in 9m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb066befbfdf482b86f844be780c0dba : SUCCESS in 14m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3c72cbf1f2b4c349ac947cf6ce437cc : SUCCESS in 26m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5d286555bf9428d9fb67c036ad87254 : SUCCESS in 20m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/40a15bd4caab44159fc2f3f76312d2db : SUCCESS in 16m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2bac00e3984a4c469aa4d24388966708 : SUCCESS in 25m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a3bded7fcfaa46b3b9e59c8c37cbd6fa : SUCCESS in 7m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/306bf8df47fb4b2d98d78dff238851bb : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c51e5a04a74041f1849751549381797a : SUCCESS in 11m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/03446de63b0e4b0dab69092fea20ee90 : SUCCESS in 11m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5375d1e714284f2c942d1e3073be4f0b : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d77c639f6d234fa98a6f34bccd2f1fa2 : SUCCESS in 53m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36e7bce257c744878266be969745a0a1 : SUCCESS in 1h 04m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc6105563afd4e3abd5797fa203b1b06 : SUCCESS in 1h 31m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ccd4af41e88343958e81c0a86f174ab8 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6e586bac0335d857f4100338c52fa12c405b4a1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 18:38:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9af7da6af04a43b6581327f0977f3518552e45ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:39:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f09ff8907ed90248ed1b86f5cbf264ddcd206ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 20:40:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824fa4f29c934723bcc718cf5fee6aa9 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98bf697933994e91a93673d82a5bd09e : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64f0149064c43688544c707178a3d76 : SUCCESS in 8m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ae59a4ad29c4fdaa70f4f547803d4ce : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/222f1ab9fe044bda92fdac9637a5d310 : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a92de6f55d6841c796658eb8de55c971 : SUCCESS in 7m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c0af6f460b14287b6e84c5982699997 : SUCCESS in 14m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ecd25450b25844d3af97bf8a8711143e : SUCCESS in 13m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f251e70ffeb440e9b59d3000058df93b : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6cd6b75ca59f4602b489af89e1dd4bd1 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11c87f968483450c879b2d9d6c5feb46 : SUCCESS in 13m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c27536c18ae54c52a3beecc3ea787f60 : SUCCESS in 37m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4bb91bc001744354acfcb442baec7ec9 : SUCCESS in 36m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7ddd3ae4e96442a689b538c035f07cd6 : SUCCESS in 15m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/54a8de9082b847d2899b635ac2ac85ba : SUCCESS in 18m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3b89e89511342bf9736e6ed5fb86271 : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a864bc6bc8049d6b1511090042522b8 : SUCCESS in 8m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/16c64010cfb4410e8e4879673bc47da4 : SUCCESS in 8m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a230fd855f94e1b9e33aa51aa04f1ff : SUCCESS in 10m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fdfa77a1a61042aaa0743be0146baf87 : SUCCESS in 12m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/688901adf30549779b9c90de705c7958 : SUCCESS in 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f084904327d4d18b9cc37ca0d0b794d : SUCCESS in 1h 10m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/725fd7825e224eff976402513c068147 : FAILURE in 10m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c97e1a25a853457692efb0cccab9e633 : SUCCESS in 1h 38m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1eef4eaf62cf42729c3e5017b7dee706 : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":2},{"id":"3ee31d42c71ef0c900c7322e4954c9db3ad4ad09","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 20:49:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d1421c73af15c1a758b4674b6bed31576c083565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 22:17:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649db28ba8764b2caa2fc6d7f917e4bc : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e7d9b2f29e6460aa168cd9cd9122312 : SUCCESS in 10m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff934436b6934817a91087bb950def1d : SUCCESS in 8m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cac67fb16854cd594967f22b6c44ee9 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cff96ad806c24558b306e83c682f5f63 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9137405e0dd24657a42b78313e20086f : SUCCESS in 9m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2af4b0c0881f47f1b373dfe63738dd71 : SUCCESS in 7m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b0f9b25e8604f349135a3525541ee53 : SUCCESS in 12m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c716e65261cb44dfbfb29f7474544fbe : SUCCESS in 15m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/446e0b8c7f6d4f0db73d164647e54519 : SUCCESS in 8m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80f2817d2c624bfba4a660f1d68a19e0 : SUCCESS in 14m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e45b6a58b834d8c8f76c3ddfc51b02a : SUCCESS in 13m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e67f9be911043bd8e492f80db5173d5 : SUCCESS in 27m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aa789de15c144329ab7bc45560b28805 : SUCCESS in 29m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36d5da204dbe46fda824fb75e14c6c98 : SUCCESS in 16m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/43561e0bc93d4b4e97c57b62740489a2 : SUCCESS in 22m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f071739669b44d91b9cb4759f981161c : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6c8bef4e4504a6da2db290b2bc4adf1 : SUCCESS in 6m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a923e2b3c852436c896af1a15220fcfe : SUCCESS in 9m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f2e2a8df9c345db81a854f8c8cfa795 : SUCCESS in 12m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a8b54b13253f44058e7a031efbdfe52a : SUCCESS in 20m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4605c33d1b9540b8b82f50799f8a2b39 : SUCCESS in 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6e3c573be1548caac32deea1f5ce23a : SUCCESS in 1h 01m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c21ff0a9b62945af898861295f43465a : FAILURE in 9m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/25f47b0e4de84830a5b47f808d8766b3 : SUCCESS in 1h 24m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b19ab483ed4f461391649eb8cd632716 : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ad92fdb69226130fc40c469ee166e258501ad2e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 05:02:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ffb7aed49b4501cc45e28e4ed4d316ee01b800f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:44:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6721f295e4e54e2f89c20bb04458b3a1 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f24d4a81164133999ca972ab6a12cf : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f3d82141104bf8878c95fcef864eb7 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e50dcaac2b0043a79f0a5f0f972a2aa5 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e4d21052e14474a95bf8d40c9ce87e6 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4b90d3b56924f809cba77b543ad026b : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f7bd48d68d74fb5ad6a39769fc5fb17 : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65faea78ee3945e4b9a8141ffc47252d : SUCCESS in 11m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc28923be93c43bb8c2631330b3fbc93 : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df2468098f2e40d18973fff73dcea72d : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4a0fdb10bc9459caec164ddb57ddcc5 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a65b130fe3294cefbac32e0e879ff03c : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e12076cafe704f1fa904d73668932018 : FAILURE in 17m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe4523e7fd7b40bd88a5dec612819100 : SUCCESS in 22m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8fd754c0998246e094c85dc7f88d4334 : SUCCESS in 15m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b9332e6a1515439b89e097b055847cc8 : SUCCESS in 19m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/839c69ba3b3e455a9f0183d3fbb50866 : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f8820d1b4e4f4b5297185a270c2921e9 : SUCCESS in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/14f8481502f44e70b5a8c6655b030f94 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ebd8696cd904f798ba4801dd3e0d623 : SUCCESS in 7m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1b2a3473ab743b585ad628abafcd474 : SUCCESS in 15m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74823daeec924434a0f389f0cbe310a4 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82eac2dddec34c339157253bca43f94f : SUCCESS in 54m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6e55c1fb2c24066b645a4e5b34e6c56 : FAILURE in 52m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b3593b3832d3438fb7c916d43d3d2caa : SUCCESS in 1h 28m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26e76f2c3808444e8e9e5565ee9ef28e : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":2},{"id":"c58b7843ae88b9bbc0b0cc47ffb6cb105b5529d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 16:02:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7420d2d113690cceec62566a3f1fb199bcccc91e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-13 17:40:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"878e621a2897712b37ffb965a75511e362016be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:53:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78405edc348466f8a470132bc26f091 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a7b7a9ea894a28a919a6d1cbb0e7b1 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d783e9ed932a44cfa309718ca3f382b4 : SUCCESS in 9m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8e1ffa4eb8b4b81ba811f48ba4bdf74 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/208bde54cf8749338fedeefbc20bd490 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e47a0104443406586887bc14e3a9209 : SUCCESS in 7m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/831564816568440b8751773e0790613b : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87486cd61c6043a7b8f7a32a02b8489e : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52812b6b4c96419db08f6429c774e131 : SUCCESS in 12m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad2e67cdb8234518b270993fbe1bf830 : SUCCESS in 7m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54af062bd6834c69819589c2e5ac1ab3 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e953c5c54e74067b22ee8bd620e35dd : SUCCESS in 10m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/daacf5c4de51476aa51430409c813057 : SUCCESS in 16m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66ab096fbdc44461bd22bf82d5c98f79 : SUCCESS in 30m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/02ed826a4a8b488ab8eb0ab60f0465f1 : SUCCESS in 13m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e3da86f3708245d1b833fe4713f9e5f8 : SUCCESS in 16m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d60bd7c4a832443cb4f37b3f82ccbe80 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/82661992093c45f2bec2276dee1c3e04 : SUCCESS in 7m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f16bf286b34b4b36bafc255852ceb1da : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b98101c5fe184d51987b74881be49d38 : SUCCESS in 9m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cdba073e0ecb41269f7f4860154df8e6 : SUCCESS in 14m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4203d7fb369249d69fb103447d6fc1f8 : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebf0989760ab4f14abef3c30b6514c5b : SUCCESS in 1h 07m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9460f2031534ad7b2999903e64a382f : SUCCESS in 1h 17m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6ba94a6a36e64a60941845f04965a811 : SUCCESS in 1h 41m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/00ca45a633e74133bffda2af0bd21bf5 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":3},{"id":"09a81a6f5ca458fe7c291c925a6958aff66fe3fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:54:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ce6070ff45c6cfb0dff96bedbcbf21050de0bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 01:29:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c9373860cb47b1864e54d650a56d83 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/495b766616ef4ab88821463da66cf787 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db62610b933488a8981ad4234545be4 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57f4042b2d2a437d9e1c4c84a396e203 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/758eaee800c64663867e5ac288732698 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85e535e85451499187c1546599b935ec : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72ffcfb5b14545938f4cd9d6b4ecd93d : SUCCESS in 9m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ca1b27846de43978a487828a2b6de17 : SUCCESS in 10m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59da1ab633504772922c9c696ff9f1d3 : SUCCESS in 6m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d9847c5eddaf48bdb8d79e6e4929d804 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6b8cabe6619453bb43080b25c640650 : SUCCESS in 9m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d5b8a006c654528aea60e703904c935 : SUCCESS in 14m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8aadcd6ca6944a518fb626f4a8c789ac : SUCCESS in 26m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/330eb36d1f99468aa4a5233a4202f65f : SUCCESS in 13m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fdf406e19197426c9e06ce5e23a21198 : SUCCESS in 19m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f0c8fca4f3a4aca8cc9ce1bac2df150 : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c33fe8f615d2465b8b32bb92b2f53c3c : SUCCESS in 7m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/96ba3905f97749afba5ba8ce69e29ec1 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7f7f5bd9aaf4548b9e8714e441bde0d : SUCCESS in 9m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bc0d7775d32742489497eaf92da15509 : SUCCESS in 10m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/47b4ed91b7004e1cace0d63e8910afea : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6fbff2d1e944cb18a9e82a444645897 : SUCCESS in 51m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3509b04f99da4ab0b01723d6f5957518 : SUCCESS in 1h 03m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ae9ad962d2de4c4faa69a41127418542 : SUCCESS in 1h 43m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/55ae2cfbb07a4c8ca43c68ce9134e648 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9d5442fbdb9b312bf712abe08c2747378c045d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 01:29:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3d850d14e338654892a5158882384b8613b99d55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 01:31:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4600c389e52466f8214f2bbc89fce40 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"61802625da05ff68ca5839ed14bf88cd17c8ec96","revisions":{"cf9e218d8280019c30618bc33bb8002b195d72d4":{"kind":"REWORK","_number":1,"created":"2020-02-11 11:05:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/707107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/707107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/707107/1"}}},"commit":{"parents":[{"commit":"4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 11:05:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 11:05:28.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 0.4.0","message":"update constraint for oslo.upgradecheck to new release 0.4.0\n\nChange-Id: I5b24ad812c09c228eed1a51135555cb9de1fb3c8\nmeta:version: 0.4.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I5958cea4a0cf33e0b2633118e9368877289eaeb7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf9e218d8280019c30618bc33bb8002b195d72d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf9e218d8280019c30618bc33bb8002b195d72d4"}]},"branch":"refs/heads/master"},"1fe680279bbf03c742ca785748ddde36ebc73701":{"kind":"REWORK","_number":2,"created":"2020-02-12 16:38:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/07/707107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/707107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/707107/2"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 11:05:28.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-12 16:37:59.000000000","tz":-360},"subject":"update constraint for oslo.upgradecheck to new release 0.4.0","message":"update constraint for oslo.upgradecheck to new release 0.4.0\n\nChange-Id: I5b24ad812c09c228eed1a51135555cb9de1fb3c8\nmeta:version: 0.4.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I5958cea4a0cf33e0b2633118e9368877289eaeb7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fe680279bbf03c742ca785748ddde36ebc73701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fe680279bbf03c742ca785748ddde36ebc73701"}]},"branch":"refs/heads/master"},"61802625da05ff68ca5839ed14bf88cd17c8ec96":{"kind":"REWORK","_number":3,"created":"2020-02-13 16:02:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/07/707107/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/707107/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/707107/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/707107/3"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 11:05:28.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-13 16:01:52.000000000","tz":-360},"subject":"update constraint for oslo.upgradecheck to new release 0.4.0","message":"update constraint for oslo.upgradecheck to new release 0.4.0\n\nChange-Id: I5b24ad812c09c228eed1a51135555cb9de1fb3c8\nmeta:version: 0.4.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I5958cea4a0cf33e0b2633118e9368877289eaeb7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61802625da05ff68ca5839ed14bf88cd17c8ec96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61802625da05ff68ca5839ed14bf88cd17c8ec96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
