)]}'
{"id":"openstack%2Frequirements~707142","triplet_id":"openstack%2Frequirements~master~I89ee7a377e3c436cb617e18b3eb8033de23143a7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89ee7a377e3c436cb617e18b3eb8033de23143a7","subject":"update constraint for aodhclient to new release 2.0.0","status":"MERGED","created":"2020-02-11 13:40:45.000000000","updated":"2020-02-11 20:51:42.000000000","submitted":"2020-02-11 20:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707142-1581454071195-159bbb0d","meta_rev_id":"e218623b0992d8f87748c6f74f984c4c250824b4","_number":707142,"virtual_id_number":707142,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-11 20:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-11 16:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-11 16:36:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 16:36:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-11 20:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90d6d4b8b2ce51781957490ea5a3162fccfd8033","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 13:40:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caec3e6b15edbec3c2eca59eb325fe703c9849b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfb211d745b4e2084f60e6e133c7c1f : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa773226d6ef490f99ddde7865dad6a4 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eafb7f0f6085431c868bb3f6f634a917 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b234cb105c8342a08e1e8b78c29e5425 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cae5eae4ba7b42a891b6d309dc4ce777 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c04378b698e7458095eba73cb80f783f : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a58f47fff4a47c39e99958630443fcc : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d48d458c776a421cab71cb78e5343982 : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/022f96598895448198323168e3db51e8 : SUCCESS in 13m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2da4a3d42b82485597f9a8050bb109c5 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8c691af013ae4b53b669882bbb468763 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b2e297fbf034a509017f9c2069ea14e : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f89272b133d41f3b039dbe96400d86b : SUCCESS in 13m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e1506fd5ee2e43f3b2f1df61a2468730 : SUCCESS in 24m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e7226282d3046a28f520651920736b3 : SUCCESS in 14m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/163460bbb5fc470dac18e47e58851659 : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a1836b17f20471996d2265c0e042f19 : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/585fe364cdac455da27ae3effe1c05ef : SUCCESS in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9a09b1af320f4121b3c1225c51588671 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9623de09d9644af78759f943569fbe15 : SUCCESS in 7m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/39fc8e91cd9c4b129ef241995e03c634 : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/90473dd1f21744048bca2fec13b04215 : SUCCESS in 9m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52c2e957ebad405da9a77c07461e7120 : SUCCESS in 56m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36c8d4588fae41ad8bf4f1ca4931bccd : SUCCESS in 1h 04m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/56f0932ebccf411db85e82d3034d9eb1 : SUCCESS in 1h 37m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3616ac2635104fdf89b3519750662647 : SUCCESS in 1h 09m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d50d86113fa9cb8e52f82b9a3ab24c1b2c414dc1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 16:36:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d282870b9327280594df8eda2fae9791d745ea1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 16:37:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b25367f69e9d4f125da37fad561e125caef13cb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 17:53:06.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bee740f133834b1a7bdc30acfb9b8cbff0aba1b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 20:47:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fb7ef21241449c972e6c6e18928b36 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35e685aa9f549aeac9274e461cddf40 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0c7ea35c9f4357aeff3e7bd556161e : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48a4e880287a4ba9b614ba03d9ecb781 : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/792c8911c8074acaa4a4e84bc50548da : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a2e96a546a04106bfc47074a7c36e11 : SUCCESS in 6m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65bac53f2a3c4da380da51be52dcb422 : SUCCESS in 15m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/43e45f5493294a1ea807e9d814a04b72 : SUCCESS in 11m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e09c3ea01df04609af924bdb6f1d30bb : SUCCESS in 8m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4568222d6de44044a37f3f7c03828491 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1746b1df64f4e35991546c339c1f9d5 : SUCCESS in 10m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fcad84da4f174d749be38820dec2fc12 : SUCCESS in 16m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f80d5c39bcbd4c459831a9c703bb898f : SUCCESS in 25m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3ea01ead013b45d3ad511373d199ce3c : SUCCESS in 15m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aa0195beca884e359eafbb36e586325e : SUCCESS in 16m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b98af84817e34795821cda54d8e307bc : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/94de96f55a1641f0a33527ca50507fc6 : SUCCESS in 8m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7e7f79e737484f3f8d69e725f5546bc3 : SUCCESS in 8m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0dd96f24a0dd45e4a70a0bb2031269a6 : SUCCESS in 10m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/853d2598565f4ee1bbc03d2ffd886cb5 : SUCCESS in 12m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c0c4b2e487347eb893324ef50670f78 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7af376dcc97542ee92244720f1f0edae : SUCCESS in 56m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c01735d28d74cdf976433db24fd7566 : SUCCESS in 1h 26m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c24c1d9cdb014a7497188d0873a6a043 : SUCCESS in 1h 33m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e1451fd6e15542b3a1d10a6ec3512f7d : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5a9c03a0bf49f3b11e031a1ca4aa38cbec95f9b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 20:47:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"07e6880db8eb834df0b79fe6073aa8d0b02656fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 20:51:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac76e1d0cd74815861e2fbb9d3980aa : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe4924077300f51cad6f635c908809e3a58fede6","revisions":{"fe4924077300f51cad6f635c908809e3a58fede6":{"kind":"REWORK","_number":1,"created":"2020-02-11 13:40:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/707142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/707142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/707142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/707142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/707142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/707142/1"}}},"commit":{"parents":[{"commit":"4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 13:40:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 13:40:44.000000000","tz":0},"subject":"update constraint for aodhclient to new release 2.0.0","message":"update constraint for aodhclient to new release 2.0.0\n\nChange-Id: I89ee7a377e3c436cb617e18b3eb8033de23143a7\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: I57b52a6c0abd650aeed5898689a2af801a44452c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe4924077300f51cad6f635c908809e3a58fede6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe4924077300f51cad6f635c908809e3a58fede6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
