)]}'
{"id":"openstack%2Frequirements~707146","triplet_id":"openstack%2Frequirements~master~I8b78c7f33a3cb7710b095f0c560e8170fba1f437","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8b78c7f33a3cb7710b095f0c560e8170fba1f437","subject":"update constraint for python-searchlightclient to new release 2.0.0","status":"MERGED","created":"2020-02-11 13:54:25.000000000","updated":"2020-02-12 02:30:02.000000000","submitted":"2020-02-12 02:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707146-1581474445233-9ea6cd4b","meta_rev_id":"9b6eef6fbda9e8f661fed7e217c4ed87865ecf46","_number":707146,"virtual_id_number":707146,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 02:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-11 20:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-11 20:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 20:56:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-12 02:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"158574782b54bdac10861f90c533cf214b3d80f6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 13:54:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"434378e323e9c9408d173ed60c5ebcafa68f0b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:29:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff35c25a60e4089a298f8e517976de6 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c5076b26d64b86b969bb6b1df7e91a : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88dce834e0034dd39994549229ca70af : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcc81a2c1d4b48f49ccf5cbd87f14d68 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4839a6d92c474370ad6682df7febcd3d : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d782d51637bd407f8324947c0b1cfead : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0238d660e5b4bf2b10eae78bcf1a801 : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d43e8a4cbe5847d6b684e9ec1cef880c : SUCCESS in 8m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/03bc3620c43e456b836d61c6508f928e : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8da118fa75d34bdba8a820b758d1dc5d : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c0add4180d8a4077bac1b1fef49dcd0d : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08fb10a6791c4771baf1d97adc222468 : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d9df22a5f7804244ba48fc3bc00dbf4e : SUCCESS in 13m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d1b9deeac6a4ecca6328d73d2dd9b5e : SUCCESS in 27m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0a97d5e44a1847c9882ee79f72bc5b3f : SUCCESS in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/be1f284cb1b04365ab2f871a85af68a3 : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3bbcbc0aec942098072d166deeb31ca : SUCCESS in 14m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3cf575b28530434a84be5618a1fd0c10 : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b033884ec49d4a9991dd3e229f73dde4 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43a11191c299477581fced6ab458eeab : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8c4af0e5743d4d3dbe6e1b0f0b896fb9 : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6b3ee45edbe2402f975d8e834fcfdab1 : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b86a2514d47f4099ab3c652c12564bed : SUCCESS in 55m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d144bccf7d441939118c1744c16255c : SUCCESS in 1h 06m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dfe90b625c0f4a0b8958ea70e3cea328 : SUCCESS in 1h 29m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/59fc821d147141ffa798716fc635dbb7 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b45521e9911a6e202a3cac0c3dfba4520d738faf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 16:37:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d23ce7d37277144613792b1a343cec5e9682074","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 20:52:35.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"282bc927a7730d610fdce05adefa42169bd1a610","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 20:55:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38f9cc8c33dc34d230cceb8361a408d9ca40e986","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 20:56:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a20f918a0cd40eadcf6acdec18822e92e7955d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:36:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fdf6d2e6cb64003bd671da1b54633e5 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29660d62df554f8c865c154378b440fc : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71c0d8eae01a47ef9f51ba10cd0340a8 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81fb844459e74df1bcf4438b3971d12b : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e56458d9209843818a5f3d90427478a3 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d362454293d743e3bb84abc7a8635268 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e16f1089cce4a81a3f791ab47607c79 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/78e62dcc8a1040309526efdb19323034 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7890d75e01f840e5a982c4ef5df5c84c : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1b69d362c6d40669fa05954aed73f37 : SUCCESS in 6m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6c756d6f69b45bc808f6e688c6b4a6d : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83200e81eb1f4e39b3b8b3de9b95db67 : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a2fffc6a1aea48b69157a9ffae135d35 : SUCCESS in 22m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e898d9852f104cc38262a2912e382b98 : SUCCESS in 29m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/15d704361bf0406dae8751026b7a65a3 : SUCCESS in 12m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bd3d55ecb7d84138b9ab31f569ac2c8e : SUCCESS in 14m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4032063e13d4ba1aa5a7433ba6e876e : SUCCESS in 20m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c7ac392b06e42a8b4b24afeda42632a : SUCCESS in 6m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f9817ecfd6ea4ed2a39f8535079c861b : SUCCESS in 8m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/893f856677724f8181f63c405642b53b : SUCCESS in 8m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0b79240d3b804402a889861ed7bb1010 : SUCCESS in 12m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13b4c9bfc56945a0a01339e52c59ea9c : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c067e18397804f08821154a14ecc572c : SUCCESS in 51m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74f4127e2844484f8064e2ee1c6c5289 : SUCCESS in 1h 16m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b7a62212fe9c4aaca32dc9288f74f969 : SUCCESS in 1h 30m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/586d6efdea294fec960098aa0a8b7326 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1062f503bd97cce68ad49f51f6229d182ab2f0e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:36:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c28d53192f1a242f5b012f98f8b97b605898afe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 02:27:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd9e243851642358f260b29de4ffc26 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce47b22f5b64010ad4e77a4a8844858 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b85624d5ae14ebe869414444aa746dc : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2f289179676433fb3368b02ccb6cae3 : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7a3d836aebc40b7b4a682139ba996e5 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b71552384a504c138b01c1f92a320b53 : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/25b488d121384d309d4b26ca235e6f9d : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb32846d1a794d39a2b4e8c6a36a7117 : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3a03a11f31544401acff8fe5c36adc41 : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/817b207afbc643928a988c4d6146922f : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a32ec98793a4c54a565bc47504ce60c : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f4bd07543464fde9649f292b2951beb : SUCCESS in 19m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fc9f2b60a7cc4a26ba3b71b574b0ccc4 : SUCCESS in 26m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ba92ae7f3364d488dae10eab353da3e : SUCCESS in 13m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/223983de557744888d01ec43cfd1d4a6 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/855899f7f9a540c7b4363f42fab0fca9 : SUCCESS in 15m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9aee35689d41484db492f412f0936ac7 : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f7c5e038b2a5483cbf3ec464be86d54c : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58bbb18ad4b340348a4f5ec0f00e1823 : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1dcb85fe220e44fc8f892f962a19c75a : SUCCESS in 11m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e41990103d234383832cb0599861a062 : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9100e0e65d584637937078c8bd8b4ba0 : SUCCESS in 51m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a5052fd55e846d1b7855f7686869f9c : SUCCESS in 1h 06m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/61171e53b1744abf9126b3193aee233a : SUCCESS in 1h 29m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/801a2c7eae8e4c56b857bb876d0b597b : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2845ec696307efc6aa565b3c2f801e7681a9fe52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 02:27:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0d4793444754d02a275dc616fc7d6b1d191f3cf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 02:30:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3849f9adbcc4a8fadc4e4c6efec1570 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"35da6bb3ac8e4086cb4a88d63eb91a74026df23e","revisions":{"5e45f0a304f1db1997e4e90850aa7fa8cda40b9b":{"kind":"REWORK","_number":1,"created":"2020-02-11 13:54:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/707146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/707146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/707146/1"}}},"commit":{"parents":[{"commit":"4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 13:54:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 13:54:20.000000000","tz":0},"subject":"update constraint for python-searchlightclient to new release 2.0.0","message":"update constraint for python-searchlightclient to new release 2.0.0\n\nChange-Id: I8b78c7f33a3cb7710b095f0c560e8170fba1f437\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I13ad2cfeef76ea145e49f9fd054a4dda9a11da68\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e45f0a304f1db1997e4e90850aa7fa8cda40b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e45f0a304f1db1997e4e90850aa7fa8cda40b9b"}]},"branch":"refs/heads/master"},"35da6bb3ac8e4086cb4a88d63eb91a74026df23e":{"kind":"REWORK","_number":2,"created":"2020-02-11 20:55:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/46/707146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/707146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/707146/2"}}},"commit":{"parents":[{"commit":"6aa1c0e332fae88d6ac28147e6b50fd70cb2720b","subject":"Merge \"update constraint for aodhclient to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aa1c0e332fae88d6ac28147e6b50fd70cb2720b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 13:54:20.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-11 20:55:38.000000000","tz":-360},"subject":"update constraint for python-searchlightclient to new release 2.0.0","message":"update constraint for python-searchlightclient to new release 2.0.0\n\nChange-Id: I8b78c7f33a3cb7710b095f0c560e8170fba1f437\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I13ad2cfeef76ea145e49f9fd054a4dda9a11da68\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35da6bb3ac8e4086cb4a88d63eb91a74026df23e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35da6bb3ac8e4086cb4a88d63eb91a74026df23e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
