)]}'
{"id":"openstack%2Frequirements~707178","triplet_id":"openstack%2Frequirements~master~I6c7b63b36fd730028e4cbb4b61cd0e9149c51f24","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6c7b63b36fd730028e4cbb4b61cd0e9149c51f24","subject":"update constraint for python-saharaclient to new release 3.0.0","status":"MERGED","created":"2020-02-11 15:42:55.000000000","updated":"2020-02-13 10:42:48.000000000","submitted":"2020-02-13 10:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707178-1581590236069-18a4250f","meta_rev_id":"fc9e414f951c25931874d1d8696a7eeb8b0ce0e0","_number":707178,"virtual_id_number":707178,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 10:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-12 18:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-12 18:42:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 18:42:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-13 10:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"507046976df9fed1d26d537f02f3e0a36a724d7f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 15:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fff22a85f8394e6a689340db780d8fa4649660a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 22:37:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a699934951f14ca49dd44c48cce64057 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62d327ac402d41999d476ac663dc912a : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b518979df4294052a44ab9368873d01c : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a2f3118b29a48c8a01ed4eba82a4c47 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b3cd005249c4486b6e3e2f0bef905bf : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b43c0535adc24f6babe290689259135e : SUCCESS in 7m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aeef64e40be445689c0ad7d9169f44d7 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/45ce3fdd6d7346edbd683934a96e5f6a : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d1cebc6e2a254d3a807f49223d6f20b4 : SUCCESS in 10m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2ab69a9a496e4f08b5530433a0487f27 : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a42fdcd2ae74d668e021e5d66adb8be : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdc2e72e02df42c684fda4c2b3220bd4 : SUCCESS in 8m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6366f1eb77004946921da3d8db858bc1 : SUCCESS in 16m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dd0958cac07841e8b0283faa4362ca50 : SUCCESS in 28m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3434aad9645549d8ba769317769a45f5 : SUCCESS in 17m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c98714233a6449aa90d00c48f412e1ce : SUCCESS in 17m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/350c90b156624c0a82222c3f8acce97a : SUCCESS in 16m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/392411b6c29346c294796bbe916b24a2 : SUCCESS in 8m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b0c4b19868cd4eb288b1e31575c297a9 : SUCCESS in 8m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efe4b987144844abb7637ddb9d822bad : SUCCESS in 8m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8094d111486a4232a53c4bbab2d8f678 : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/55bf8ff58afd40a3b33b1f15cad06234 : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76c8c19071da4bc98d38e7f0afc1e67b : SUCCESS in 1h 06m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d10ea2d48864da9aa2de312804ee086 : SUCCESS in 1h 05m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09bf28f9bd7347d8bec18e81a8d9106d : SUCCESS in 1h 28m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/80160103aaee4c8eb9a67804629c5aa3 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3fce55f3addd230caf35e24b2e4fc088892783d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 16:44:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ab6f0c1e942e1069eafd95e667352ee4ae41c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:40:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39d5bfc88ab424fbd2333b7a84d5c03 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c3ab0656b34dce85b2257b8a9d5043 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96ca4fa16ec440e09756c64371f1f888 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1dd2a710a9474eb097caf38b84e74482 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df92d0efe0e84d3cb4853715f29aac15 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c042a08ef2f54787a93539661096f735 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b5e48d8457940c09658ea3e789b8de7 : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/722171d83e1e43e48dbb6f72e6a6794c : SUCCESS in 15m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/44eb5670a712473a902d3a4a5491c103 : SUCCESS in 15m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d44e5c2e92d74161857d8ccfcbff6400 : SUCCESS in 7m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/19395c5c6c124877b94401026b335188 : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2337f7e2ba14d79b66bb40ce142352d : SUCCESS in 8m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0199ccb5d2ba444d87141716e383cf36 : SUCCESS in 22m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f899f1121334e68862dfd39bf958499 : SUCCESS in 24m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd40691053694fd78a827da1f4e19f62 : SUCCESS in 16m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fb26e96ed91e4b2abf773794d191d65f : SUCCESS in 17m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/159101e6a01945fc86ae2673999fca20 : SUCCESS in 21m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0fc00bc3c6234a448c47443e515ad163 : SUCCESS in 6m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5f6c477255014d8f9ee5a23aec6cd1d4 : SUCCESS in 8m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a28fb32cd6584472b1479759f5e19603 : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/957579a179ee4dddb02d5c96737533eb : SUCCESS in 11m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2bc0c050d6704689a8d05c12d26bad92 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/293a64bd95f147989928f1994c616943 : SUCCESS in 1h 08m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35856d9ab6cc498aa5066da6e064e091 : SUCCESS in 1h 08m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2807414e860b4f4fa09c21f660c30240 : SUCCESS in 1h 30m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8dbead9922b0420890eeb463c334f26c : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d6509d99ebb43411e0eae25c3abafa0665bee9c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 18:42:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0dd14e50a3a2bf5b55565631cc379acb87e0a030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:45:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a56b5f3048417c9fc9cbdb38addcfb2c55c91b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 21:15:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057cee737b47481a88960369c9344e22 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f1089265ec40889ca161596cb3d204 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c7a286e333426aa54347da8751e711 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/769655bcfb6a4c20a79da1930fc0bca9 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34d57c34296f4349a2ce5381ffbcdbc7 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80181242455746d1afc2c3e21a582bf9 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb639b01cd944927898c333df2cf0c66 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/da3a98bdf26e46f6ad31a603fa591ce0 : SUCCESS in 12m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0916236dcd954de3b6e2b274f9e87f0c : SUCCESS in 8m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/147afc7c628d4949a16f2b23f67bc324 : SUCCESS in 14m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e163d5954ab9475a8911880358cc56fe : SUCCESS in 10m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9fcb5b8bbaec4d8ba5a7124988ba7a72 : SUCCESS in 26m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/27c1bbe5cf78434090362e3f7f6910ae : SUCCESS in 28m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2de6337a1bf54b9a9289682ef32caab4 : SUCCESS in 25m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d25d82743b945d598c700402a234cd1 : SUCCESS in 14m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/840dcff1c378402883955408861d2691 : SUCCESS in 19m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c01e5045fb184667bfbecd412f0aa330 : SUCCESS in 7m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f552cdd61e054ca487a7267f635fd169 : SUCCESS in 8m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01ef5a6ca18d4e1cbb604ba19d96ee11 : SUCCESS in 8m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/42ff4fdf554243868588e7ffd48b01c2 : SUCCESS in 18m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/376e374b675e4b3cbb07f21f2fd26910 : SUCCESS in 17m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46e3ca9990bf45df8137d2f0e59a618a : SUCCESS in 1h 06m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c049ff1431d24ebe988192119a5e2c1d : FAILURE in 11m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b6c6ac97743b401691ec154369d2e029 : SUCCESS in 1h 46m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8ec1bbda1d254cc89d03ddf3a4a0b5f8 : FAILURE in 11m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3e30e54334cc08a6bb4d6967b8a3585e49f88837","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 21:20:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f1e42f413959edd889e516ad74e976a40778a014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 23:05:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd67530faaf4ea7a787dca8aed0a422 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f835396c71ff42eb98f161c501e90f9e : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39619d629a45406899e5e9bae75b26c9 : SUCCESS in 8m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f54270e72b84b6597605a62873889d6 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c0ca3e9591645cb96d8d9adb68bdf36 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94aa4fea98ec4108bb0ab553135baab3 : SUCCESS in 7m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f29c7c2e53e24b2fba607e340dd750ea : SUCCESS in 6m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/929e2b237e8a4ca6adc994f7c40a72ea : SUCCESS in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/27a14aa0431a4514871f01362dc53fb8 : SUCCESS in 14m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0fb335827674b9dbe256ff774f296a1 : SUCCESS in 8m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bf2e98d15b894f438e44997b8e1de571 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44604b015f5b425e9ecb892393368e97 : SUCCESS in 11m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da1ccf896113442caa201defb44817fc : SUCCESS in 18m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f99c29878784b64ba56f561f7890ee5 : SUCCESS in 27m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49a63df1a8ce48c9b676c50c7cb24d76 : SUCCESS in 17m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b95bc5016888435e8faf99d0d6e93aa1 : SUCCESS in 16m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/638cc93b08f24e2ea76083be0da64941 : SUCCESS in 20m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8c22a9efe0fc4a12a384cac89ec4d947 : SUCCESS in 12m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e0506c4d8e344df9ddeb23f8212af2d : SUCCESS in 9m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9600fb0e6454bdcae29847df38029dc : SUCCESS in 10m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/83956e4339f24ae3be93a196a32cbba8 : SUCCESS in 13m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a99cd1e613894244884f4463585de6a9 : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/433f215f51d64c32b8e106a5a93e64f4 : SUCCESS in 1h 06m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b22234534c874a8c971b8546abce566c : FAILURE in 12m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b731fbf7db814093a325828efe3038aa : SUCCESS in 1h 33m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2541243044734b6b8b933a0f81ab97e6 : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"4a2e1dc8ea522bccd067f025bc6e6176a471d773","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 02:41:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d7bced1e42c868f454e4bf6e7a20f28c1ae7ab41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 04:27:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d720b1c4084fc7bccc4ecdf2455b06 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cfbc87849de45c3b8110c29a42a054a : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21b0ea0a2e274ac5b2b600bff4b0c2f7 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74788d9bced14462b2905d53c257cffd : RETRY_LIMIT in 1s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/129158e1e8f04571af8f96287cdb724e : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/556ba846bec94622a13c554eb4788dd2 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b88fe8250adc41cb96bd03124b2582c4 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4e9f54c4f8c74df2809000188998be92 : SUCCESS in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f7f037e1f78149449a18fdbf7743a24b : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/62e7182f65c74ca2af5e03b15d2e706d : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3afb83a9b2304f0884d3c931d13f78f9 : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf9927fb20a04773b1b09cf34d549eb5 : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aa7f210c7b1c47e4843cacbfb4529c70 : TIMED_OUT in 1h 00m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2a43249443b243f39d587d8048806a05 : SUCCESS in 25m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/917d6f64e0d34ac69fe570e36e49f218 : SUCCESS in 15m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0faf3745bdd04afca009ac79a5af7e46 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42ae17b7076746b09626ef20aee78dbf : SUCCESS in 16m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d8c16b6a7d50490f89ea304c49488acc : SUCCESS in 4m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/34afc039fee64e199e6abc9e928ed437 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2dd05298bba4ad7b7fbee7d1e1edb5b : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e2669278ec0a4b07b149bff195d15170 : SUCCESS in 12m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d55d8b772fc4df7ba3a58ebcf69c2e8 : SUCCESS in 19m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61025003b4f7464b847f17adfffd970c : SUCCESS in 52m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4502b295c5914689b50fa9393308f15a : SUCCESS in 1h 12m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0dac6b7d363c47d9b42c45edbd412aeb : SUCCESS in 1h 40m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/43269f2f46694dd4b7a3d84a74992d6c : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7c751209e3d4ba90cd0c39963bade17c3f7dabf6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 05:02:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c15b7fdb13c4bb07a1f0ec7f0366be8408f9b434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 08:57:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e757cd57e9b40469c01fa1f18e977e3 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2238bbf9284d91b969895efe587d31 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3c8d948434f4afd8ecfc229131ea410 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87bc37167f5d43f793b48544db4dad34 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dde164de23c4771bb84a72c8bd4d531 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85dbb86b696049b09b959bcc252ee12d : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e30f4a9d540e476d82fae9a4ad22cd38 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/404633fd373c4607b03f5ab77de6a792 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/501c00644cb04a12a009cac78a71af90 : SUCCESS in 10m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a495ba8fc6b743ba8765da83a5e9fe6f : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e8975b9b4e24648aa2e7b1800094737 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5344c71684f14d8087f31b744cdc51e4 : SUCCESS in 9m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93cf73c4708148a99ec3a60c01c19a10 : SUCCESS in 13m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/46d6c177b9ef4c2c878b7cffcaf3aa74 : SUCCESS in 25m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e1f92323bf9446ccbd881cfeeeb49ed8 : SUCCESS in 16m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a691c109031840f9adc7b3078d660ec6 : SUCCESS in 16m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d60a89e2daeb4f9e87f601e35b1e995f : SUCCESS in 20m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9b35fcc61c1b49d098727cd7f9b88d0f : SUCCESS in 6m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8bce048ab7a649b9984f6d978b147284 : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f88c990c8df14514a51b3bfafc699648 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d65b4b3830a44fa8324276d06cf4656 : SUCCESS in 12m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c36d0368f784708b0a34618edb59b0f : SUCCESS in 11m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1f3d2e6b77c444aa7c3ffdf83925d10 : SUCCESS in 57m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3ad64d2f2204a7895da4246d1caee75 : SUCCESS in 1h 09m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d773297c631d4a97b63f95b20ce51c69 : SUCCESS in 1h 44m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/59ac328ff3974558b227cfd3d3b77daf : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"7c1cf7fa284328dd27763c9dd26b5bf7d8154e86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 08:58:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ebcb860e7335b3b9e56d9c3df46191d5968af0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:37:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001feebfd7e34a18be68acc2cdb7ed4d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fd1a3411de40b191e219f28f60aefd : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e2d87fd224c4a5098796baa875df9b8 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ae05c7b7553472ab833637b201a8c8b : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfb2bbabe0c0437fad24f549b7e9d687 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd003e302aeb443d8fc71241659aeec0 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c0492445898487c8ac0bcfaedef36a0 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/02ec1ade748549c5b46f6b95b175bb6e : SUCCESS in 11m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f206e9b2076141098877ef5ca5f98215 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22998a6047cd42c196536ba738178565 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73bb9d91055a4aeca8bb5db63a325f92 : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0be2d5572593412ea03fe93f6a47e05e : SUCCESS in 19m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/60f757578145410bad2972177369f4ad : SUCCESS in 25m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bdda49ed06bb4a2ca8b5a1bfb746c549 : SUCCESS in 13m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5ebfdba90c2248599c1b8735182f68c9 : SUCCESS in 15m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d8bbb13b12e499791f8897ee8b4acf2 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bed05aacabe34689a0cf7c780cbbfa6c : SUCCESS in 5m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/adb7b99782eb4740a5782bcb17be8246 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/196671b2e4a24d629021fed2c9a15c69 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90ef0e9fd5634ab5803043210f706fd4 : SUCCESS in 11m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73619b91183e46678596b72612968767 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/075d1f0a78894141877487f2fc8959e6 : SUCCESS in 56m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ea05881fba343848539257d506d1056 : SUCCESS in 1h 15m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/380b36c23bc0478896e5046b3174dcd8 : SUCCESS in 1h 31m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc4d2f8a67d543958608dc4719acb739 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"878277c7ca817847ffd46abae2c7a379d413658f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:37:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2703d4a39a7cfaa7e7a72b6f70edadda06c99f87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:42:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc03f6587b804d0e922f9f23c28c7f79 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"741873d53fd43e1173d2d74d866c6408a433b13c","revisions":{"83b0063b253566b8ffb980a4d4271c29671d887a":{"kind":"REWORK","_number":1,"created":"2020-02-11 15:42:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/707178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/707178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/707178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/707178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/707178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/707178/1"}}},"commit":{"parents":[{"commit":"4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c42f72f4b109dd83fb4e4cdf23c8e0fb2fca3a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 15:42:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 15:42:54.000000000","tz":0},"subject":"update constraint for python-saharaclient to new release 3.0.0","message":"update constraint for python-saharaclient to new release 3.0.0\n\nChange-Id: I6c7b63b36fd730028e4cbb4b61cd0e9149c51f24\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib11a35a2434d821fbe66169e68f1c692ab537230\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Telles Mota Vidal Nóbrega \u003ctenobreg@redhat.com\u003e\nmeta:release:Code-Review+1: Jeremy Freudberg \u003cjeremyfreudberg@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83b0063b253566b8ffb980a4d4271c29671d887a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83b0063b253566b8ffb980a4d4271c29671d887a"}]},"branch":"refs/heads/master"},"741873d53fd43e1173d2d74d866c6408a433b13c":{"kind":"REWORK","_number":2,"created":"2020-02-12 16:44:01.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/78/707178/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/707178/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/707178/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/707178/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/707178/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/707178/2"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 15:42:54.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-12 16:43:54.000000000","tz":-360},"subject":"update constraint for python-saharaclient to new release 3.0.0","message":"update constraint for python-saharaclient to new release 3.0.0\n\nChange-Id: I6c7b63b36fd730028e4cbb4b61cd0e9149c51f24\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib11a35a2434d821fbe66169e68f1c692ab537230\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Telles Mota Vidal Nóbrega \u003ctenobreg@redhat.com\u003e\nmeta:release:Code-Review+1: Jeremy Freudberg \u003cjeremyfreudberg@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/741873d53fd43e1173d2d74d866c6408a433b13c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/741873d53fd43e1173d2d74d866c6408a433b13c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
