)]}'
{"id":"openstack%2Frequirements~707253","triplet_id":"openstack%2Frequirements~master~Ic2d14c94c2c434d228b85f8cd8515ee1f90d0fe4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic2d14c94c2c434d228b85f8cd8515ee1f90d0fe4","subject":"update constraint for python-manilaclient to new release 2.0.0","status":"MERGED","created":"2020-02-11 21:28:41.000000000","updated":"2020-02-13 21:54:04.000000000","submitted":"2020-02-13 21:51:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707253-1581630699319-bbef5f2b","meta_rev_id":"766023a67280ec41df2f1c411d498d3a9b809348","_number":707253,"virtual_id_number":707253,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-13 15:14:43.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 21:51:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-12 17:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-12 19:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 21:37:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-12 19:07:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-13 15:14:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-13 21:51:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8e3ff30e832cd55506fe6cf4653dd3ba36e37b0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 21:28:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f462768b1b8a1f394fbc8845771a6dba43959f8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 21:37:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22ce75376392cce4f44dfc7fb28e3b49a6feddf3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 21:42:46.000000000","message":"Patch Set 1:\n\ndropped py27 here too","accounts_in_message":[],"_revision_number":1},{"id":"b9a27cbc8617e4c9d44cbc216e5d2421ef7d6c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:40:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4baffe8c29c4fd4bac2aecc3a408813 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b69364c6be3451584d2eccd035f01d0 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2afbf596a7be4f2ba1327c91c111831e : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8e6ffc216ef405f9c9dafcee0ff603f : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b697dff0d7834e9cb0d556a2c91f84fe : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebe10cab94e340fea949502487ebb180 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89850d8822664eec88e753d8d483e47a : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/879c5ff165664c87bd3271cc317ab028 : SUCCESS in 10m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/160fd7c2478d4ee18dac89d8578dbfbb : SUCCESS in 12m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e53acc9675a54efe9c56fa86a0274645 : SUCCESS in 7m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/78f518251adb4cd9a931b8b3cf2cb15a : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85327167de6c4b7bb78e97aaa8da30ce : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ffc8939472624d2d85486cd263bac1a9 : SUCCESS in 24m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d3fa8d33a2249018038c614a2ce4310 : SUCCESS in 30m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ed79c17054c4e18a3530161b9e8bf8e : SUCCESS in 15m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2787b3dff6be405ea26fab87bc1fdde2 : SUCCESS in 15m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08edfc67d19447d69e9a4b4204691460 : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aff9d38d9f00490ba18da9254882a899 : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/84b703ca44544599b890e9a034a6ad33 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f009a53889349e1aa764507d99029b4 : SUCCESS in 8m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c0eea0cc6b834d26a0868697ff9a4902 : SUCCESS in 16m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/df343bcd33654de3a91d920b783549bb : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c84d3506f0db416d89614fce6246901b : SUCCESS in 54m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1147e1b58c4547b4a174f7098466c745 : SUCCESS in 1h 13m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9063abdbb0a742b4a1816b5f44e7503f : SUCCESS in 1h 26m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/75e0c699aeb24f619dadff03f33ec8ff : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1a93936bc77952385a60226aecb8b5679766de49","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-12 16:38:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa1b1815406ae0c4db30f055b7f1bf538eb331fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 16:50:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ecf93e33d7ad9daf6bde75c466354d32b22150d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-12 17:01:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9845ea15d895a71402de0c98229e4ade3cf59acb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 17:01:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"00352cdd2cf0ffda74d8a5cf965577ef2a9d16da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:56:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186c1258318c4a15a7697eb303502a4c : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9fe8650c89047bc8d1319221452621d : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/657cc2e3cc954035b0727a2c31df6a2b : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c26c0fc5fd6418196fac6e1e9548a77 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3d0c98b89a24cc49e2d2752ce50c985 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2735f0b684b44918abb2a7e4750d9e3b : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfacf77d000946b2a1c88b4321d6f8db : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d67b8406c8b6453ba3b8ab36e9c48535 : SUCCESS in 16m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/62618a4a9dfc4098a65b8d56368c6938 : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1948a6f9401949feb6454028d7196739 : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e256747e23474621a612cfccf18f7aa1 : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d550a76fa7a4ae1846bd11b699d9639 : SUCCESS in 17m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c98e14fbb4f6488691c0ca069fbc5df1 : SUCCESS in 18m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe04363b318349d6aa6f4af1d36292f8 : SUCCESS in 31m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a02aca0af5d34a60914b0c787f0079b4 : SUCCESS in 16m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/551401356afc41d4913dcd934dfd9349 : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9cb928b3917c41b69a1da415bf24d8f9 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/821694904ac84fda9f190c97f03c001d : SUCCESS in 7m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/96196e96129c4d79ac3864efa51fe4bb : SUCCESS in 8m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0582e388f0384714922b36d0b495b697 : SUCCESS in 8m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ff33ecdce38545009ac5b78c7abf60eb : SUCCESS in 13m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/324207ab3fa4444fa6d8e2a4539f1ee5 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13482577a2124933898f00b3605c62b1 : SUCCESS in 1h 04m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a04795ae6694cb8ab062164d0c706ce : SUCCESS in 1h 12m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4a57417f79347c39a634366d9871c7c : SUCCESS in 1h 25m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6671602adf51456dadb2c44c8d406184 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":3},{"id":"bcb08864ca0ebfbd6406299f1e0aa0791e1ac650","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 19:07:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"65a667b494871f3e78259a992cc5513758d27934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:07:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3eb7fd2227d051a9f464c12992d1d081e494ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 21:48:04.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1066c805eb747a29f4c98cd4491a956 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2010be4d494ac19630ea6911d5de0e : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bcb6a3b4070450db716d5a4d29110b5 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a44a8d61f9f41f4aeea45197d3c12f4 : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/120f1208ad6b4a688513a2e1bf9625e1 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d16097431ed4299be651f9f034acf58 : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fcf296ab46a342e994b28d9fa2c39240 : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a1ebfebfba34e798c70e076130c0fa4 : SUCCESS in 12m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc85fc01e69f42178960c36a2f94b4ac : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c01f7dbf7b94bbeb513126446031157 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95e1238cbdbf4670bc3420b96a7d3d97 : SUCCESS in 9m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cd0fd36cec3f404387a02e5e159b05a2 : SUCCESS in 23m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d9b4bb36bf5d42298f87d4871ac5ce42 : SUCCESS in 27m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0fc2d2ef7c14b749b5ea36f45a03836 : SUCCESS in 14m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e14d7a0e48134dcc8d442fd754359390 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e7cda40ce044ef6806fb95330254846 : SUCCESS in 18m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a27dd90904ab4a96afb1d9aecf174829 : SUCCESS in 7m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d749c4fbd52641349079d0830718daac : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d0330f00a1c4693b61611b734a262a5 : SUCCESS in 10m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b5ab35d08df044c6b2e3e69cb6d12c96 : SUCCESS in 13m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dcbeea41f5524cafaa41b0fcadb8bca6 : SUCCESS in 18m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7dedc610fd142c3953b3fa3fba1c47b : SUCCESS in 55m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62f46604be9d432f8a5b12d146695fcc : FAILURE in 8m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b5ec2c0eaf514cb08a72ca3dd42416b1 : SUCCESS in 1h 37m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee667d60cf844f5681b4be54709cfeaf : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"5ca2248f6b9ef8df82ebb2774e04c41278cf5fd5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 22:09:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8c96251787a7df484b11b14fc05d36b1a1887efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 23:55:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2678ff98f4e04b2cbe2e96c3d5aeb245 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7d58fdfef14b88b255b3e44b3f5451 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c9e1461e35d4516878fbdf915c85e52 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22519241af9b485b8199f2d7205410fa : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a0c30bae1c47089e431d62decb0a68 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/051f33bc462c47879dc84e30a9a84b52 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa98ed55874a4afabc1ea724e4183f1c : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/74b7bfe815294967aa9b26a9f0f6e35d : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59ca7e3c8cbe4c6ba458518fbd88fc25 : SUCCESS in 11m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ebbfd6f2797e46d08da7caccfb9e0978 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2a970f5202324c6bab4103fe7942c8a2 : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c67f4ace1924c2fb5a16e4e33e5a3c5 : SUCCESS in 10m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96df9a1953ad48dfa12d59d02ebf4c8e : SUCCESS in 16m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ac78706c1b824b1da47a5a3232d46ede : SUCCESS in 35m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c5386d009d949148ae5b9cf1e904c59 : SUCCESS in 12m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/43d945e666bd41a5b97731d3f6e66123 : SUCCESS in 14m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/afa76c74dd11413cad05c260ae742c21 : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7329e0f514d843ca8943942ea4c45288 : SUCCESS in 8m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d76dca82546d43ecb7472a4192551aae : SUCCESS in 8m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46f90db4050f49b5bd2b6d0ba6fa33ea : SUCCESS in 8m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d3645cc1ff841939e9d4337efe84990 : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f3b781f6b83344be9c8f2cbb4468023e : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a718370a438548d3989f9166f59ee9e4 : SUCCESS in 1h 12m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ae010f5684a42d1afd45f61a829d553 : FAILURE in 11m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6606515a4e1f4c34872f595012a4cf0e : SUCCESS in 1h 39m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf07c1ef772c4135b925e9a54b0789e0 : FAILURE in 7m 37s","accounts_in_message":[],"_revision_number":3},{"id":"583787d48730e6c8170c7426eede77a3d3b0d876","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 02:41:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4dc324076db0ac169bdb21c4290b6ca27338dca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 03:55:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b0acef59404a019c4f9789bc6a708b : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9aa819ad8c54af69f212e494a6a5c31 : RETRY_LIMIT in 1s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e451f5094b845f689a81f76811ec22d : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/505999ed5cd34755b174b2ccb26dadfc : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66fe4fc00f44437aaea18f16ff58c7cb : RETRY_LIMIT in 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17b6f395342b4de2acc6ff59f4c3ddc1 : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/020dc6f125a846d7a96ae08502307f5f : RETRY_LIMIT in 2s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1545c35b9d4049e7b222642dd7da758f : SUCCESS in 8m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b66bbf5d26d640428453d5af332d0413 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8c639ae758a9454ab8d9b79e745d93ba : SUCCESS in 6m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48914e6a8da34a7c99f837fc986443a2 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba4d2fda38574ec2b20fcfbd045c97e0 : RETRY_LIMIT in 4s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0175a3b0170b46f881be60bfa757d1dd : SUCCESS in 19m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a987f308f034b548f407ff8f7586f29 : SUCCESS in 26m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/537714505b8f4bd6a6e82d97371515f1 : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ee14d415338d478b98b381817b8f45a0 : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7410d9b0005469689d40469864dd505 : SUCCESS in 15m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3da949c399d24d1b96ecd46862b2d5b8 : RETRY_LIMIT in 1s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/562e203d0eea49fa9d9a035b0722d0b7 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ea1394fd3d14e7286edd9477b67d882 : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/da3dbed89fe04e0e96c256e98dae27f4 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1af7ceb0aa58406b99383583b2abb318 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b4aba676d134505b2543cb5ee423674 : SUCCESS in 52m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f64a96ef6da342dbbec77ca4df5e922f : SUCCESS in 1h 07m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/187dbcc6bf834ab6800bb2279d0339d2 : RETRY_LIMIT in 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/31f0a2cb141b4539a140e459c26ae456 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c0ca4bbbe40085a22dcc43efb2cd1b4b4fedc33b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 05:02:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b0c92c006a5d3cb9af9031c8372cfa33f750d648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:35:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e210bd7d9a904dd9a325b33befaec239 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b0c00fcb01409f9021011d245f503c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/002ac665940b404f812e413095940ad6 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e19c2ef0de5449f3b26c5ee5d7dacef7 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd1b5f37d4244bec89fbf892d6528bf1 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59e00cca63fc418b823464145e4e7aba : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51fdfa12b5294d25b8c919e123d159e8 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88d81c13ae48425dbdb7edaff7c041be : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bed84262691b498f9cdb4afbeaafc392 : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2397c6f49e54493d8956f5cbffeaae5e : SUCCESS in 6m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63b0cfde3e2846188ae1e10f3505dfcf : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa08f474e0e546c4864949197f973ead : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/21c4c5345faa4b5096f53908bedc8650 : SUCCESS in 18m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea5d4d8d2aa94983abda9b7b78dfd6be : SUCCESS in 22m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b699968d3bf4992b6e8cbc6b7c720ac : SUCCESS in 12m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0e49d8f8549442b68c368fbdb47287f9 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e6a6c26b44a4863a539f777ee6664fb : SUCCESS in 15m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4e1845bda75340af8daba1d1ed8c971f : SUCCESS in 5m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/926238f791b3412c951814a38e30dd27 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2fa4cdffda24ab192eded710ae09cc5 : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5635887bf3f5495981674348a7956bd5 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/70786504ec184f999ae64c5d537f63b0 : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f61e25a6a2c4a8787cf9184381648bc : SUCCESS in 53m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/382aadbb521949ba855baca5417874b4 : SUCCESS in 1h 05m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9aec143d840747849ec163c599a2a8b9 : FAILURE in 1h 29m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2eec2a83392842de8651f1e8ecdbc006 : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":3},{"id":"61a9abc8aafd7bc1e9fea1d707a369e6d35c777b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-13 15:14:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8e782fc9365b1b49f569c50bb956f99b433a2254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:47:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f79a51a83d4f7e8d2c4f60c862ee1c : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a662e78b29fe4584ad5ce3cf47258311 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94eb33f07e1142db97524341c3dca048 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f5d8159d8a64c3ba8cb691fa87f9400 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7f787cfc733458fa8d22c41541f60e9 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffb637012db94d569e1fc84d120f588b : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ed600c6408543e9a2799eee628afb82 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/27b64c46433045949d188b0e0aa8a5dc : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abe0321e5ece44128bc9e741fb1b734a : SUCCESS in 12m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ae786720d034ee4a07ce43ddaf46399 : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3148a11f5cc846558068542894b8d491 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cfa41988a6fd4acda5dae0601b5008c2 : SUCCESS in 9m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42b19911b92347ad9dbb69f2418f3abf : SUCCESS in 15m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bd9a13a3df4d48af9ed19cf1925ebca0 : SUCCESS in 32m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6aa57d0cbc894a82a1008841da8bce87 : SUCCESS in 15m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c89a6aa009f419ca0428abfd9e029d8 : SUCCESS in 16m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f802a4dd4c944897898c481ec5190910 : SUCCESS in 16m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f820c9a0ccd54ef58019684e5f45b971 : SUCCESS in 5m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9df41bd95db54f61a0c197ac636196a6 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/707ba8ab46b74e339d0c1634284ee990 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/08a4eda1ebb0484a81ef99fdd6ed7b92 : SUCCESS in 10m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f993377db6314238a4964a280e92d5ab : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8457045ad9564d8a880afa1a37dad5f9 : SUCCESS in 1h 02m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db5550552cb74796960849be320b39b9 : SUCCESS in 1h 10m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/411b9e2a20834eb2ac2c8991583c0edc : SUCCESS in 1h 27m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f5e88282972c4369b41019e8b09bd901 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":3},{"id":"9915f635d7db9927a76bafa3753cd4681abb154e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:47:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3b173bdbeac92aa6c3fd04e15d810737703a46db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 21:51:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ead1af60ac40b3e0d55f6c026922d56b4c0bf86c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 21:51:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/911588de30424d5e8d99ea63b60b9c41 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050c6cf6bb3347d2b0e72286a7565491 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a12b19745af7467ba8b69d470d0ea684 : SUCCESS in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef3c17b5e6e84ceaa05c0e255a431f70 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e64a382ce740423782679e5f604b4dd9 : SUCCESS in 7m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/265d6dc596f544cbba1365a5b1482032 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ef64e3917f8449d8d9e5812aa169b64 : SUCCESS in 11m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4af4f35d76c5476a92194344bf32cae5 : SUCCESS in 11m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/361cf703ea2e41cf84c99575df4f3874 : SUCCESS in 8m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/33322e295a9149a98580d2bcb340d2b3 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eab819aee10146f1ac26d17faf3cdd13 : SUCCESS in 11m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dff0bb8e8a23417ca1eea62e74a4622c : SUCCESS in 19m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/600c492048884a8e80a1c02623cde282 : SUCCESS in 29m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8d55e999026f4def83c094a744b93a66 : SUCCESS in 16m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/17d509cf4ee64a308d2fc6ca0169bedb : SUCCESS in 15m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d8ee1a78ad6460aa5d5894a3232a226 : SUCCESS in 19m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/74330b53d82d4d718f0d8f01e9e2c63e : SUCCESS in 8m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e4c186bceab34fd2a01c81c2991f875a : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5da3f3cfa0094c2483c8ffef54b54daa : SUCCESS in 9m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/76a8872aa85940e7af46af5c5e3c7408 : SUCCESS in 12m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e71e9e7cc8ba405e8c8a6af5ca85c4e2 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a28d63fc827c43a6a0fd159134fbaa5e : SUCCESS in 1h 05m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ba190e300f6490ba477d69912b732dd : SUCCESS in 1h 17m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4220ace20b104389846aa3a7242a27bd : SUCCESS in 1h 31m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73902b1abf884c49bf66f28a55dbf464 : SUCCESS in 1h 11m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3bf7e74412a0df6b2133e058b115726b3dd48a6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 21:54:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6f1549d4de489fb6aad03a69e24bb1 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"93d08927be4695d31b0f9c88dc7fbab53b262146","revisions":{"c0530647ef64626e1315475f3adaa9ee43f38c96":{"kind":"REWORK","_number":1,"created":"2020-02-11 21:28:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/707253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/707253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/707253/1"}}},"commit":{"parents":[{"commit":"d482f46963bd665ec7ca8e6fdfe2fead60ebc7c4","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d482f46963bd665ec7ca8e6fdfe2fead60ebc7c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:28:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:28:40.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.0.0","message":"update constraint for python-manilaclient to new release 2.0.0\n\nChange-Id: Ic2d14c94c2c434d228b85f8cd8515ee1f90d0fe4\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Idbcb65bc1a3e15440623eb370692ddf2ec69ae9e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Tom Barron \u003ctpb@dyncloud.net\u003e\nmeta:release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0530647ef64626e1315475f3adaa9ee43f38c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0530647ef64626e1315475f3adaa9ee43f38c96"}]},"branch":"refs/heads/master"},"8a39c1adc17ed8272f72156eff5f42612f7e4d18":{"kind":"REWORK","_number":2,"created":"2020-02-12 16:38:48.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/53/707253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/707253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/707253/2"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:28:40.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-12 16:38:43.000000000","tz":-360},"subject":"update constraint for python-manilaclient to new release 2.0.0","message":"update constraint for python-manilaclient to new release 2.0.0\n\nChange-Id: Ic2d14c94c2c434d228b85f8cd8515ee1f90d0fe4\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Idbcb65bc1a3e15440623eb370692ddf2ec69ae9e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Tom Barron \u003ctpb@dyncloud.net\u003e\nmeta:release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a39c1adc17ed8272f72156eff5f42612f7e4d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a39c1adc17ed8272f72156eff5f42612f7e4d18"}]},"branch":"refs/heads/master"},"93d08927be4695d31b0f9c88dc7fbab53b262146":{"kind":"REWORK","_number":3,"created":"2020-02-12 17:01:10.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/53/707253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/707253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/707253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/707253/3"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:28:40.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-12 17:01:06.000000000","tz":-360},"subject":"update constraint for python-manilaclient to new release 2.0.0","message":"update constraint for python-manilaclient to new release 2.0.0\n\nChange-Id: Ic2d14c94c2c434d228b85f8cd8515ee1f90d0fe4\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Idbcb65bc1a3e15440623eb370692ddf2ec69ae9e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Tom Barron \u003ctpb@dyncloud.net\u003e\nmeta:release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93d08927be4695d31b0f9c88dc7fbab53b262146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93d08927be4695d31b0f9c88dc7fbab53b262146"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
