)]}'
{"id":"openstack%2Frequirements~707257","triplet_id":"openstack%2Frequirements~master~Ifc4644c1a9d58ac6e8b2574a8dae641e664b6c8e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifc4644c1a9d58ac6e8b2574a8dae641e664b6c8e","subject":"update constraint for bashate to new release 1.0.0","status":"MERGED","created":"2020-02-11 21:36:24.000000000","updated":"2020-02-13 10:34:34.000000000","submitted":"2020-02-13 10:34:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707257-1581590074958-2176cf01","meta_rev_id":"8a1f788983d7fdd5d0c737bcafcdea2de3387198","_number":707257,"virtual_id_number":707257,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 10:34:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-12 19:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-12 19:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 21:37:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-12 19:07:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-13 10:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3918b82972114872b7f688925a258b2d984cb6f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 21:36:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6686bf50587de8fa7fc62ba3bd57ec44bc1be1d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 21:37:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d495b18d203c00e30ed33105ea9d1836008ca04","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 21:37:31.000000000","message":"Patch Set 1:\n\nTho... this is a linter. I didn\u0027t think we wanted to track those.","accounts_in_message":[],"_revision_number":1},{"id":"7c267de404238f0435a38a4e7b942921e2114b97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-11 21:42:08.000000000","message":"Patch Set 1:\n\ndropped py27 support","accounts_in_message":[],"_revision_number":1},{"id":"16289d6f5e7295fff75079de248d76c086aa2fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07519d0a36e64d8ca84c362b9b55f2a2 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5ca294ad634743a33c6d60eb86f119 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7cc6bff4df41e19f9c6817faf7803a : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84925f4b26d8475184561414ff8b2c86 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a986cafac194410b1528821faad3ced : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff1c83e645264b9ca9203efe9315a80c : SUCCESS in 7m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08caab36f54a47fab804036c16867b60 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a96dac4bfe2f488ebe22a719131e4289 : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6c0f9705622740c9b3b9759347c67cb9 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/646177d04526415cbe2208b390eaf429 : SUCCESS in 6m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81a5a2b1956741d7a7c18f06c52d8d78 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12724ab71456437dae1011da3a4cf744 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ffb211fb73724817b3c8a986e019c109 : SUCCESS in 15m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f852138f9c34a6981c8529765542976 : SUCCESS in 28m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ff768e52116e4fae89e6d4c09c085bf0 : SUCCESS in 14m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1623ff7490694f91bf4a6443d27714ac : SUCCESS in 16m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/038ad2957c964ff1aede9bcd6eb2f697 : SUCCESS in 21m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e26906f1624c4301a8c505e23090b411 : SUCCESS in 6m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d9d9de91cf15461a9a5f8e93fbe4c163 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30d2e78417e34ad7ad272bb1733db27f : SUCCESS in 7m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/953946de736c48a6bf2e6e54b627b66a : SUCCESS in 13m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96cdaa50e1f049638cf05e96cc9b01c4 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2ea371cba8b49f3987391d0d5efcf58 : SUCCESS in 56m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/795c17fa551c43e1bc54aa08b9c9e728 : SUCCESS in 1h 14m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c077c62830674e75972e5f3921d81b69 : SUCCESS in 1h 40m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fd48c6f6232f4fc584b9436a8cb3e7a4 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b16efa3ed1f802fac4d2e15b850a832c2333aa76","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 16:51:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92f343dc6445821618bb8e2e44fb787c46c2f014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:04:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff4f3e8d6624817bee9793620efdfb0 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e58eec00de5442c85f209848a8494fa : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31bc46fd801e42af9d2a95633d00e09c : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f060f828e904e73bad6820624a75b0d : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d26501b14b9b4acbb5202099b3379e2d : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c69da75d993e4198be2c844b86e2bb1a : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e866c56fccd54cef9b1188fb2d5bdf02 : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a3384a972bf4b03b7fc37cea07e2052 : SUCCESS in 10m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/137353a6dc3c41e897ea219e3d838e6f : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a844860a11b54e0597efa7fb41f1b65e : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8a155830537425aa41f87df875a92fe : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51a13da15dca45cbbf60d3311b32b12d : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e291ee73af9f45bb840d093cc0e4f0ea : SUCCESS in 17m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/89d7754f53904f969ece8d643b32511f : SUCCESS in 25m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/92ffa8d3753f400c9ff05eee4d4c2e0f : SUCCESS in 14m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3912c2b4572c441c9a3086d7547dfbab : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83eef66f21984e1bb52a6cfc954709e1 : SUCCESS in 17m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/90cdf13b798a4c4f9fbcf701e9d83f8e : SUCCESS in 7m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/85e336ac1f6d48aba167369f4e699314 : SUCCESS in 7m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59cf827f758449a3bb9382082c3b977d : SUCCESS in 8m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b198a8c3bf83484d96d060befeceb922 : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96a6113c26d04606912f8b7719434551 : SUCCESS in 15m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c371bd35e544e65a200a8a0bc12449f : SUCCESS in 59m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/30981a848b6f4f08bc2534d199380d19 : SUCCESS in 1h 16m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0e0fb866cae645cab346eee69c256a58 : SUCCESS in 1h 43m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04818720e40d4837a9fce9315630e251 : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":2},{"id":"684d8e3404f00658ed93ed96f4578eef48d4c82d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 19:07:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5751e2d342a32aeb11400e1d5183d4938e5a6f26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:07:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38810d52f15f9dd90cf6121a78d699f383e76e3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 22:10:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf19e9354e5b490ebd1c546a081f657e : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864f2ea3c3fb4988b6080a998a0f676d : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8022d7210bc44d38c915a7823a0c91d : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdc4a61c86e04f5c84beededa0705bcf : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8364815804384a7fa9a7ace0cc0afcac : SUCCESS in 6m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e8f00b68af1402d94ba923277b9f316 : SUCCESS in 7m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7250250df33f4294a02184eb3241e7e8 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/84a1a8f4e15544309a37a51ee09a3ee1 : SUCCESS in 12m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8da26f4c991c4161ab02fcff183116c4 : SUCCESS in 9m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/def0f45f22d34440bda827a0a5d164dd : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4774c4f7374e43d7abad1c482ec1799b : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fe93bce6dee440ad8f5fda1cedb2d1d7 : SUCCESS in 16m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/286f8b03f9ab4a498f294e45c99eb2fa : SUCCESS in 28m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ecc9d0bb5b84457aa179275b92ea08d9 : SUCCESS in 17m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6b525440bfc6460b82a7af96c8b1e45c : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93a88bd5cd924f8ca0f0d1618bea9f9c : SUCCESS in 20m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1130ed88d09b4169a41de4182f4d2b71 : SUCCESS in 8m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bf5b746d088a4d4abd9faf94ca6575a3 : SUCCESS in 8m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1441832d081f47cfa448f060721f43e9 : SUCCESS in 8m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/624c7b4604f84e35ad020d02549e3f96 : SUCCESS in 11m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/149c9ba793934a6f9bf23c1cb5fde184 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e44e534fa9e04322ad429c2cac18338e : SUCCESS in 58m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fbdd7defc314cb991349eb068c57050 : FAILURE in 9m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb0039684f5e4ef9a58f7e590ca7beac : SUCCESS in 1h 51m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6bda8b5e3a6408e9e180e8747906f42 : FAILURE in 9m 03s","accounts_in_message":[],"_revision_number":2},{"id":"7a6355d23480ad4f8a34a0107e5a8fa7e33a620c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 22:16:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"769da8d3691a15e499246f1f138a81a8fd0dbd43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 23:59:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9f80237add4af092c40b4827565478 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f31c87dd4b4ce6a7a1052e175bdf59 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77ccee50a58244a4b2caf78aec8eff98 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cacda54f407490d89b7950556661906 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6c7a23c062a4a8bb71a4e957c42d8c8 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb53bc44ac6e4f199e8edad97260639f : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fda09a0f1774de78f4acdba346d35b4 : SUCCESS in 5m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/69c67ec9d46f4aaa8be776d318860b91 : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b9bf5feee3c4e1aac6c72cfa3484175 : SUCCESS in 11m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea618b2504c044d19ed18da57b49fcf9 : SUCCESS in 6m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e344a249d8be4230a71c8357a7394761 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b29962552b74a1b9b99e9e5998e46de : SUCCESS in 16m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6372ecfa82f4f159b7c94480685c9a6 : SUCCESS in 16m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e7f9f52eaa9b4b758dff4e658b7ed15d : SUCCESS in 34m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/50c91a96598641a3bdf9016c4f11383e : SUCCESS in 12m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/33e8dde979814333914bcb70f615817a : SUCCESS in 14m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2abf786c1c104f708fdab15f1f96d9b3 : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/94e7b053aaca497f9f4079acfcdc9697 : SUCCESS in 7m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7197c50089ce4a048e19f7f3be76a62f : SUCCESS in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ca2e62e43b140418ed2cd577e55a7a2 : SUCCESS in 8m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b8ce1af3e6947a295fe312bbe3ca415 : SUCCESS in 10m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/836a73b512dd4173bf982e11f044a474 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5ffb8ea475d4707a717365ba21eeeb2 : SUCCESS in 53m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e2f57faa0874ca99457a29b38d672c5 : FAILURE in 7m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9bfbeaf96078405b94d7c9256ebaf3d2 : SUCCESS in 1h 37m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ebe8c8c7ac674617806eeb2ed7945950 : FAILURE in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9f1b2b872fed5bda26b3c21ace34fe9118e12271","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 02:41:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6b4132dabda2d54332d2f7d99aebc38a73d415f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 04:10:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a924238eb2a9445ea5fbce2aead3ed92 : RETRY_LIMIT in 3s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9b0c2108f64fcfa0a7c334103701ca : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a74d3ede74e745e7ac599aa9feb111f5 : RETRY_LIMIT in 1s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/454a42dbbf8c4098a0c5cab123dfa66e : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11480a01cab468280da1b99667da5d5 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1809a18b00e4390ba1bb0e8ebbf2b7c : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c45572a2f78440deb29ed765dab48a52 : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e1011d3c25ea4688be09d6e76dd79590 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b0cc09a3f654466bd8d97e287b0ecdb : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec87624a223c4facbd346858ede34eed : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bfbc511686d249409a2fa0a923723af6 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8628b8b5ffb4460e9e2950260054cb1c : SUCCESS in 9m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e2bcb48892b348328a4604efdb9199dc : SUCCESS in 18m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4e7a783fdd444505a56dbd35a8724034 : SUCCESS in 23m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/61b4f85b1dad46c39d70addc2336dd03 : SUCCESS in 11m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cb8fd7bd6c1a4b93b861f3790c6e56e8 : RETRY_LIMIT in 6s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/146be1f93f2840078f598cdeb6735bd6 : SUCCESS in 15m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2f93f308ac7545f7be46816c20019327 : SUCCESS in 4m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/92eb246152b941bb905394f4195ef836 : RETRY_LIMIT in 1s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88d977b00eae4580a8e65d6d6aff47b1 : SUCCESS in 7m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/943dd5e2c58d45288df38d18a9ea6fda : SUCCESS in 11m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5818cca612154120a85d9a2e5a8e459e : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1981d49b18074edd838985ab65aa666e : RETRY_LIMIT in 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/649dcda8a45243a5a3f84d527911ba6b : SUCCESS in 1h 06m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1413edf61d344b0b97fc3a148e5dfd71 : SUCCESS in 1h 26m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6351bd17bdb04da0a41d3ddf024d6a80 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a2cf612ef146d2ac0e209441a110be24b0b1b857","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 05:02:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"462ac2d0b91a3bb4a6b54a536709a6f4b3aa6d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:29:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40972e50e26a485fa697813758a37c66 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b715954d90b64ebca606fa61827e3651 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdfbaf77ba9f4bd59765f17c15f1cec2 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85a2105f3b2f44c7a318183caa756fa9 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44339e90c1564c50bda0cd1fa0a4f752 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a7cfe7f02e746738feb1ce664510a4f : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4953c853d2a84125bc4edd0df51bb03d : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb7561fde8bd4a6a845643c274760c21 : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f087800aa5fb4cdd949b09dc9efc545c : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f7d920bb74d420ba4ff6e96faafa26f : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b7223af06d1440ba0921aafbdb05f5d : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b461552b97f4e909c46cff268d4fbca : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fdf39646e6d845aea0078ea1d99889a8 : SUCCESS in 17m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64405a835c194f088062bc974285bc85 : SUCCESS in 22m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/381169a3569d4965933c733a81bafff3 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/564f9254b46c473bb7a7abd4bded172b : SUCCESS in 14m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7234b631bc144d97b583a9a83e697027 : SUCCESS in 16m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3333c9c549b842b38d11e9b5ecb89d9f : SUCCESS in 6m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bc4ad621183749d2bee56f8969903525 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eaa1df3237ca4fbeaab23a7c6c2d1f6f : SUCCESS in 7m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af59facf38f544fd8758247c0783d964 : SUCCESS in 13m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fa8e75602b604ab1a1f6821a492be6c2 : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/430a2ea0883643f0bc6bdb4476e847b9 : SUCCESS in 54m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90cd3bac75f8498eb61a0cc581da9c6f : SUCCESS in 1h 05m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c4dcb1a4eb4340488597c4bd609ec6b0 : SUCCESS in 1h 23m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3cc2adf676124917bed94ddfd2fdd4de : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":2},{"id":"3393a88ded7fbace478dfcdf75a74aed87b542d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:30:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52acc182ef9ce6ce83dfb123fe8af6617303db7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:34:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89a2cef863939e80efc750feb55f346eee749230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:34:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365747869f8d4031bdf1cb362c7e372f : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5931f064c6c145dbb1015487f34f3eab : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf2aed7eded4dd1a78e18d83d0d34a5 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f33e7f205e914fb38d8ded3fd56502b6 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/799d061592d1409889554deff8084e06 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0bcb5949d0ea4a1ab3b4f00daf5a85b5 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eecac4272781418296aaa31b2fb1d2b5 : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/295d51d2652447b69a43653fb10e42cc : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b500e9c47c604bea93860d5356253727 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1e2d59d9bb354d7891a68cfa5f59ba0d : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6590e61a30ac4c4b96b7bf59019be836 : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/df8f877b04c9426dbf1d9afef2e5808a : SUCCESS in 14m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ad2c7f314e484fcb8e826ab38227eab6 : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d62e4483f48e427bb3b923e9414de405 : SUCCESS in 15m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f97e00dfaa5c4fe793e3983f9c73ebdd : SUCCESS in 14m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/220649956e904f3dbca655849c62666e : SUCCESS in 16m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6f1e15b719be401e90f644eb980c106c : SUCCESS in 6m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7473641cfdff4e96a28d9ad6d697dd12 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65cb2c3dc7804c48a097b1312f35712d : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a60b95bc72d74232b8e29df4ec836c5e : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8837cd7b9ff944758c45b417f90cf7c7 : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9ad341b0b314be69c66e75c43af0912 : SUCCESS in 55m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0db4c4010d8f4f4c8a915fa2adebfaf5 : SUCCESS in 1h 10m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/874ad94293a84ed8ba49bb460397ad5d : SUCCESS in 1h 37m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f4dde825d7a46d89b82467283e22ffb : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa77b23279dfc399988958951343168b328211f0","revisions":{"f5576d7bea0772aaeb0deea056084c784126960f":{"kind":"REWORK","_number":1,"created":"2020-02-11 21:36:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/707257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/707257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/707257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/707257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/707257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/707257/1"}}},"commit":{"parents":[{"commit":"d482f46963bd665ec7ca8e6fdfe2fead60ebc7c4","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d482f46963bd665ec7ca8e6fdfe2fead60ebc7c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:36:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:36:22.000000000","tz":0},"subject":"update constraint for bashate to new release 1.0.0","message":"update constraint for bashate to new release 1.0.0\n\nChange-Id: Ifc4644c1a9d58ac6e8b2574a8dae641e664b6c8e\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ian Wienand \u003ciwienand@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ie8d9211f17a830db7d16cad840c3b3b71ee9ffd4\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5576d7bea0772aaeb0deea056084c784126960f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5576d7bea0772aaeb0deea056084c784126960f"}]},"branch":"refs/heads/master"},"aa77b23279dfc399988958951343168b328211f0":{"kind":"REWORK","_number":2,"created":"2020-02-12 16:51:32.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/57/707257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/707257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/707257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/707257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/707257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/707257/2"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:36:22.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-12 16:51:26.000000000","tz":-360},"subject":"update constraint for bashate to new release 1.0.0","message":"update constraint for bashate to new release 1.0.0\n\nChange-Id: Ifc4644c1a9d58ac6e8b2574a8dae641e664b6c8e\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ian Wienand \u003ciwienand@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ie8d9211f17a830db7d16cad840c3b3b71ee9ffd4\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa77b23279dfc399988958951343168b328211f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa77b23279dfc399988958951343168b328211f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
