)]}'
{"id":"openstack%2Frequirements~707260","triplet_id":"openstack%2Frequirements~master~Ie4bb5c37020afa95737f149283607d47e92ca54e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie4bb5c37020afa95737f149283607d47e92ca54e","subject":"update constraint for os-ken to new release 1.0.0","status":"MERGED","created":"2020-02-11 21:53:49.000000000","updated":"2020-02-13 10:36:26.000000000","submitted":"2020-02-13 10:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707260-1581590073558-027efaee","meta_rev_id":"81a5677c2ba51181ba4d5275c8c70a64378dac23","_number":707260,"virtual_id_number":707260,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 10:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-12 19:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-12 19:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 19:07:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-13 10:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d6e88e8a644f8abb0e13909e893c5941cf58735","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-11 21:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fabcae337622f30f1e684818349a32b94f8e5d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:58:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa2c50aeff743679d61d01f864bc949 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7642b8994c49c190891080c90a10d9 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f728ae5293524725b152db7032e690f9 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2e8f1697ac0426b89fed3c3fd6dd30c : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f974d51ba4224aa7a6de9bd0171f09fd : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/067eef31edc84ae18bc3ef3a7827cec9 : SUCCESS in 6m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd35ea737f5e4565a4e2ba34ac41dcb1 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/063a2f66fd2c46c6be5dff4934c5235c : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/58199c2256df484aafa5724b2189b859 : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db50e28e28d241308fecd92be43ce0bf : SUCCESS in 7m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2662c3ff43e047ac8bd2d9417af40a61 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/711075f21cd54223a752ec36a12162df : SUCCESS in 9m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/51b8a47d43924582a784b58cb24610ed : SUCCESS in 14m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/107694b439b342a9890204f0f3f5d824 : SUCCESS in 25m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/41205a3c1b674fecb18539911f315b22 : SUCCESS in 14m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1aa44ef8c9a84c02bc9242c4f97abc3b : SUCCESS in 18m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d3849d7b98744ac8180c0eaf6c84694 : SUCCESS in 16m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fcc4dd6ecb674d978463638f5026a6a2 : SUCCESS in 6m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bd41e6640c1e4fe4931ed8199372e452 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdc1810d19e44d8db3bcc20cdd7bc684 : SUCCESS in 9m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e3afba06b3204adeab4b68e5c011b649 : SUCCESS in 10m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/148152eaa9814a43a1a2efa488575b53 : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94e47424b3e24ab9904e0932fb8b625c : SUCCESS in 54m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaae23fa5ac7404eacc5419b8e6ff783 : SUCCESS in 1h 14m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7fc2e7b5094540ecb69c2872909d2ea7 : SUCCESS in 1h 39m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/52c4d399771f45b1bc52249f9619e83b : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":1},{"id":"326f104df52801fe58ae8292757dc8c26e482ebb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 16:52:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ecfde6bfab01ce170a550fcdd31003d4af337c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:59:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d89b285317ed4aa38847b6d042c5845f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadbde14698240b3b99f9a5c2242f17b : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6bf8623cd54471b926d5a85ff14fe24 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/078b7427641344f7820f843584e66751 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2fa54a0acc44de0939a88d4df73f0ba : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b800fa54a52425da1e9ca9142989354 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a61a94aedc234bfeb81f8dca3a81acec : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/acca4044405147cc85b0aaa9ea2f1a1f : SUCCESS in 8m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd31e3cf13884247a3d4d2c8fe7f506e : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1d7d043f4e0c4a5b96f4e32ee217ff38 : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dab04835278448bc82339fef96d05645 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6be1ccb4be4b443280cafc09336576d5 : SUCCESS in 8m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e7dc5f34ac34a2190fc07efc9ece39a : SUCCESS in 17m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6d0fe48dfac747c5a61db88d5d6db889 : SUCCESS in 25m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/42dacaa013bf47b08eac2eadd4ab8a1c : SUCCESS in 16m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/16fc22ece0c44979a32c8c3c8bc3254c : SUCCESS in 17m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e90d0486752046e2b82a925c3de3823a : SUCCESS in 20m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8cb4c90a0fd945a2a816ddf6146d8c3b : SUCCESS in 7m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f6c627b93e5c41d6b094ccb0f4763e24 : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ba10a24d09d44dd9c6f8b904797c884 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3d19416cd2334f03827145ab4d949dad : SUCCESS in 15m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/14aa19c53ed0484f9025c15924b1757d : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/911c565c7c424fafaa8f1c9d5d0dc374 : SUCCESS in 53m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76187789370e4fdfb045c0cb049441e8 : SUCCESS in 1h 15m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d3a5157604f4505aee236cfb10a712e : SUCCESS in 1h 28m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/76e5eb68a40d4c0d91ca0a52e62077bf : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f356acb811aaad1b822feb1bd02d758f3b2c6f28","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 19:07:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e35358732cdf19f132fc387e5623f78a635e38f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:08:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62c6808dd41cb3cf5dc410585f6f23c8747a4360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 22:18:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48f9916580f4e439c109e9989054195 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b898f01bab49f2970482bfda5b1c88 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a17be32bfeb4456db77df75474cc16df : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e30e2b849d941a5bcd89421932c36ef : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4911a459bdeb482cbb27f4d6491691ca : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e051c2bf39b949c9b2cdfe5b6fffb0ad : SUCCESS in 8m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c7215e2e8ef4440198ae828741c7b70f : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/435002ba5f3f40a18dcd2519868c98f6 : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c39cbf088ae44d2a82d373b32d368625 : SUCCESS in 7m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48c903187b104f86870914fa8ff1aecd : SUCCESS in 17m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae6f7de02e784d66a5ebdf10bc3b59be : SUCCESS in 9m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0856e7df65d64449926f63b302f9ba27 : SUCCESS in 15m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15f426fb2ca34d72896d4a9dec74a0b1 : SUCCESS in 30m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb44e82237ed44778a12fc72a8a3acce : SUCCESS in 16m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5e90ff454ad54a1587ddc53c6beb0ff6 : SUCCESS in 16m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ded715be1c54d06bde9b96b1e6bfe76 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ce9680fb22245c1bd7fcd94d666138e : SUCCESS in 8m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7043da0e501b4710b15bd9e3f2f129e0 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc7420c956214b1bb477b7198f41d24d : SUCCESS in 11m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9705732ec5d44165a83fe10871d17608 : SUCCESS in 13m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dde16097ee754e84a84dfc6c3c133652 : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fde12606b7864368ac1b8e8ccbc46de2 : SUCCESS in 54m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34480bacb883437998443398bb62ea66 : FAILURE in 10m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/16008b938fb2483aabd4bacb689ce579 : SUCCESS in 1h 44m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f280355a25742c7a1eec224791a6422 : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d6ae7b812c7ff90e1c3af15672f2a17a8b43dd70","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 02:42:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2aa26ac99af5569b1bb7999c61c3e3d01851a1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 04:08:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01528e7fd9684027b1a85a5039da1723 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4bf428aca724bbd957f3390e5ca8ed8 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ff53f2155a34c9e8b3186c81794f63c : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0ffefc87b044844b6ac41455b29a624 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e9fb45c300f41e4af50932bdd7d0670 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9911c7a666d493a8ee2475e722be60a : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4e3bb84a0ff4fe1aae4788d6eb733d6 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6548c4dfc33142b1b06df42c08c240f0 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2864195389124704a26200f21b1c1911 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c14eb77c66834d69bf765eb51dfe64f0 : RETRY_LIMIT in 2s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfbb859607fc4b4bb467e01ca25a412d : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e722b676b7f4081b917d4d754e9448b : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c3c91042d9674a718b396443c327061b : SUCCESS in 18m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/da6e0c3ac0a74e2aa516f99a682c34a7 : SUCCESS in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3063c7b04e9f4483b2e6802aaf929c16 : SUCCESS in 13m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e653db8bafd349c1b90eecd826101afd : SUCCESS in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/921f9843e689494ab2bf90e5f9e6a9d8 : SUCCESS in 16m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ccf020ff6dec4edaaa56ed79a57133f6 : SUCCESS in 4m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a66bde839f9f40cba498c02e925d5884 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e62bbf1af200418ebd5856a471e4c3ed : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/abaca6b167cf4e428388e67033e21066 : SUCCESS in 11m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c559e2f24e4b4817a1eb274fdcabba92 : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f74b1538354142638776209d51419f6f : SUCCESS in 53m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d56b636d449341df954d6f9d5b99f541 : SUCCESS in 1h 08m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/59ad05406d2d447b8e219b1ac1af0bb5 : SUCCESS in 1h 24m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b2d4c9aac203459fb533942cac9abf8e : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1a265ba5791fec3e28d76b6afefb29d7b6493d7c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 05:02:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"51619add6aae71ce17e118f614fe6d801708e348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:29:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb808dde5bd64514bf0c8d59d54324f9 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e933230a4c5a4e29bd9a08bf4d755872 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1137186cb3b748e3a01887c9617ae8ae : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a03b2ada8e94383be87fba44527f282 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dba0e5d55a7f43edaa6fe13a77462c68 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f9554e6828f44319f1ded11554aec37 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b35aa69e5bfa4451bb5765333f895c60 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/169166b45d0b47b199dbc93f4d0b9fe0 : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7fe5518210ff4b13be5a44fe60f5051e : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dc61a50e7614430c800d7a51c69fc491 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63774c80292c4497a702a46c8e5981a5 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce6d90d37b914435bd85cfe2787a2605 : SUCCESS in 9m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5bab1cc388d64fa085e9c7be03367ec3 : SUCCESS in 24m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eca61dcc294c4326834e32174a8e6393 : SUCCESS in 23m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aeb97628632b462ba8d98d0e34117b78 : SUCCESS in 14m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a36073be293458ea0030dfc9b92d6cb : SUCCESS in 15m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7c0d8de98244edcb5fbcffe647cc771 : SUCCESS in 16m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e19ae6c66538447089366fbe73e40f64 : SUCCESS in 5m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f562c1c0cdd24e34b224e805c97c1e24 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9548230a5f9e4152bc78e9b7cadcffa5 : SUCCESS in 7m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a86b2cd2736e42eab5e4243ebc911e7a : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/870813cf3a114df8b5005b961cd0cd9f : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ee70b21bff2434d835222963cad111a : SUCCESS in 1h 02m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/165b1e962a164e65bec9fed7710883ea : SUCCESS in 1h 02m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/477bd92f5c154b74a7268bfcc1172df5 : SUCCESS in 1h 23m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f407bed3d9384e278597b55b3561e361 : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1ecea6ed098478768db2ec66c66f7fa3d6c6cf6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:29:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d08a65a6fd16b2d5c797e1f99f345a41be11cd62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:34:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df00244fe5104746ad605f894d2f6a66 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8b28abbc39476698ec017b79596964 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a39baad1ad6430faa7bddd1ba4de3b3 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5baf968c293c44c691685e88836ca670 : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8327aa77221846049c6c410d951c606f : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/231d2b3ad62b47909ac322c9150a1940 : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/317879c4716f4c48bf6a807bef915eea : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40ba6e1d662d46cebb83d56705031d52 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e84003fa308e4ad1af582a5be27c947f : SUCCESS in 7m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/71bbdbabda374d28bec1055390498ef2 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98ba8aff0b144eeb87f498538ac67d46 : SUCCESS in 9m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95c749bbea3e48b6af86a0f44cac8c45 : SUCCESS in 28m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1281bcef79b7455ebba8a9f91b82da20 : SUCCESS in 28m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e1dfeead970347bc928bb1f3263a03c9 : SUCCESS in 16m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/850c62d4cd954440a87b1b9dbd25b704 : SUCCESS in 18m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/574caacb81844c47bdbff0e62ff346c6 : SUCCESS in 14m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/519b20e14f4241269c9621836f16764b : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/12d2bd7022f043bdbc8452f03327f804 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e6793d6718b4643a282786ec2f4a925 : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/63902cdabc964daeb4f90474eef44b1f : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/129d59c886d345d4a484f21905d497d4 : SUCCESS in 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e0342f026544e2d9e48c9db8647a630 : SUCCESS in 1h 06m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bf942ad2d624c01ab349dda3f712a4d : SUCCESS in 1h 05m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ea329fb7c82b4e18b55b532f1ca67720 : SUCCESS in 1h 36m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/03198c2e74dc465ea91d4347276ee991 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":2},{"id":"bc374218c72c693bda2c01c5a939dd361fbbe8f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:34:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d96cd2b5fea938f3a53acf4196e23c0ecce4249f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:36:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eaf93700fc9426f8d30a8dd2739c726 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3fef9de3f7e13601ad95e4fc234879092da77c51","revisions":{"b52c78715d5d5c8403bb9ff2dc6350ea03532431":{"kind":"REWORK","_number":1,"created":"2020-02-11 21:53:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/707260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/707260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/707260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/707260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/707260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/707260/1"}}},"commit":{"parents":[{"commit":"d482f46963bd665ec7ca8e6fdfe2fead60ebc7c4","subject":"Merge \"Update the minversion parameter.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d482f46963bd665ec7ca8e6fdfe2fead60ebc7c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:53:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:53:47.000000000","tz":0},"subject":"update constraint for os-ken to new release 1.0.0","message":"update constraint for os-ken to new release 1.0.0\n\nChange-Id: Ie4bb5c37020afa95737f149283607d47e92ca54e\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I6e39be6f0bf00376e364ada61108f2e6f4152427\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b52c78715d5d5c8403bb9ff2dc6350ea03532431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b52c78715d5d5c8403bb9ff2dc6350ea03532431"}]},"branch":"refs/heads/master"},"3fef9de3f7e13601ad95e4fc234879092da77c51":{"kind":"REWORK","_number":2,"created":"2020-02-12 16:52:13.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/707260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/707260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/707260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/707260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/707260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/707260/2"}}},"commit":{"parents":[{"commit":"3355b605728d324c69cf551d87e7a7a64d97bdc2","subject":"Merge \"ironicclient4.0.0 breaks in py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3355b605728d324c69cf551d87e7a7a64d97bdc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-11 21:53:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-12 16:52:09.000000000","tz":-360},"subject":"update constraint for os-ken to new release 1.0.0","message":"update constraint for os-ken to new release 1.0.0\n\nChange-Id: Ie4bb5c37020afa95737f149283607d47e92ca54e\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I6e39be6f0bf00376e364ada61108f2e6f4152427\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fef9de3f7e13601ad95e4fc234879092da77c51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fef9de3f7e13601ad95e4fc234879092da77c51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
