)]}'
{"id":"openstack%2Frequirements~707616","triplet_id":"openstack%2Frequirements~master~I86f8147caa4aa3f170a0c3ef415edff18c67cd87","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I86f8147caa4aa3f170a0c3ef415edff18c67cd87","subject":"update constraint for python-senlinclient to new release 2.0.0","status":"MERGED","created":"2020-02-13 11:00:12.000000000","updated":"2020-02-16 00:57:57.000000000","submitted":"2020-02-16 00:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707616-1581814595187-0cc33de8","meta_rev_id":"cbbea96b3244c7c825623594bc2654c542253d1f","_number":707616,"virtual_id_number":707616,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 00:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-15 23:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-15 23:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-15 23:30:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 00:56:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59e6ba0ffde23901a8b20b93e2d304fc52e4db02","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-13 11:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"729cc3c95a7583d1f290364702afbff161e43e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:44:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f8dc3fea044f4cb47aa48a23067bbc : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16789a5eb5f94915a5844ef516d4a5a1 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db30f0ebef4c4b329024959b2cf6928e : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49bbb17ac6574918b94e4b24ba21d292 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0ed54a7bc4d442b9669efbe43274116 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aa57ea729744ad497a9a2e32733658e : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30a8897e651942c680f9d8f109d7a115 : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dd1fc776537d40c9a6748098f425c000 : SUCCESS in 11m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5202ffd658384178b8e50d881c80810f : SUCCESS in 11m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6316a94410b648cca1ba2c4a052d6cc5 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1b2186ad87be4208b4c2bea2c788bf65 : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/205cbefb6b234b5c99f6877167cd6ab5 : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f01b5bede4e84e22b0fac676729b5d6e : SUCCESS in 14m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7b5fa614055e4987ab5527994cec2b0e : SUCCESS in 27m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/42dd059a84634b0282ea669bdab16646 : SUCCESS in 18m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6ef8d720f6b8458287e0f7c8f9d579d8 : SUCCESS in 14m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5944132996ab4802aab9dbe382f7d9f2 : SUCCESS in 17m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/39a4cc93251c4e9e92760b00e0efd20c : SUCCESS in 7m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c6ddb8ea9097485dadc1bc5949114b15 : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0a3390d6bc344afb6dae45830158254 : SUCCESS in 8m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9faa3914bde24267ac6e450673be0ef5 : SUCCESS in 13m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c90bbb62bedf4b4fb3b86de2c71f0c58 : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88af351daa0044cd847730e56ab2d709 : SUCCESS in 59m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c148692118834d99af1b4dcb24d54b3f : SUCCESS in 1h 28m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4b7c6f969024419fb3d3a0562d44b09d : SUCCESS in 1h 34m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04779909da294f198e6baba577e69bb7 : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1b358f2e8ef00e615177c92956985208cfcbb709","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 15:51:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec489a13a94d72a341d025a93bb22a7c8b5a4c61","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 15:55:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f7b6e68f96b917870a54735adb93c640d1234c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:32:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b0a2a859984609b6089450c7f57be3 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16850dbdc8114a90a2fb9e752bb5b50f : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcdc9048c8594a5fa69adb78b38ab2a4 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00366f6d69f1496687bf37cb25bafec8 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b54bedaff4b74c29bccb4b3beb6ab17f : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a0d939ae9384aa08261990adf33b803 : SUCCESS in 7m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/300cce7232124992ab7a589912132a98 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a8aaa13150d54e7fa687486dfdc0665d : SUCCESS in 13m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f5a8fbfe9d204230832b0204addfb289 : SUCCESS in 13m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1b3b556c84484196851a905f3697acac : SUCCESS in 9m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f84dac8d748440a0892064bead43bbf6 : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f09e3a4a6fd641b8a164b5927f4af0a3 : SUCCESS in 11m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/84b6556a9d134e17bd6122552cd04e2d : SUCCESS in 21m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b90f397e00a407f8ced5837dcd947c9 : SUCCESS in 29m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4427f1aeb62e45388b5c4650c1a3bcdc : SUCCESS in 14m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4155d88f11d442449186af219799f057 : SUCCESS in 17m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5686ce39314d4e1cb3e8a81db3015ac9 : SUCCESS in 23m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8d3b98fdafe44c5699249d6ee72f617c : SUCCESS in 6m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/54b54a9d661e4f5b93dd151beb490ed3 : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15e672751d6946d4a8801de4640a54a8 : SUCCESS in 9m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/54f001cb526c4302aec19eead25b4d2e : SUCCESS in 10m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e69ba0e3836b4aeb881b7a96e5ab2642 : SUCCESS in 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfa23b2e6ba74a1499c7fe942269268a : SUCCESS in 1h 03m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1cdb47569ca4bfb90354208ffab8c2f : SUCCESS in 1h 08m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f452986b6ff460b8034c6c4d8341c6b : POST_FAILURE in 1h 30m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/838e26eeb994476196601e9dfe1238db : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":3},{"id":"0ee47cf5f77fa453084d1169facc1c90c0a0e8c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 19:10:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4b9456052f542ebab2fc8d3e8908d83be7361964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 20:39:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5409de8f7aeb4f30adc286ddf88375b1 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9e49069ebd4d50801b8549289cd670 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb3b9c0026c4ee78219673fa13e5139 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81eb6338e66144429f6ecba26ec265c7 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a68ca047a1407784df8b5ee0ac31e1 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6d957c3b5ae4a1da2040ae178fe563f : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66c9d76af8fc4948b1d567b9080efb0a : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d3af82cef9947178b217cdb02da6612 : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c539168a4a0b4b429a979fdc0a058738 : SUCCESS in 10m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4789af0c3904ca68632485c37b5674a : SUCCESS in 6m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eb5131bdedd74448bdbc9ea613e19af8 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7ab121a6c4a4ed6b1b6c7c744ce0368 : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65df83c96ad047c19bf676cd9bcacc6a : SUCCESS in 13m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b672a5e7466244e4a79fca5490bab6ce : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/097bc57b9c63485f8a8fd7a792ed0a0d : SUCCESS in 12m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b995a07c41714b84ac2c3d5d55485326 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d363a8d9cf64c528f02493b929c8f9a : SUCCESS in 15m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f33e9f09997a4e7aa9a4468e10d6f93c : SUCCESS in 6m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/da4b544eab3543eeb6d0481be0145073 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/51013b4868bd47cab8b0e2ccc155724c : SUCCESS in 7m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c3c08cf86a0f4cbaad79afac120c088e : SUCCESS in 10m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/720abe7e2fc242ce868032d6a9085d0d : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7373a535a254f3baf895c47615d4c7f : SUCCESS in 58m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94ea5ab79af842b6bcb7a733dd502c66 : SUCCESS in 1h 05m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/97e10aed61124551875d07b42fcbe039 : SUCCESS in 1h 21m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e4a31b761a4c4a2b8db160faf3f9f71a : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":3},{"id":"8ee6cb00a1225c5be110151148c371156d9ad287","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 20:45:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f0bc09644b7d73d83ac7bfec483046181fdbb1dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 20:45:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2203238f282fa81fb8277536ebfe776dd31809b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 01:52:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347d80c67c18443d8cdad638183721d0 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63dc01a7109432aaf0592b7306284b9 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca355494c5d4822894b72cf67082323 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d90ff09d2e554cbbbea6f13972be283e : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0491e4ea153d4453aa9576354aa5acd2 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8d8aca90d394f29905d58ffed9aee4b : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a693b0da98684c62a669d54f347e13fc : SUCCESS in 7m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/72b03e5c472a47dab2185d32cee02d35 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/969ed5ea548848899db8026ebe4c86c2 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8f3ad78c16914b54b2f4688020302790 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/30da2cc356b14a33a31c8170afcde915 : SUCCESS in 8m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ff26af9844854993ba36e2c6092119bb : SUCCESS in 14m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/876c414af5c9488687df6158f4bbedbb : SUCCESS in 27m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/245e847a329d4cea96550821c9c3b1da : SUCCESS in 13m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ccc46eb5f84e44c88f59fbb059870a15 : SUCCESS in 14m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7f8b88965174ef99b005f84dce3525e : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3e22ae6da2e46e7bc9a37d1e9af4728 : SUCCESS in 5m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4c198118a64940c29a3a1b506d985b47 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a99edfcd50f47e194edb25abb063dbf : SUCCESS in 8m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/43120e169798430b91bcdaff3b3aa7fe : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b47728748bc4133a71d1702a9613955 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ef95b1cae2745dc988f631b1247286d : FAILURE in 17m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0252e7355f314f09b19701fe3a8823cb : SUCCESS in 1h 12m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7abff3537984420b167ff17dd915845 : FAILURE in 17m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/da2574ccbf4d4764bd5938b0739454d9 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b31441efbfed94691edd22950db1bf0c76c061b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 03:06:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0f5c7999eebdd7798722896082cee2608518ab17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:20:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8827406d4da48228fb300bafff90299 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b91671ea1f41d0a187fa63fdbce41c : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4688f9248af8407a9cd35e6066ca50fd : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8742352ce3c74914b99bd18b59e60226 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc296f719ac347bb8a51e64d77063a68 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cd77f672ed64b50bff917492e579141 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3386f88778f345dba252df00a974512f : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29dbf9ddeae943419f6b9605a95b1b30 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0025c2ac280f4feea428fa878c6befb3 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63483f656ec646ebbe0a7d6f4387ffee : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/952eed4857794a1081dd30f693cf5d84 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e95d0a33dc404335b029b1319a81956b : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f29ebfebd2e444e5b19248eb628e22f4 : SUCCESS in 14m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0064ff22faa74b71a945024e678045e0 : SUCCESS in 25m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/66bf1e3d31e44678890d74af16e50129 : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/89ff2e780fa149418bd0c4dbf83ea468 : SUCCESS in 14m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8924eab4a23b45668513dcf9dac74cb3 : SUCCESS in 16m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4e008445e9874ead8a7d38c4517abb7f : SUCCESS in 5m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e82fd29a364b4f7d9dc03c549c87c297 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa3ef0e105ca451ab11c0dd68f6092b8 : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5fb9faf7a67240dab8fdb5caa129420a : SUCCESS in 13m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c01c65c07321448dac6e5b8c55be99c2 : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/878920693c8246b9b1b388a6efd085a8 : FAILURE in 17m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/13c999e7a85c475d9fd06c024c145ef5 : SUCCESS in 1h 07m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9923e49a5a07446f9a9c9ed3a655284c : FAILURE in 16m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/df62b7d897ed47b2a8fa20ad390a1a8d : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8abe346d3dd017e65bd7ab8776a93daaeb4920dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 04:27:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0aa0503ba85b50f127ebf2b14ae8a8aba60119e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 05:40:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1414d464218047dc80c25ed9a3c6c246 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a2d8d541674018bb662509ad0a5974 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee57aca50e154db89056a587fa706fbc : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aafddddc9c39462289c9fae6ca27b36f : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6716373ae1a546028f518aa9674f41d1 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54d1026ce3c843deaf2b2a35f85691ad : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc179da1bea7434ea9e0f3c3dd73ca2d : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2bc70a731f48491b854f5b7eea1a2dea : SUCCESS in 7m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b31ed85a23ee4993945187c5f14aab02 : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2febc61825c242bb9cab8a93a06031e3 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54a94bd072da46db89a79a22656b2a16 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d594e2daa38a4d61b9e3043b5cc6f556 : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bbb70cfe20f340339d999468cf9b2202 : SUCCESS in 13m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99248aa930274303b7c55758347ad294 : SUCCESS in 26m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a2cf43d9f0674bffb44d53697580e9f5 : SUCCESS in 11m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ef34103d08774ef386105624e1c2347e : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a47e44ad61141aaaff1f4bf48b5acc6 : SUCCESS in 16m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a9a8623b4cfe4762adcadccdc7061e28 : SUCCESS in 5m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e552a0a3b35d41b198eb768fc484ca30 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a635ac7a91e845e68593d8d64211abfe : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32a5fbae41414316afedf3455a65fc7e : SUCCESS in 9m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf160da0759c45c59c38c702bed3eb93 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c27ea23123a34f799f76a9d417aae068 : FAILURE in 14m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb04b2a2a342455f83a294032cbefd58 : SUCCESS in 1h 07m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/45a80dfdd1514bffafb94fa97febaa11 : FAILURE in 15m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0177906903b14907bdde4b24d4804169 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c653401f85f6fbd79eba4e41e856e2cab69ac113","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 05:56:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"918131fd8298d37dd2dfff2d84391669eaa24bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:20:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11478e0e425c4e7b8450b1d7e88a5f6f : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28816d58b3a045b289a24494766dd330 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f04287db4b4049688e1390561c3f878d : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6376c2f6f4eb412397fa7657c9802be3 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdb3645817ac4e4eac36c2d74441ebc7 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e086bc8648bd4394af7c2aa660aa2f9b : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fd82faf459c4969b35230d1e29e28b8 : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b14b7fafe7984184ba0edcf60abedcca : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f76fd4be9774d1e8592cdba033b70c7 : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b73086aca11c40e7916b8a460c6b8900 : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/69be457aa010401a8df9653b16003b98 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41d47b50e231410cb43b145002a34c38 : SUCCESS in 10m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b5b4aedf05eb4209b3bbe8bf6e1d3426 : SUCCESS in 20m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5736aac847824d018bc9b00cd49e7f5a : SUCCESS in 28m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d4f95476605348299bbdfb27b99d7f53 : SUCCESS in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/56ba137129b447849a7c21966bccafa0 : SUCCESS in 15m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25b8e770eb534c749d4952f896e29b37 : SUCCESS in 16m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/71c9ad2234a741a6896d918e1e59b47c : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e244cfa3aa004a71b20d9b359167315e : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e11d0ba1b59d49338c3ac61757ba8465 : SUCCESS in 7m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5ac79027c2a74bfd9789cb7ed69b5d88 : SUCCESS in 11m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a24607ccf29a4ae2bba689c952ac0709 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/873e37066bc945db804da9b0a87caab0 : SUCCESS in 1h 02m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/48899aa18300418ea6073d8d161047d0 : SUCCESS in 1h 18m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a5a62a9bb43463aa0fa636a465255bf : SUCCESS in 1h 20m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/39d9e536469b4e7da0387eed6ff161d4 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":3},{"id":"20ac2f0d864abe5aa74129039fea43325072c943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:20:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bf317ef5158a249fb892fc3afdeab8951dc35edf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:21:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c8767525a84953a1f1028d1aa4ee97 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd2af6ced0e4c13b3d72749520b8f60 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96aec9cd418c43d4bb4755b8507a5a65 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/673522360ca943289e9c50c6b0451ce5 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bed8320793c44d69394f8b03600d100 : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e7f0d0accd046cea48620d1448089ca : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2110f0e9eea1491b9f83aaaf6014005e : SUCCESS in 12m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cce58c7813ae4bac83181badfbd0b76d : SUCCESS in 12m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c1697b6640846998f8289ad7c0470e5 : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/43f858485da541d6809248f798d6568f : SUCCESS in 17m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed63eb3728f846acae14b53aac7011b9 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adf268c2055546ceaa97e5c0e45389fe : SUCCESS in 21m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9ed647c9ebcb4098b8b9396c39238b38 : SUCCESS in 37m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bce5939b47b84592b3cd3699ac51d606 : SUCCESS in 13m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c895c04964cb4b09827d84bdbca4b26b : SUCCESS in 15m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90c325dd543243349fee9b46fe5a0b26 : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d1d4e2d63f9d4b86a50a4d43611d3ea8 : SUCCESS in 5m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5c69d0e12259409abe653187ec517f8b : SUCCESS in 8m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e27353ba2c87422dbe4835287b2fc9b6 : SUCCESS in 7m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ec14fa64868b48b398a019ce2f658b9e : SUCCESS in 11m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b65f4ec119744c70a736504cf051af9a : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/217818de57f6429ca5e40353f9fa0f1e : SUCCESS in 59m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/91b647572978427aaee39efd643b0c16 : SUCCESS in 1h 17m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/99bf52875b80436d86789caec9c30610 : FAILURE in 15m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/295dc02588be4cbd9493f3f30ba5a7da : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":3},{"id":"8fbc4973bc65ec782273b430d2498ca6cfa8865e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 21:56:58.000000000","message":"Patch Set 3: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2415e96dccd7e69af0f6dfb03f392380e67edbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 23:18:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f05e3f78dfa436da6eda1045e33ed4c : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e337faa8da5e407487b691558fe3ed18 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf1e8f97fba4c7e9da3da473323990c : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7568ab6cc3cf4588ac41c72050e38bb6 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/811848eaabf040538796ed7666c19a2b : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9987c439ff694346a7c4a0e385144825 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8729a5749c8491a90ad6156437b0f73 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77e40dd8466e4c6e81026b07cfb084db : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2072da929cdf49389ea22b753875fdac : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/46a03012f36a4b8f9c0dda90b0cf825e : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8b25eeb76d0246dbbb715243e72261bf : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86ae3b4a7f864413ad661eac0bf9e370 : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef729491db654876a4d493bbab97255b : SUCCESS in 18m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3585b2c81c9546c7808cf0dace0ec7ea : SUCCESS in 30m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28bd26b5d4764611882ed7ee6f00c987 : SUCCESS in 16m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e1eb0e302f324865bb100bb785513c2a : SUCCESS in 13m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e955e2e37cf44efaf2ba1a9c18f20fa : SUCCESS in 19m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f0276321d0704b97a1d332c958730fdf : SUCCESS in 6m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/055c7c59bfe449dcbafa1dd2b0a9d076 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad55d23a5fcd44a1987ccb2a5c200490 : SUCCESS in 8m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/26d7cf5db51e4ac291890e45589c276a : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8d28b92a96354b59a829cd23cd57feb4 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33559cf4cf7545a1b1415b587025ae0d : SUCCESS in 55m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e0758371737497786367a37fa34c8a8 : SUCCESS in 1h 14m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e7fbc7e8d354ccf99d6541860a3b2b1 : FAILURE in 14m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/51062d43750b4e25a56fe5fbc3529f55 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f133319c0fcd909b1b1c2e527c0245f2d63add2e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 05:50:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"aa3fbb3cb48bac38a0ead0a607cfcf44e7e18ee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 07:05:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7648b8704364d7d95c4c0e5bcdf097f : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80414c81afba40af8dfa57a40ad859a2 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7027247c0303455a954e08e6b38e6770 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6590f1a5cea145abb958803f6ea1ab50 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1547d783f4aa458ca3e298eb295c1ac1 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c230917a8384cc6a24fd464a12b49bc : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ba735c9477b40ffa0edd22b7506d397 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5969e4d0ece430db003d2f2652cab13 : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3342f836c1e4f3988760599a74a7b6f : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/512d5a3acab8416db4a488d92426aacf : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7312f4f5bf344022aa8f4f1f7522eef2 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4fb1e9e93d274b9c85bdd82e2692e4db : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/355c5369fe9442549e9abba475a03139 : SUCCESS in 17m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f68c1c8d8c2840b0b9c1c9e2ee7d5602 : SUCCESS in 25m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6892caff230d497abcb880a34c57420f : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/50a297a94c37476eb28fd9bfad196892 : SUCCESS in 15m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96095a916fa846c2bad966cedddb38a0 : SUCCESS in 17m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b3a1825916164b81abed174712b59cff : SUCCESS in 4m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/003faa6afb0f4574a6ef0aa4134cb155 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1612fbf3830c44ad88048fc15c42183c : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/353b622e44e540f693e49c3940d4d6d8 : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e115aaa5150a4db6a1ed4de65ab0f063 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4d40d6bbeb143e48934159829fd92f1 : SUCCESS in 55m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad41d122e07f408bab5a897d3c5fd4e6 : SUCCESS in 1h 10m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b025a86fab44bfda5c0362802b89a0d : FAILURE in 16m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c6747d37d1fa4be481b6b9941ada8063 : FAILURE in 44m 41s","accounts_in_message":[],"_revision_number":3},{"id":"611757ccab37ecc6ac7f24d2c893a2445a7de2c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 22:01:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"97834540b3d85d74fc9f522ef7675b6b346e11c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:25:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3808700cc27e4310bdc45c7efae870c2 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333615eb737043048632dda42c35b9f8 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860a6c0ab52c47c29ad9e1675fbc3d62 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3de6c66f1334ad694c3ed59443a698a : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40989cdac3ab48b89ae6dcf1002170c4 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8dbeb500d394dbf8bae3dc2f359996d : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a604d7654e9a4137aab8e0198440997c : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3eb28026feb4c4090bf9128ba7c59b2 : SUCCESS in 7m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b3e9e44702349fd9c1f07d62f476a78 : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/934f5c99ec1841daab4c6fdc32cefcd5 : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/922218017b6e4c2d87a4374838b98fc0 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bfd207063a9449e82c7f23fef0ec975 : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8ed78aa6800412081436985830e28e4 : SUCCESS in 15m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/96159b1c5d8849d7a2b3337767c4dc1c : SUCCESS in 29m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f64fc1b2c284601a9137de4d7d068ee : SUCCESS in 12m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d3c67e0a74c4182831ef061573a3939 : SUCCESS in 14m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23b7a8a0966f4eeca0f2e9a775fd96f4 : SUCCESS in 16m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/390b8b8086a84e88a92b0b0beb6d6210 : SUCCESS in 5m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/56f4cc8858f047e19e617d30c0b1573e : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03bf926080814984961f0c672e341863 : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2fa95826c5c643df9a5edb3c375993cc : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf75d9323df444279ffebbbd6468ee15 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d488d66955a42fdbaeba2de6693c91c : SUCCESS in 1h 00m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69c653ea58844fef970b87b9e00464d5 : SUCCESS in 1h 09m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10d22adffb5a4d9b8dceb6be0d0155ea : SUCCESS in 1h 17m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/43e0b9392a724275af039c2b8088751c : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":3},{"id":"df3760c5062e85fe6d9d91f59d0d2ec8f8e3e0eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 23:30:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2323b1e544ac5fa8abd67e12ce636d02597898be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:30:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ba8e9517c1519a1b4519bba2c2256d1731c29bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 00:56:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff357282caf346c2924765905250eaef : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59ce457592a4c1eb4b92289596c7078 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac4dae02ddd7406382624cd352f1a184 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ed950c634484cee85ecc23dcdc9182e : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e109b08f526a433eb3d2dded84df1cea : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f348c4d588cb4abe800e74d57c64b7a0 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/148050dea3854c10a63ccf8fa7124305 : SUCCESS in 7m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6eb2eb6b02e8413f9f3ee35c7b585933 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6cf3ebcb25343bca98cc7d2046b2e46 : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9af1c12d52ca4f349d0199cbd183404e : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5157e789f3184a83a3a91f577fae6874 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05edc55772014ce3944606f1d61f5d9d : SUCCESS in 12m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/84e424d5ed6f44e0aed88c794d4625d4 : SUCCESS in 26m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d2a6b5f85f8b4c359f554ad0a477fb04 : SUCCESS in 11m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bae3ec74a66f4693886e098c3497e279 : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6274efb95e24003b77349891a67fdf4 : SUCCESS in 16m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/298a231b78b6428cb66ae436254d112f : SUCCESS in 6m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0fd519e089144cd4a72b4eeb56e80e8c : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e05250a27730431880bd5d3de10d2a11 : SUCCESS in 6m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c9d5d1f486e34d01b28b40b3165b44ea : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab832546383449a087806d0c17a9f556 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dd3fe4271624bc1b61a92c74e0b0b7f : SUCCESS in 1h 00m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/608a1b052e8d48ab8e9bdac4e845ea2a : SUCCESS in 1h 15m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99addc5016ee4b7f811e92d71053bc64 : SUCCESS in 1h 20m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0704234798954b1c878b4b110628702e : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":3},{"id":"8d0ad3e320ba3c909e3c3d7a31970a3568fe8fef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 00:56:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5982e4733d35b162b9184383027a12316ea3b96e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 00:57:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75ed745c88b4715ab5078b4bb350129 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a8995a8b1123a53dc1da594f16d60d49c6f99ef7","revisions":{"d1ad10b0f915d9399e713e24dde8a31b1bc5330f":{"kind":"REWORK","_number":1,"created":"2020-02-13 11:00:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/707616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/707616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/707616/1"}}},"commit":{"parents":[{"commit":"3479259d7ecbfa23f12220d9b745db87b68116b2","subject":"Merge \"update constraint for python-saharaclient to new release 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3479259d7ecbfa23f12220d9b745db87b68116b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 11:00:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 11:00:10.000000000","tz":0},"subject":"update constraint for python-senlinclient to new release 2.0.0","message":"update constraint for python-senlinclient to new release 2.0.0\n\nChange-Id: I86f8147caa4aa3f170a0c3ef415edff18c67cd87\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I27a379543ef18859f0bac189090ea76722411b0b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: XueFeng Liu \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1ad10b0f915d9399e713e24dde8a31b1bc5330f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1ad10b0f915d9399e713e24dde8a31b1bc5330f"}]},"branch":"refs/heads/master"},"c24d84596d7f8a659d4958af101c3980fe2e7b6d":{"kind":"REWORK","_number":2,"created":"2020-02-13 15:51:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/16/707616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/707616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/707616/2"}}},"commit":{"parents":[{"commit":"2f4090a70cd6008a808473dfa765aa01575c28b9","subject":"Merge \"add more caps for recently released libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f4090a70cd6008a808473dfa765aa01575c28b9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 11:00:10.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-13 15:51:32.000000000","tz":-360},"subject":"update constraint for python-senlinclient to new release 2.0.0","message":"update constraint for python-senlinclient to new release 2.0.0\n\nChange-Id: I86f8147caa4aa3f170a0c3ef415edff18c67cd87\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I27a379543ef18859f0bac189090ea76722411b0b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: XueFeng Liu \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c24d84596d7f8a659d4958af101c3980fe2e7b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c24d84596d7f8a659d4958af101c3980fe2e7b6d"}]},"branch":"refs/heads/master"},"a8995a8b1123a53dc1da594f16d60d49c6f99ef7":{"kind":"REWORK","_number":3,"created":"2020-02-13 15:55:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/16/707616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/707616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/707616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/707616/3"}}},"commit":{"parents":[{"commit":"2f4090a70cd6008a808473dfa765aa01575c28b9","subject":"Merge \"add more caps for recently released libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f4090a70cd6008a808473dfa765aa01575c28b9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 11:00:10.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-13 15:54:57.000000000","tz":-360},"subject":"update constraint for python-senlinclient to new release 2.0.0","message":"update constraint for python-senlinclient to new release 2.0.0\n\nChange-Id: I86f8147caa4aa3f170a0c3ef415edff18c67cd87\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I27a379543ef18859f0bac189090ea76722411b0b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: XueFeng Liu \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8995a8b1123a53dc1da594f16d60d49c6f99ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8995a8b1123a53dc1da594f16d60d49c6f99ef7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
