)]}'
{"id":"openstack%2Frequirements~707702","triplet_id":"openstack%2Frequirements~master~I5fb35d0c3fd9ff0a29b33d9611125f91c3a4f8fe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5fb35d0c3fd9ff0a29b33d9611125f91c3a4f8fe","subject":"update constraint for python-swiftclient to new release 3.9.0","status":"MERGED","created":"2020-02-13 17:43:10.000000000","updated":"2020-02-16 01:18:20.000000000","submitted":"2020-02-16 01:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707702-1581815809660-4edda447","meta_rev_id":"1e3e296e58955ffa857dc812df9001ba3c8faa8d","_number":707702,"virtual_id_number":707702,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 01:16:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-13 19:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-15 23:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-15 23:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 19:16:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-15 23:42:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 01:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c66f689a0e02507f2918d039cd07574e014b41d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-13 17:43:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c728006a532b1b9d74e66302d37fcf8e841e95ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-13 19:16:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ec3ff748396c48ecaea24a85a2ff2b136b2a8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9840b1cf0c45b7950586dba7786f4c : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52dade996ea842a486f4444aa1af7255 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ccaf5a64c9c42d7a2f12be6bace6d09 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a672c38024c64fb8a590c62732a010d6 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd12636091044c298474a1d46e222d81 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02973b9f56844d4cb57ec80a08512383 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8de4c7d3bfa4506832080cb5c114751 : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c438aa96e1d24456bec3dbc708f949bb : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b8502f179ee4288ab034a35b9d9e936 : SUCCESS in 11m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e70607ca451a4d7ebf9373bd4f7b9b70 : SUCCESS in 7m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/928d7fba0d994e6483af13806adf5982 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2004f08a19f74f4aaa5d968b10e72d2d : SUCCESS in 10m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/209883e972d64e28b9d21743fb1907fa : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8b141a2a668b47d2a8700fd9e0e15d7d : SUCCESS in 29m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d093779efcfd424f866ce497f9e83a62 : SUCCESS in 18m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/67cfb063233741a1b3590d4319a0caa8 : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63fab4b8231849718767897247217e4c : SUCCESS in 18m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c55a5d9739254979bcda52af3c1ad7a3 : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0f2b6efcf61446b3937b56a0ec0724f0 : SUCCESS in 8m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87766053109547e58c7babe8dcbd7b3b : SUCCESS in 9m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b4996c852e4418c929ab70ad7650893 : SUCCESS in 16m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/64e37381db874cdd886d341ef42058c6 : SUCCESS in 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5af68c7917914adcbd8a596a9db3eb80 : SUCCESS in 1h 14m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b09dcb1dca824ead98c3f5d7d4038d4d : SUCCESS in 1h 13m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e9cd4f11d13c4561b00e2f53900841c9 : SUCCESS in 1h 28m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/31a103f56ecb42409671d0bcae78f3fc : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3d3eaae7fe56b2c97a117944b8d8704121ec8d00","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 03:24:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"462ceb6dfe66a0105fa115d28dda28a4ff69305e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 03:24:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78a66bcc9c22652fd8204abf6e5ca69e2a405738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:50:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb8c7061e3f4337ad78b3e669440ca4 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/509b6e9e364149ba9f9afa1fdd1cf3ca : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2aac930446d4eda867c64b3a2d51dbd : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1531023d147f41cba3ad1c6ff52d6111 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ee2bc246e784aa5808074343ddc71cb : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/842197686ef340b49b6b0560f06c3911 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/782d2a7095954f509e28746ad1409172 : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d018d5e0296497183d18301aeb22248 : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41b9d5feedc44ad3998c04ada155195a : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9cb7672ef89b41b8a9e7e399f7e9443a : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4af168a3f87d410e80cc1b27db87344a : SUCCESS in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f2ab3993e54443abe2935f38ac59ea1 : SUCCESS in 22m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8371bcbe75aa4250b2d893238bba86c4 : SUCCESS in 25m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/73fd304cccb848308f3a9e5fb9b4ec0f : SUCCESS in 17m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b40ac74e13d94aa3ae95397a13cc9c1f : SUCCESS in 18m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13295880b586431c8f4f3b0d2043e8c8 : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e4871b64f048438994015d785556a6cb : SUCCESS in 5m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bf8ba755f4e0473b9cdbad56cc61aa5d : SUCCESS in 8m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac0a012f3df04f419373ed0a46378a04 : SUCCESS in 7m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90018d37dbf140708d306b134f02fad6 : SUCCESS in 15m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/79db37f2bdcc4870b18207fac2b1faf2 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50e25b40cff444aa92668b6d86f8c818 : FAILURE in 21m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a25105a953a74ec898379bb57b532308 : SUCCESS in 1h 20m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/febbbca9986b4f4792ce51b8b61042ac : FAILURE in 15m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7d32fc6aebbe4cae90b230d5eccbbb79 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fc808cfc4519e520e838e2b8645e0029ad2df01c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 04:51:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a6b3c189763217778736e3aa701952792fd201ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 06:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27de6ae2d4bc46229660ebf8ae785fc9 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7e975e0674476a8f9d59a3d74fd08f : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dca598ea91554954b4c91cf77d1107f9 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1f4dd05d28244239c997bb69871e42e : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75967f4ab4ea47338f09ef0fac43d743 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33a981f4a993483e80c5d4b3c1d40d00 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98c817c021a1495e92e647c184a67114 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/08bcf72c07544cdb9f926edb7ad36a9c : SUCCESS in 7m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/658f8cc67f504a07ae4fff8772ea72cf : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c7579273f4744efb95518ff062c9cf4d : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/14fc0d84ec2a4052a05ca88829591e9d : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce357fb20ae042af9f668edb2fcfc4bc : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/387a4dc96a2b4013be5e21c9c47f23f4 : SUCCESS in 13m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/73f36f84ca514401b75e55406deabb26 : SUCCESS in 27m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf59a3bd773e44b3ab7a9391c7ec8940 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/30b417e0b2a748c9b7a8b71b4a52e3d6 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbb966ab11284fef884692ff60716351 : SUCCESS in 16m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e520cb6622d344adb3387fb35b53144c : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f6953ad3ae354aad88e5de4023c50e9e : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17075b0189324e6290c7fd3759da37c6 : SUCCESS in 7m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/abae0dfc281a45d19133de77bad8d8e4 : SUCCESS in 9m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b96a9549e284d33a0b75c6b702858e0 : SUCCESS in 9m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f60e9f740bd24cfeb4eea813d4fc29d0 : SUCCESS in 1h 00m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f8a0420316942b8ab4faee2fa82e272 : SUCCESS in 1h 12m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/31c007b730914514bc874674eaadab9e : SUCCESS in 1h 26m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1034711c83464d6aa96f8abe7241f7c9 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ba82396e0f1142d42b0bd83c676bb2c5cd836d4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 06:22:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74d7f79b38bba831241b1317976a15021a0fbe3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 08:30:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58968bb031f84948a1a94539697d5d71 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb800862a0bf4e568b0494b507b9f996 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5f85ce68c134aabafd89d5e14d6b655 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00ed42d62dee447488ed0d5b66717025 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d751fa0106b4401eb6a99009f56fa567 : SUCCESS in 12m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dba2351009f24a52b8b67e72b28e4262 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c7fbe0bd52c4c57b8628c4f23dbac10 : SUCCESS in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/409e2ae4a8194006a66b89ad70d45e16 : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e83ad43f9994ae8a7be37bfa3d043df : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a212a2b568674b4c95deb9bb531a990f : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/846598e47af34dc9ad740acdcb82327a : SUCCESS in 10m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/41f27dbf31dc4e0c9f5cba5415fcd7bd : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/342554fe588344a59af6636dc1aa8746 : SUCCESS in 34m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9ad86340dc3468f988329879fbf9f2e : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6e4040e01fb54c09b160daa6a7a5713d : SUCCESS in 18m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef0df141cec945028861a74e9f83232f : SUCCESS in 20m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b955e506b07a40f388b0099753d3b77c : SUCCESS in 7m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/975692ea95fc48af89d3522abd1dbced : SUCCESS in 9m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55d8e9a007ab4270a2da7ff212bf4e4d : SUCCESS in 9m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b7d85792abce44b3bf71e37da09a5a64 : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a815a359bab743998f87a2159497fd80 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db034a890e1441208534b59edb9d9880 : SUCCESS in 1h 11m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/010d912230c04ae4b59570cc423793e5 : SUCCESS in 1h 27m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a896583969824d2f9e6ce83ae6d78ed4 : FAILURE in 30m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/14701564b5bb46fb8188e4c35d8a198b : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"50f3bdfebffc14fdea63c57c9e4555af1ccea77a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 21:57:00.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"377a5f6279ee9734141a25fbbd55fc17b1af903c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 23:16:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a5c6a8ef5449f6ab820b7c4f343a3d : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac7e0175f3c434d9c0d7acac4b42c32 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b75aab979b24c70a67daed4060c1833 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57fcd87fb4fc421d9caa77b5feca6401 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a77aea52faf14889a6fe83c827190e4c : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5e4ddd569eb4ca1a40faef2527ec8d7 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2512ec4d765a426b878a2d9a13dac7e0 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8a3b9ac883d41ea8990321481aafd28 : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/298e8eeab8e7441193c53b9e45f27f68 : SUCCESS in 12m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a958fdcac686450095dbda7b8031df24 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1cc6020215d34a2ea9dfd2b351f9654d : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13b009d64e8645bbb94d039bba9a5c43 : SUCCESS in 8m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/25ecb2793f604506b6647eedfb034519 : SUCCESS in 17m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b0d0c29ab5bb49d0bfe1d9320c2a065d : SUCCESS in 23m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35598bef1369477d8d883516ae5a702b : SUCCESS in 13m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4d1aa907d702486088b3436dacb2ab31 : SUCCESS in 14m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/785802fe760445468ff1d3665c2adea7 : SUCCESS in 16m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c4aa0d14b5e84f4d9b5bbc598a9b0e00 : SUCCESS in 6m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3e3689fc57f7412abb5927fa756d6d3b : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab7502256bf7407aa33147344c459fa1 : SUCCESS in 8m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96e130de952a4e899810c4ecf9b11d0e : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03e47410e39b43e88159064ed0f8874d : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f7ee75f9f9641adb4dd092990119861 : SUCCESS in 54m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0283db8ef8ca4ba490a6f3e528607b2f : SUCCESS in 1h 11m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/653f6f55673740e79b00b8d320c9437d : FAILURE in 16m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/36b1e4b8aa5e402a89bcd7afeafcb722 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5afc6926f86d620d90f3ddbb5ef654f77f280141","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 05:50:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f015374b521eb618a2acf0e32dfd2605a74db67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 07:18:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd9e6efff984a49bd99adcf36e71519 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb16d300c33490ea120d3e7c8bd81b2 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f29c4ac9fce444269ac9f2b40369f6cd : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ff7a790dc6e478680b2f0f382d5df71 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53e8283740e84dc996b46ee9a47e011a : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48e5e56b5ce14d8f9e40450f2072a3dd : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb224495566d48cba316b582d6861ec3 : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d32aff657716471086f59de9d32e8595 : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a3f223ce98bb44e6b72a33e2e92476d8 : SUCCESS in 11m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a85b8963702c4ddc9134697b395d13f2 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a58f182bd4c74de09d723b26a4acb670 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b525f2c7cd704841a43ebf347410cc4e : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/133352ba05fe4d82a21cbb5be1940f60 : SUCCESS in 18m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4acb93880b844a3e88b6cd77e332c8f2 : SUCCESS in 30m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c5f058080f54474b810a251e836e503 : SUCCESS in 15m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a2d76239276941a1bbf44308dabeaacb : SUCCESS in 14m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/425b7d07aeca417baa88d77ffaf8bbc6 : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/670c334f5d38410284f214146992bea0 : SUCCESS in 5m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6725c4bf6c2a4cd9a312718b4faacf35 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6371cb64f6b24971bbcd28cbfd2dabcd : SUCCESS in 7m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9c34908000034838998e180998096baf : SUCCESS in 10m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72ce7400aeae46ed8cb7af3e0bce94a1 : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/289b75760db2402f91d2891faf7a0b50 : SUCCESS in 54m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80cc3ecfe7f740f980d3ddeabfcc83ae : SUCCESS in 1h 12m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eeafe31c2a564d7299d319814b14215e : FAILURE in 19m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/14f4a593e07f4f4cade2192b16ebef1a : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"435a7155a2fc84da550470becf27bb1d86926f1a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 22:01:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cd7682cf5542b0014a4cb744fbfacd2c3cd638e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:39:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a11c7c3b3f84c67853bf62ea3ae4f9c : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b7f712087f4b51a382788d43a79f3a : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7c0544e78b480fa5bf93a6a5360072 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61ef183427ae450586be71eed2584fd7 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a175ec5c001a4557a4fb180f50ca1b06 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c59aecbd3f7449b59c987a5b9c7d5825 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c49b703c1b046b6b46a53f857d3a983 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/afdb78b57d0a4f63a5319265b5ac4c8a : SUCCESS in 7m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6f5b104b825e44e09ab3d4cb433b190e : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/edd3cc7933a14e0192ee27d86ca0f1cd : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8489f925f5714c78bd74a15c217a3fab : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c98475ea46c4a778536ad00753c760f : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2f9d337de2d947e5b5d90c534f7691c9 : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/87961798ee9c4e79897a43e9857afefb : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c1fd09afcb9849fc97bff480df2dd389 : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/27080a3f553c4b1b9f6b8c4320d66383 : SUCCESS in 17m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e1fb14c077d4f70abb70b025b486de4 : SUCCESS in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d40cbb2243344f6cb089ab6a54bbb5c4 : SUCCESS in 5m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/287ede75aebb4bd9869727fd16c24346 : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d9d4fa809c44dd38d5ce769d663498b : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/625cfa4b8552431fa27000f83a54e93f : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0ba598d74d0f497bb22164e8592ca028 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e68e34c1a0a1421b9aa6d6a5bbee7ad7 : SUCCESS in 58m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b00aa9f30974dff92987bef155dbbbb : SUCCESS in 1h 11m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85eb530c251048cdb801b8ade6558af1 : SUCCESS in 1h 14m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ea6e8849952b47b0ab1d7827d1c4e2a6 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6e0d0cee14d0a9dd2e73a4430d2a8421e9ca3b0d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 23:42:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea92e4d19e5bac08cb437635985260849176a2e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0191bdf89dc7740f197f853c492c6bfa648ffc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 01:16:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee94eec67b24951a127b7031ca2c8d0 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e63138ac90424a24a32191daaf9a3a0b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13c6beaee9914f6b99b57e7118dc5de6 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50bbbe81e5f444608e3c0037ce674884 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6aa81c580234251a81d4c585f0bb835 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75202cfbb12e4b3696243a463d272b0a : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db70518639974444bd74cc22968c5c18 : SUCCESS in 7m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7184c8d5dcca4e8780976b71d04a2c48 : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e90bff5664d64331a290303d73e8f105 : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aab95b68212848e0a548fd30972a15d2 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/814dde7279a7431ea7ff13f35ade605a : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7bb1cc49fc24811be3fe594fef6a61a : SUCCESS in 13m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fdbaaa0c24a54b20a4c8cc6d3feb588f : SUCCESS in 22m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c746748f6e82447b8d65598da51284d8 : SUCCESS in 16m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/80e417fc342e4cd0ad08c985ea962259 : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4807be38522b40d88248ebfe9b572170 : SUCCESS in 14m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d055d86440e46ae8144c9e588cfdfc4 : SUCCESS in 6m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f69610382b9b4bbc9f94916fe9ca2d50 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e618c1894794c509df56fd0cab26d4f : SUCCESS in 7m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/941cddeb01704f4b97a5eb622c340344 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6d173e1bdbc04eb48704904215bf07c7 : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/920150bf84a74f5f88dbd55e867d6435 : SUCCESS in 56m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8edd25b03db46d9813f227b89f34715 : SUCCESS in 1h 18m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9670cf16b994fcb94c22140d33630a1 : SUCCESS in 1h 15m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0afbe12679be4d5693857e4eef2fcb4c : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":1},{"id":"43ddc95f658539b0d83bce076e5e0b24626da2f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 01:16:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a859f0c5d30ea08ac3cc0cea715062bb58a29810","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 01:18:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0daf26b9635b4746af14c751e0831aa3 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10fc70c34d3511acc12e3e5bda8b1ed28c2b8a31","revisions":{"10fc70c34d3511acc12e3e5bda8b1ed28c2b8a31":{"kind":"REWORK","_number":1,"created":"2020-02-13 17:43:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/707702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/707702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/707702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/707702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/707702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/707702/1"}}},"commit":{"parents":[{"commit":"2f4090a70cd6008a808473dfa765aa01575c28b9","subject":"Merge \"add more caps for recently released libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f4090a70cd6008a808473dfa765aa01575c28b9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 17:43:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 17:43:06.000000000","tz":0},"subject":"update constraint for python-swiftclient to new release 3.9.0","message":"update constraint for python-swiftclient to new release 3.9.0\n\nChange-Id: I5fb35d0c3fd9ff0a29b33d9611125f91c3a4f8fe\nmeta:version: 3.9.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Tim Burke \u003ctim.burke@gmail.com\u003e\nmeta:release:Change-Id: Ieb6c55d20bc122c366495cf0887cbd59cfea5f22\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Tim Burke \u003ctim@swiftstack.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10fc70c34d3511acc12e3e5bda8b1ed28c2b8a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10fc70c34d3511acc12e3e5bda8b1ed28c2b8a31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
