)]}'
{"id":"openstack%2Frequirements~707717","triplet_id":"openstack%2Frequirements~master~I1df5f791a6a4476273b2df9f43d3ede91206013a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1df5f791a6a4476273b2df9f43d3ede91206013a","subject":"update constraint for oslo.i18n to new release 4.0.0","status":"MERGED","created":"2020-02-13 19:54:29.000000000","updated":"2020-02-16 00:58:54.000000000","submitted":"2020-02-16 00:56:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707717-1581814598929-a5ae4dff","meta_rev_id":"d5aac2fb4821f5a99015e5c9a2210235e974fe2c","_number":707717,"virtual_id_number":707717,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-14 14:15:56.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 00:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-15 23:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-15 23:37:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 14:15:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-15 23:37:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 00:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b394c603a178592ed3b9fd28a2bad33b0a7da658","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-13 19:54:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3221c954de4979371341858fe19e4f69dbce6989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 21:39:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6d70ffb14c4886a5f65b5827bfee7a : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e9b1149532d4d1f8fafd0562ab41aac : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71c93a9cbb5847d8b6c718ea45f6c8d2 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02fb2af657124cf3b13d96efb40af2b1 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9749ef202a5a4c0cbf8034db2b18d6f1 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d33f37fc9424fcb8777ec79e2b721f7 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/415f3fa5f8614fc4ac78a1b88e2daf1b : SUCCESS in 6m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de12a1b94e794b38bc78bd3ffa4042e6 : SUCCESS in 10m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d42467f8a5f447f8e5d3e9ffe04af5c : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c61ba3fb22de43d09a188660beda7e36 : FAILURE in 7m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f57670301dd4abc96b3bafe70ae91ef : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b71031048c86469f93b83e89ee06a51a : SUCCESS in 8m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/584ab4dd0d7d40d19d5979191dc74836 : SUCCESS in 16m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/151f60db27684cf3953d9526715e3742 : SUCCESS in 29m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c945144689e476fa698c18cda5854b9 : SUCCESS in 14m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a123864b3e1848b890ad54ffe3a650e9 : SUCCESS in 16m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f84fe448adf419eb8eb58b11f5d42a4 : SUCCESS in 21m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a4771b49118f49d9a8d1b8543f3bfc0e : SUCCESS in 8m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4e6aaa07b20243cf828e198f91af401e : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75bcd644c19a4c1f866243146c2a5dab : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/58dd9978aff64befbcd5a8449324284c : FAILURE in 4m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4193a82008c643d2a74df59083576c16 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/343a16c8f20e43c688d7271ade6f1fe4 : SUCCESS in 55m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/575e6591f91746c0a070883d89ce7031 : SUCCESS in 1h 13m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4007a7db384e4114aea8927c0a179fa9 : SUCCESS in 1h 34m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77978bda9b144e96b3963c6cbf433134 : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b5391b072aded8ccaebe0f5a78f9faaf18ca7c62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 03:30:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d005f95c119d8c172f6a29c3e3c6dd86880b95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 05:04:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6995c41a0ceb46a8bd02bf67a8897495 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a9f933d19e4c34a93488e8bcae82e3 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bc38f414e684027936f4af90bd62d00 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aeba55a8940546c687e8cb9b0d7ea3fa : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2904f9587e904e398562a81dcf43bbb8 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a25e47970d34a75810e617f65aa4156 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/002bce96cb124753be3b7ee126da9859 : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d4061d57fe144b2e8ec04a0289a864cb : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c9525f844b254e19bd180c74d4c8aa79 : SUCCESS in 10m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb364f623e0e4e60ab68b2dfe58cc042 : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a748e986c8a460185cd9784b7496a2b : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ebd62286303b4cb38b17c2a0b261ffd3 : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ecf0d054c7e54f01844cbe92123425d1 : SUCCESS in 26m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/918a6176517848368ea27ebcd9f89f68 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97b970361a5c45e49c886d415df78cae : SUCCESS in 14m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4e80c26be7e045ac97ad7a6e15e9e7ec : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ef04d2a28524213a9dad1154f64eb0a : SUCCESS in 16m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/74953534ef484d2d85c5621f3d66d66d : SUCCESS in 6m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b117d37bbc674dc1827037b03ca96c74 : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73b5e65a074a4ef882a1f30be36e0bf0 : SUCCESS in 9m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/50d30f867dfe4b05aaf09b156725db0b : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef420f0e6b044f6686fd2c0990ed7d29 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f706d214b7c8444186601a4615f37011 : FAILURE in 17m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ee10c9ae7344d6eb01593a216c19f7a : SUCCESS in 1h 06m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3aa4df5416f54993ad926f8b428ab50b : FAILURE in 16m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7e1f3e949a274497b1b51068cdc7a553 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":2},{"id":"89ce121ec5de14e28b043e6818d336cc01615d7c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 05:08:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da415cc724bf4cca750b021f216053ca6c13ecb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 06:36:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f6fb6514be4b8ba1cc3f8b66ed2dde : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fcbcca5c7214f7d8e6be6c41eb0ed12 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/935ab626f7a54a16af0ed9fb1ab090b8 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc08915c077b429bae1e8f61cf34c7ef : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2053bcda69d4a17a0ca7c3ad7896ae1 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/268dd04cd8514e538a93f6a51e2466ac : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae64569d16244484a221aea510fc10e1 : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/25f3625034d24e74a2e9dd3edbff7189 : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c4936f1dfc740a2bb2c1770a9279841 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/01704117c69142c4a573fb857a4dad1c : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cd209f80a8bd4e94afe4359aeb9999f0 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d06b649a21942e4b0190745f2d37d83 : SUCCESS in 8m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1553bacb2014496ca947149c59980bf6 : SUCCESS in 12m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/23b10e0b7f4b48549b8740b3fea7c5f1 : SUCCESS in 23m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f317f2dc1e2a4e3dab434ccf998a8a37 : SUCCESS in 12m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bba3015c896c4633aae3d3f184726563 : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33800ce37aef468ea4257cf6fe2b3696 : SUCCESS in 16m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/73a1127d75f94e369a75c42e82663cb3 : SUCCESS in 5m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/32d9e6a63e094f0d97f8282d7445b570 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fd68dcd817f44a082adc987ffab498c : SUCCESS in 7m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b2bd754985834f55b4662b4a37936629 : SUCCESS in 14m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/31a5cef79a6a4fe58bc86d80d971e411 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1a06311053e45aaa224e1fe8a20a09f : SUCCESS in 56m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcfa3f5ea98e472998b4b0abb9e9e7c6 : SUCCESS in 1h 09m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b338af85885496cb73a8748abc1446e : SUCCESS in 1h 26m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/162efe98ff234f00a6c6220d72deb0e0 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":2},{"id":"53b3eede4a721bd7e721bdc0ab244dd2f7843829","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 06:47:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c75d4906e30923e6faefb4d6d24f2e5e083bcb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 06:47:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2607799f1ba87ede313752bd3a93fa9def5edf53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 09:47:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73d2e621b594dc6b0b1c32d9b277a07 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c034bffa8fd400b850eb3c237f76c8a : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75ab5b0751a44d2791848f9f4cf6c532 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7021eedfb2d4f58bf0c5a7ad8c0e642 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a190f4c7e36d472d8e57dffc731ca209 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7be375dcef54fcab4be98796a71af51 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/076b77f38ce9491590de671a0c7af58b : SUCCESS in 8m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd86eebdd862495086c873e99dc1d00f : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/88476c3985f143dd8bd0cc021dbc0e34 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f2ea0b73bfc340edbac607010f11624a : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3a0a58eacc140f38393f512de183a38 : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c48b8beccc9a47939d7f3a4214fae142 : SUCCESS in 53m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/089b151c176348c78361f621eb885de9 : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/86df422a5cbf430194e6b31c5fbb0f30 : SUCCESS in 13m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/576509197da94129b031d71d5d5b102a : SUCCESS in 16m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/509a501318dd4c77a1fc6489f83c82fc : SUCCESS in 19m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/442e01c3a73f43b9ac2cc2d14fa7d7c8 : SUCCESS in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f1a6f973f59e460ea6aa278e27cf3ef4 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f9308f380684ca59d76da6026f4d732 : SUCCESS in 8m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c61a8e5bfa5941e9b3011634e801e5cc : SUCCESS in 13m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24b4951d648b448782ae5b18c85ef86a : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b698299c03a243128151bdc97808fc79 : SUCCESS in 1h 09m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bac971ac557f4c2db817d604a4041382 : FAILURE in 24m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1ee58b018918495ab69cf65a9c0e26b2 : FAILURE in 14m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7c9123d19bc2439eaf84e2572fda9196 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":2},{"id":"7b8bc0723fcce76d8a760c481bfb7803961c8765","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 14:15:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca3f2e76bbd3bec7b2021aad0b963f4c0ad9d264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 15:44:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2c30b3fc7e4c4eb142fad4a3d6cc08 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4e8c72a3844709b9d8b70022458d9c : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4542f9ca4d2247a58e8e4c24c3f50fc4 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31b88128b6744931b09824d64af691b0 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b73678144cda40a1bbaff6b97b694c94 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae27fca7bb74483a8bb3b256f18e6c74 : SUCCESS in 6m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d5f722e922e4bae861fff4e1269c888 : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a3d5c369e884e6f97f2962b8aad37b1 : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cad66f8453124b00be5b808e83d0fd88 : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3055fd0e84424cd185c1c3c2fa83936e : SUCCESS in 8m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/71886692d2af4bd6a813d376d3bffa40 : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcf8e8ac1f2f468f9fe0a6bed8086dff : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/13f2a98f003a407bbf1af961bed100e2 : SUCCESS in 21m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b5aa8e33a3b04ce2b0b136a83388826e : SUCCESS in 24m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a067f12ba66a467d8932d17f83ba5e6c : SUCCESS in 16m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f85f0a5f832b4fadb79c4e0f27a38762 : SUCCESS in 16m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18a41b3013914f49af5e6b5566216221 : SUCCESS in 22m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9e64511e214147248709e177dc509fd6 : SUCCESS in 6m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/553bbd55f7df4e2ca968a593a92229ad : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3107aedc32e4c808fa479b567cf1be2 : SUCCESS in 10m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/93a9d5090a3d4012bbffaa8e1f41da80 : SUCCESS in 12m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/073c7b5ccb3b43cf9dd7a3044d86271b : SUCCESS in 16m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acd743926965406a92fcf24a987c15e8 : SUCCESS in 1h 02m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da3ba1b9604c4a82b35ef82b3495f01b : SUCCESS in 1h 16m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4a3850aa20e645919ae8f097b7dff4e1 : FAILURE in 16m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ca97090827d4643b944bd7f481be763 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6e11733592a6ec6f859d851f0e7fe46740b0c5c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-14 21:56:50.000000000","message":"Patch Set 2: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"924b8fc9ba7fd983541bd04345f3d296f561271b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 23:10:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b665e8fba4240e09a4f83f7eda242bd : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a9309bf04c54e74b66e6073f226df13 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f35ce9f0dbf434299c86e3fed06aaed : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aac226c3aa3648bdac7f3fd2e3502604 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e1be531602843aeb25a15bb47bf64bb : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff602225c147492681f03b12402c2f98 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0810ecb4d72e4c1295762f276a5de4c6 : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53b5bc9cf322489ebd29ee68c0a07e9b : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0c0f3e1c8f7542f48c3578190bd1ea6d : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b92a95d5b044ec1b7c75941a35a9744 : SUCCESS in 6m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/001f6bcf45074f70b78b0b87c9b5af55 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c37958c03c994049865ac3bf85efc7a7 : SUCCESS in 10m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/347c1a5fa41648198799f71cb07ca69f : SUCCESS in 22m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fea0fc6635fa4cb9b4746fa62a45ca9a : SUCCESS in 32m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7396c02fae414325b31a29cca5664fe3 : SUCCESS in 15m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a451c0c2844f49c09e95db70576165c4 : SUCCESS in 15m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aca6d0739d944c908acd7b755b318d4c : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b755ef058fb44f2e8f1b244c3946a1be : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8038322c0dcc43c9ba422de09269437e : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b418316f53ad40e2b8afe891366b90e3 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dd02a6142a074dcd972686666a41172c : SUCCESS in 14m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec8166cc4fe048b38526e006871d9a77 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48dccdd572134d0a91f3c9c4d2e7d139 : SUCCESS in 54m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e8a52907f0c45019aa7a17cc310b0f4 : SUCCESS in 1h 11m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b15b32c5bff45c2a1a3e0c6f1100254 : FAILURE in 16m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3931eb0e29534de79aceacdbd532653c : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":2},{"id":"70faaf7754e884489dba72df8a0a4c4ad4b438c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 05:50:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"feb9d7b6146a6a3ce9a4def494e4be89551f9046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 07:10:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ba40ace71d4559a79c9164478a1d28 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a36ef800af842ec9676b4f8e4d1c9d1 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/899902efdf684617bcd05ea9a03ee646 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa424d71c1a943bb8c9a614a884119d0 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ceb2715ad224700aa43341a99d57175 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a339f82eff743e6a31d5dc2f449a6b1 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f10a98d1153a4156989fe24194206894 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39cee112886e478a87427f1c60bd8416 : SUCCESS in 8m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ebd43c9dbf944de2b14851887d3898d9 : SUCCESS in 11m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/202a1c3bb3c94726841eed1e614d7ebc : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b2e25467e96f40f5bd1e5d87f8fdd899 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df48475a84ab45f7a12cc7c7c3b46b2a : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ada5b65217a349f691bbf3175c271dd2 : SUCCESS in 16m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/005eb472509442b68c0a9f9e1cd3703d : SUCCESS in 28m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/56295bd8a5c5423d889d0f8d4999824c : SUCCESS in 19m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6c3fc7d8ebdd4a16a697012a7f2092c6 : SUCCESS in 19m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0486886bcd3f4474a95d6c60415dc825 : SUCCESS in 21m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/32ee610290ff4e8796bdcc4e3060f58d : SUCCESS in 6m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/914e2b34e30f49a7b07c85e61b4af315 : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d5db8f7cba740a5a42ac82370f33961 : SUCCESS in 8m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/02e5d87b70dd416c96fbe105402e161e : SUCCESS in 15m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9f393d99bb64581bd8fbec3a4d40bd1 : SUCCESS in 15m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2415eca8ce1844d3a64b5751bb56fbca : SUCCESS in 57m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac3d2851de924beda3509d0f09ab7164 : SUCCESS in 1h 12m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/36d725dee93345f2bb04ed0debe8aa28 : FAILURE in 16m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0986248b256f4b87aeb94b38fd7b3203 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7ef044a9b46c1cea3ad8cd89b326542d89c4a9d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 22:01:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ff929d7e4a3fe629b2da010c3d2527744fca94f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:32:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9518e3db5b24534b8e7b822c2cc33fe : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f12a376d1746cbb2e7c6741904c9ff : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa46289edad40de9ae46e5436e8fd93 : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9efea759bb0e4666aab263a40d77cafb : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f815bd8bec47bd981ac1876582b921 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cac57755532f4133ac3cdaeba88ea1e5 : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6fc77580df0486c85a930a6704134f3 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/892f53737df74a1c81380cb32868c758 : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e5f80fa2b6248e3825b251dcfde7a83 : SUCCESS in 14m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c421175cdcb4a6a9d55978ae9cafc4c : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/484e116f97dc4cee8207a2fa9200c0a5 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e88f767b7cd4954bb0744713cc9edfa : SUCCESS in 7m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19773350cdaa402594e1133245289a47 : SUCCESS in 16m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/42a6209748c54cabbb92bf16c7735142 : SUCCESS in 27m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c7caa6a40e34001922ba0fa98e431d7 : SUCCESS in 12m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9f5c9bf93bb849a091a961057daa2eed : SUCCESS in 18m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/737d383810ec4fa49e075805c8cef0e2 : SUCCESS in 18m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/469adf873a6e46c1b67b184eeca0db70 : SUCCESS in 5m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/553ff454856a46e29b96799ee2fa4533 : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c84292a8dad4473bb703ec1cb6346d0 : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6f318a0ffe2b48e4bd975b32399fda64 : SUCCESS in 13m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba65d112d2dd4a0895f643078c359807 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b6b8aa174b546f5be2275c3826b2697 : SUCCESS in 52m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fbe2aa5a9c845579a27e1558b1d8eea : SUCCESS in 56m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81b9f6319e514060a416993ed4dc6493 : SUCCESS in 1h 13m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a30190c1260d46c98c4a8aa44eb09a93 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":2},{"id":"539c68e04d0b8b3387c21c05bd2b0a9c7a583fe4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 23:37:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0a76232bae08ebf14cfc7c058b9fbdd47ed52e0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:37:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"519227801867559493f90c310c5469d8156c0ba1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 00:56:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"edffa5460ec89338e0c166559a1efaefdf091cf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 00:56:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e241ff7992a4fb28e5fd1c05232c3bd : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c652dfe7c546b2b3d0896a2f558307 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd2aa45f4cea458ca3182fd635943cb8 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5492984faa264d3b9cbf0e842fe23414 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/126345c953ff45aabeddff09d70d29b5 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1263a1c3b36443efb56b4b5ada364f85 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0b37b5480b2421b9bd282b99a79197f : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/77b02c942f0849e88bbd1f3139b8092a : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57880502575e491abd7b0566a16d1523 : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/709d08c5d0574cd4a1579d0c32198c0d : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f40c5b4bfc543dc8913555f4d869d7d : SUCCESS in 10m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f4161914bb054d43b96eed15a65925bc : SUCCESS in 19m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/531094f4426a486591ffa0c18d8799b7 : SUCCESS in 27m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b42cee482dcf4c84be1c24c616944098 : SUCCESS in 14m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/042c553ffc9a4a3d88a94abef3f361c7 : SUCCESS in 13m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4314d422b5e4492c82c6f3e78976a627 : SUCCESS in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/357a39bdb70f484ca93b6a563718ed0f : SUCCESS in 7m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c048561b10b6492f85ab08911d127068 : SUCCESS in 7m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1473674fb8314684b0742882ff8c0ed4 : SUCCESS in 7m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2edd88e4016c42c894843c7644a7bc1c : SUCCESS in 16m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7d4c804a91f84f548e89440a8ef6bf4b : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2d441ce5d8a43909ee647e9a7d2f923 : SUCCESS in 54m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3d8810c94ea4c7fa3e7693654a37a0a : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0cad8ea99884b43a9b200f7d4cc88aa : SUCCESS in 1h 14m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9025c6555794d3b8390d9e41e272227 : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0ec9d013d102ed7d97069d0b5fe1879d2a3374fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 00:58:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7b67a4177f4a2ab51b12fa077bd34e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"346c411e46fb9b43dddbcefdb80657fac8801ec5","revisions":{"82d011fd68e1bad47abadc45689ace36e586e89a":{"kind":"REWORK","_number":1,"created":"2020-02-13 19:54:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/707717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/707717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/707717/1"}}},"commit":{"parents":[{"commit":"1cfdb9d527e66342ed762964a87120a168ba270f","subject":"Merge \"update constraint for os-win to new release 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cfdb9d527e66342ed762964a87120a168ba270f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 19:54:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 19:54:27.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 4.0.0","message":"update constraint for oslo.i18n to new release 4.0.0\n\nChange-Id: I1df5f791a6a4476273b2df9f43d3ede91206013a\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I5524048c0b318d03038431cacde558b0df0cb13d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82d011fd68e1bad47abadc45689ace36e586e89a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82d011fd68e1bad47abadc45689ace36e586e89a"}]},"branch":"refs/heads/master"},"346c411e46fb9b43dddbcefdb80657fac8801ec5":{"kind":"REWORK","_number":2,"created":"2020-02-14 03:30:07.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/17/707717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/707717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/707717/2"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-13 19:54:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-14 03:29:59.000000000","tz":-360},"subject":"update constraint for oslo.i18n to new release 4.0.0","message":"update constraint for oslo.i18n to new release 4.0.0\n\nChange-Id: I1df5f791a6a4476273b2df9f43d3ede91206013a\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I5524048c0b318d03038431cacde558b0df0cb13d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346c411e46fb9b43dddbcefdb80657fac8801ec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346c411e46fb9b43dddbcefdb80657fac8801ec5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
