)]}'
{"id":"openstack%2Frequirements~707917","triplet_id":"openstack%2Frequirements~master~I09eec256347eec765ee55cc804130280980a5ffe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I09eec256347eec765ee55cc804130280980a5ffe","subject":"update constraint for python-mistralclient to new release 4.0.0","status":"MERGED","created":"2020-02-14 22:46:15.000000000","updated":"2020-02-16 04:19:37.000000000","submitted":"2020-02-16 04:18:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707917-1581826688890-bbef5f2b","meta_rev_id":"a1b29074da54121e4a013af5294b59bd09f2f455","_number":707917,"virtual_id_number":707917,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 04:18:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-16 02:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-16 02:50:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-16 02:50:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 04:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aef8024e9c4ed94618aa05cddc0078999afabe62","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-14 22:46:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c30ed71fb58e3082e3bde64b53269cee135bf4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 00:07:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04f46bfb1314fe5a1d5aa07f140e833 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac7fb91790c4f9992ee0b35d8233ea0 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48f828f3353241c4b4dad389fbec86e2 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/634a6de4062f438492c01d3920ade061 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2db5b0d78f4746cea5a7cd5dd317cd9a : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19f8eb34d38d4d12b0435ba821fb96d6 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f8739db8238429aa3da8234937eb179 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b7a03ade4eb4742827d84cd6db2fb44 : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e9ae741ed5146dd82ae0ae3d7c9ea07 : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e037df5957b447984a89a9f096e7061 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88300ae63a3e4b55b504a18558a70e68 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0c3891d68d84e8e8ed112ca01a76a5a : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68735cf3f0be4121912a158dc4bfa772 : SUCCESS in 22m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5cd24bf0c7c244298cb2541e4c4ff28d : SUCCESS in 29m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2c4b6951a2844c379f4dc5593c836eef : SUCCESS in 14m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d1cabddb980c4a88b2111504c43d82d5 : SUCCESS in 15m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78631413930e4fc4af20729d05076b60 : SUCCESS in 17m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/50c6cdf7727a4c7898d72e501625f435 : SUCCESS in 5m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7d5ddae5df044fd28d5eccc1199adfe2 : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fd61e36898f4fb2ad765ca7ed6d145f : SUCCESS in 7m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a0148c882bcc4dda868400ea8e7b1bc4 : SUCCESS in 13m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/774c1f8c6e8b4629b670a197d02556e6 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53c8fc063a4f48e8b2a3f37f10035658 : SUCCESS in 59m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b3881d3c17c4e53b476e107b67a916e : SUCCESS in 1h 08m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b4c9244e22d4e42a24ed9325a8dbf25 : FAILURE in 18m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0bbc9d726f354f0bba3cb38912830d9b : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a355fa2d9f0428c6ed02e12ffff6d556409ebc1f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 01:31:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a76761499dcfa2cc92ea2e0f47da7c66b869eb14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 02:48:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391ff007a3c44f3f8f95e41ae9091321 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05af1bf945334941a94a4b3e867a1328 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5b7c660e00b484a88041a8defb24425 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb52c886ae3141cb88e816bd109bfb08 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ff7e400aaae48fa879df217858e2a12 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c794e43c93647ceacbab7362dc009b3 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3be054239f42418eb760616d2cadd0d3 : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5d498ba7d949474c9057211e71e7f945 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cb159c77f83248888c1ce91f04ec1c68 : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cad5351d8ccc462d8db3ab652bf27a71 : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bafe476505b44892b047b608a58d3472 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa2881a2bc644cdc84e556ff7bbe9930 : SUCCESS in 8m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b27d76703e442eb8ac0cd0349283794 : SUCCESS in 26m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/74d961d3253443fd9f7f4639ccc827d3 : SUCCESS in 22m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3881a27ed69a417fa232d4f24242056b : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8a45de8c0724e6aa79adf9190659749 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c458a3fd4c8247e1bead4a15f44617c2 : SUCCESS in 18m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c40afd51c1742b2999ad624e8582ca1 : SUCCESS in 5m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bd92474555d44bdc9c6b559773d8d8b5 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49b045e4bb924757a00c1bc539df655b : SUCCESS in 7m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1be471b69757434ea11ff793accf6f69 : SUCCESS in 14m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf8fec96b5cc442da9ffd65874f6120b : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9f57396c8204d1bb1fd4af2ed4df708 : SUCCESS in 55m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a02f7efdfece4c05b1b23c72f588bacc : SUCCESS in 1h 06m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/071330018132454aabce3936a09eb514 : SUCCESS in 1h 15m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c838c7f8d1d42a6a5f66bf2b4b236f4 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":2},{"id":"625371f8a95684aa9e8582e1617772f3aaf138dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 02:50:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67e4d75261ba3ff6c5eb6dae6d2772baadc21f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 02:51:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63d54d9409f1b8f25a54eb03b551e04eb97acbc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:18:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e538a4d25930458cbf6848e74e8558b4 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f9b79013cf4d1fb349b2adf6559ceb : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b68bf952234c2dbc0e6c9ccfb5215c : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77135c19d3f54945b1afd8ce45b5aa6c : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2211cd2f34394df081ced9a81fa06bcd : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6f93f389c4f472b87836ecaba2425e3 : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38e69a73ca634b4d81bdec0d663d1613 : SUCCESS in 8m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4cda371fc2e49bdbde8fd8d79c375f3 : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a46ff7caa1f4506b62ab9503a917162 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/025f9cdd75e240d69eee80b2bde6240e : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c491003fe392425cb6f420f94f6f956e : SUCCESS in 9m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05229ca4b0b04464af45d28b88067b89 : SUCCESS in 24m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/75682ffea9a8469aa01fc2a7e8b95524 : SUCCESS in 27m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2e63737056f46558f6197b2ccbf22f8 : SUCCESS in 15m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f51a3439690944b19f5e5fe25e8471c2 : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e55b94bc1bc142b9a1594bdcd38ecbae : SUCCESS in 15m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7e1df1a0df8948b4a0b2887af48e9807 : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2846a3853ca84f2daea407d21dffc1a6 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fb802ffe4b34c269e9bac32cda2af69 : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/be10bc3b7d7543789aee75b34bbea1a9 : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a82539bae86745fe8b1f462fe1281fb0 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a925274df6344768f92d2f6e08f9819 : SUCCESS in 57m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e64e91e17824624904fc9d19804b619 : SUCCESS in 1h 07m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e46c61b1522e43329ac4e2cd99812443 : SUCCESS in 1h 24m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/895b46ab4dcd434590abe47fe35ad592 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":2},{"id":"3078795dc759b435fafb72259018211283d78664","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:18:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bd4600dcf48d8241a6e9de5fa7241f461314c1d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:19:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4333d1c228644ca9879e33cc9d333e9 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5614a13736cd624058e4bf6c3d8d3e6a6bb3f36d","revisions":{"29a1ca1c00c3362b8412e8ab046893ea427c0b99":{"kind":"REWORK","_number":1,"created":"2020-02-14 22:46:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/707917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/707917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/707917/1"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 22:46:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 22:46:13.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 4.0.0","message":"update constraint for python-mistralclient to new release 4.0.0\n\nChange-Id: I09eec256347eec765ee55cc804130280980a5ffe\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iebf980d67a36e2709ffe2ce7fb954159eaa1bd46\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29a1ca1c00c3362b8412e8ab046893ea427c0b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29a1ca1c00c3362b8412e8ab046893ea427c0b99"}]},"branch":"refs/heads/master"},"5614a13736cd624058e4bf6c3d8d3e6a6bb3f36d":{"kind":"REWORK","_number":2,"created":"2020-02-16 01:31:49.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/17/707917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/707917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/707917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/707917/2"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 22:46:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-16 01:31:35.000000000","tz":-360},"subject":"update constraint for python-mistralclient to new release 4.0.0","message":"update constraint for python-mistralclient to new release 4.0.0\n\nChange-Id: I09eec256347eec765ee55cc804130280980a5ffe\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iebf980d67a36e2709ffe2ce7fb954159eaa1bd46\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5614a13736cd624058e4bf6c3d8d3e6a6bb3f36d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5614a13736cd624058e4bf6c3d8d3e6a6bb3f36d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
