)]}'
{"id":"openstack%2Frequirements~707922","triplet_id":"openstack%2Frequirements~master~Iba7794a7a44be27fe84b0218f66c0dcf4a131e89","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iba7794a7a44be27fe84b0218f66c0dcf4a131e89","subject":"update constraint for tricircleclient to new release 1.0.0","status":"MERGED","created":"2020-02-14 23:03:46.000000000","updated":"2020-02-16 06:23:18.000000000","submitted":"2020-02-16 06:21:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707922-1581834080491-b9790512","meta_rev_id":"91348aee701758fd2550426a081662325730b8ed","_number":707922,"virtual_id_number":707922,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 06:21:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-16 03:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-16 03:08:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-16 03:08:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 06:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e554b8eb592e1e49efbd2f2fdc37e9eb5dc01038","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-14 23:03:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69d1b166d9177ba056f128d979c3762a3b1b01ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 00:23:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3eef0736b0b46a3b54ae6f61fbfb859 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3316a4dc9394b7295e19fb16403034b : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da862882fd1e454890deecf55a1bbe70 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b373fcd0948146799de6b9bb62f5a5e8 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/507a136c91cb499f8b0e27ceb3d863fa : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fcee223c7a2474f8e878391a341bf43 : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0106f6319e7440cbffd1fb3c96e25df : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44066ffa883c4d5bafc48288cc9c64da : SUCCESS in 8m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90a56d87a42f4b65b15032e83b909b58 : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d36be0d2e3774922859146c139622570 : SUCCESS in 6m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3af9cbe33ba94a49bb245a864a45047c : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/915f4d57937748a288d361488f1ceb63 : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a329e1b04774549bd040b21b1896628 : SUCCESS in 20m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1559d907ad04e2195fc5289eab41013 : SUCCESS in 25m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe12178288ea404a93dd1459ba15d988 : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4002ce2610d54870ac2679edb8a630d9 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53fa42e23d6b438294678172457bc3bb : SUCCESS in 17m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fd9f93fec3384124a1592a1f40752a6e : SUCCESS in 6m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/81da522422b243e2b3acfbefe8676cd0 : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96ea4e5dc7dd4f60a740c731c1f3a6ca : SUCCESS in 7m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/276208a34c9d41d9a9aef7df32e33509 : SUCCESS in 11m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b6989704685439e8a0ede7f615d2a0f : SUCCESS in 13m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c05a9eb36c04e3bb561169345299601 : SUCCESS in 1h 00m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdcacdf6db1a4dec9c83ed9b48901175 : SUCCESS in 1h 12m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c31b7405f93e49349935d12b9068f7f4 : FAILURE in 15m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ea2235db1a30485d9849c6efcf0bf2db : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"43a541a526ef369979444516149effa389d507f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 01:47:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4434d6f1be1fb32677a4d388b0b0ef730703161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 03:01:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a419d01fa2cc4773b782cee8b92efec5 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478777fc03884cb5b7080af904fa0b53 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75541a9d415d4a59a29b31e63575d213 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6281338ccf3442d9e92073efd120fa9 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da0e8c07141d424d819c18dd2d2cd5a1 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4fa3e15afd048199931dc838e19efa6 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/346e1de705844f239011f647d7f1dfd9 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a80ca903c6d4659883c60dd1d829c09 : SUCCESS in 7m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8fd591f359c14423b67613b28b5454cd : SUCCESS in 14m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d153510722b1405a868da92e2fbd87bf : SUCCESS in 7m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fb82b2329c8b4daaa063b2d2a16dbfd5 : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4c3766672554227bb1386873e2561e8 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5f6ac690525c494faf233a848f1eea9c : SUCCESS in 27m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0df49aea31c14eaeac3b1d438f3155f1 : SUCCESS in 31m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/15a88d1c969540f0b0150eb2a9acb08d : SUCCESS in 20m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5865fedd9d247c18689d30b91ffa37d : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/124cf6cba28c49b2adafd185b26911f1 : FAILURE in 21m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/08402587e82b433ca17333318ab18a5f : SUCCESS in 5m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e4508d7af02c4ff1a4473cd1ad7ba8b2 : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c36877060d28489ababffcec6db2b638 : SUCCESS in 8m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c5c709423c63462ba4b0c01c11f469b2 : SUCCESS in 15m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/76d31fc092314e66867e125f211fa9dd : SUCCESS in 17m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e743de2ce0834a3d88831493cb7cd731 : SUCCESS in 1h 09m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1b8f005320843aa921637b26c70e575 : SUCCESS in 59m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a1bfabd44344e3195f1faab19e6fc37 : SUCCESS in 1h 12m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ed43d873dc0f47e3ba1452fdad51d349 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":2},{"id":"24b260aabf5982cff6b83d439b6887942d094a22","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 03:08:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36339fd72bd9acd0ce93c34eae5c5cc848065a5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:37:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5bf0f6996744e2fad185bc11e765b51 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0591461c276748eba379d6e0489cc5a9 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdd48584b5294b718ebf1832078337c6 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78f81865d62b41988471e15fafc7f2c9 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fdb5daf3383430fad56402e1b9b54d4 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d6c2d9d11104a0bab477d222c19141a : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c295d80eb2db4d579845e845af422c1d : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0e8c841755a487ea9dae05689079273 : SUCCESS in 8m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/49f77c24869543e8b28a8c9f40fa47fb : SUCCESS in 8m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/47834d534d2a4c9cb3abf89503bf1310 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/03f2c7f7652b44d49137ac1919cd2008 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c57e938def84460801cbf3f9cbbb555 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c078ab3ab72847e29a1c4ff4dbbaddbc : SUCCESS in 13m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5eb02f4fbe74485a3d5d7e219c0ba2c : SUCCESS in 25m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/072236cc534846b486a5b560120a86ed : SUCCESS in 11m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/87dd5ffbc20e453191d16a108aa3cb37 : SUCCESS in 13m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d140333313bb47cba253f9ccc36561f7 : SUCCESS in 16m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5707a55bf87947f8a5b8639272c8b49a : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/71eb6f8e630b47e4be2a87cbfd5c537f : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4518734f319a444d98481c0d684bf4cd : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fd7cfacad5264b65b55394074e1bde50 : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75a3c7995b1c47d39718e2e798600cec : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/038980b1a26a43a0817cc40f8a06e048 : SUCCESS in 55m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/713989e73b3744068d9e84dc2b953663 : SUCCESS in 1h 05m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/beb0812f156d4ffe8ba6c3f770de32fb : SUCCESS in 1h 17m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/64d46870e2fa4a80a2d1b07f06f97c60 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1a0452104a63bf25ae890c2ab86fdc87b3a3fcd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:37:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60531445fb4fbcffb795c2000a94b7e9b91e6662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 06:21:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"affbe9bca392284ca371918e2f37e58ae197d840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 06:21:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c20dcccdcd45ce96998b14b35bd8a8 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2750c3decc6044a7911941258d8bfc68 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e73e6b2a8e4c93ba15f4ff6cca098c : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c1ef74550df459192bb002bb64ba2ea : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b484c5eb87d94945a72896c415a42aa1 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db8e50116be74191948806fd59f18234 : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cac3ea87ede04f84804612e055c87357 : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3c69989371c4457bb6b4ef7074bcf6f1 : SUCCESS in 11m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c8579158ece24940956ca14a343420eb : SUCCESS in 6m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eaac1464e60e47d482aa157c3c257f5f : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c067cbd80e2849289125d301e7255f7f : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2d77bdecfb454b3fb774d51184daefac : SUCCESS in 18m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf8259eaccca48d398ddd773b81aacbf : SUCCESS in 31m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0bd4028af0f49b5986d651304125626 : SUCCESS in 15m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2af7920dfeb2440da44afc32811550b2 : SUCCESS in 19m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eee6a27d6ced4408a98da90d481a0a89 : SUCCESS in 16m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/005361e4c5764dceb898da8c675b39a2 : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/16e2fef5e0d34b379594afa6c06bdc42 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22d7be7e0d684330b65aba9bca33bb4f : SUCCESS in 8m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/56aec38bf3e547eb9a49a5a18e88df2c : SUCCESS in 16m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f34333889bba43cc9c14de8101023e8f : SUCCESS in 9m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26bb8191f9224f6c8295ebd5bf19e950 : SUCCESS in 1h 04m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71026774f4a1420b83a33000b29f35ca : SUCCESS in 1h 02m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98b26a4448774d20a87a7a1b3d64daf9 : SUCCESS in 1h 34m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/247073d89d684c4a99806ff1ec4ab02f : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0add04804f7ea3a99c03fdae2146f5799e36f6a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 06:23:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b687f8379da7424a8ebc87d423a061ef : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51f849467c3225a8e30b68345da40ff931178a56","revisions":{"cb015f6a2f3b0e8ca3a8c2667e5bc53168d33c93":{"kind":"REWORK","_number":1,"created":"2020-02-14 23:03:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/707922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/707922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/707922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/707922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/707922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/707922/1"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 23:03:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 23:03:44.000000000","tz":0},"subject":"update constraint for tricircleclient to new release 1.0.0","message":"update constraint for tricircleclient to new release 1.0.0\n\nChange-Id: Iba7794a7a44be27fe84b0218f66c0dcf4a131e89\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I9bda1e59885d0bc0b76a9d2a22aae1b664ab97cc\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb015f6a2f3b0e8ca3a8c2667e5bc53168d33c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb015f6a2f3b0e8ca3a8c2667e5bc53168d33c93"}]},"branch":"refs/heads/master"},"51f849467c3225a8e30b68345da40ff931178a56":{"kind":"REWORK","_number":2,"created":"2020-02-16 01:47:30.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/22/707922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/707922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/707922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/707922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/707922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/707922/2"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 23:03:44.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-16 01:44:09.000000000","tz":-360},"subject":"update constraint for tricircleclient to new release 1.0.0","message":"update constraint for tricircleclient to new release 1.0.0\n\nChange-Id: Iba7794a7a44be27fe84b0218f66c0dcf4a131e89\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I9bda1e59885d0bc0b76a9d2a22aae1b664ab97cc\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51f849467c3225a8e30b68345da40ff931178a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51f849467c3225a8e30b68345da40ff931178a56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
