)]}'
{"id":"openstack%2Frequirements~707923","triplet_id":"openstack%2Frequirements~master~I9f995ba85257342f55f01a11c3bfc38766da1eb2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9f995ba85257342f55f01a11c3bfc38766da1eb2","subject":"update constraint for python-masakariclient to new release 5.6.0","status":"MERGED","created":"2020-02-14 23:08:00.000000000","updated":"2020-02-16 04:24:30.000000000","submitted":"2020-02-16 04:23:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707923-1581826988965-b9790512","meta_rev_id":"a30825f11daf56e63d270f9b43dbd9dbf107637f","_number":707923,"virtual_id_number":707923,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 04:23:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-16 03:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-16 03:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-16 03:00:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 04:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d6f6424809c993521073d67e2a099acb09d1c4c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-14 23:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d28ac1d91c6758486bc0588656f49a92bf864cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 00:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0298abe8ee694b8593cc42b422db61e7 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddd479846074123877560efc7c5e45d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec90cab3e5d643c99babed87603928f9 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4459792d7fd246f9bedd64de5b04d2ad : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33e3bae3919345d793b5f20d1f09f529 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac22fb571ea44bd39c5e8e4a6f2ead85 : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/796bbdd906a14a7fa8d4ec4c9cc5458e : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/241b8e94831f4b18b13787000b009a63 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/be1d6164aa3f49efb25b78239551f09e : SUCCESS in 10m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5ce4c297a0e40f9b1460a2008d03165 : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/659b97e0b2704a36b72536e9e5b5c37e : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50260d41b9c947b0a00dbaf2e25456ce : SUCCESS in 13m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3cd36231553749558c0ea821bc7a391d : SUCCESS in 30m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9d791aa561894f62bc58a93731c85532 : SUCCESS in 37m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db062010fa5a41039d3352766d158149 : SUCCESS in 24m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e99088eef1aa41ddb0dab864a351b8a4 : SUCCESS in 16m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76f1fe09fa2141fc82e2fcfb6356d51f : SUCCESS in 27m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/986525b6d00c44efa9fdb15b515a924c : SUCCESS in 6m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6e6a2b4aad67497da7f6eb6fe09d3d33 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1631b114d224547a4c9a911d3d56449 : SUCCESS in 8m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9e5791dc0a774b2c880b3fedff0151f8 : SUCCESS in 19m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/41e254601fe240f6958aa7ae090c86a5 : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/684a333316264d31ab89ea80e4c77b9e : SUCCESS in 1h 05m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b0a31d568844d19a268a3c1915649d1 : FAILURE in 1h 00m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a00ba1c7a484bdd964b4d507ca338f3 : FAILURE in 16m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/661c6127305a48258720248e1593d79e : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b52c67de5f06e0a1e4732a0cc5581c745f848567","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 01:48:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6768218adc5ba6e1556f9ac7e9877f0dbd842131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 03:00:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516d80a8938142e798908466d40d2cbe : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b421d404840413da0a6594a483ca123 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37cc022b30974fd8a0b5c83365b8f904 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c47707e53ff3409387d50fd8c161ccef : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ceae11c51ae41d2b85185b1968d6e9d : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49231dcb510846f1ac068462eaa6963e : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d056604cd4c944ccb9f5bbc16be7dcbe : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cb376266977e4317a02a812a427e62a2 : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3be5b75a478e41e1951025bea3f01dd4 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5bd6ea0463624779a8e5bfe4f3f41bf0 : SUCCESS in 7m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9cb7e5b72e014b5c9347a6b6b6d20e0d : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17b00b66e5f54be996ef587e6c10de21 : SUCCESS in 8m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/571fbc4958204c3d9c7574eee25c2d82 : SUCCESS in 18m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8456024c069e49448b81a17f94023f03 : SUCCESS in 22m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/03d0a1ddaff74d2798e010d344fd29fe : SUCCESS in 13m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a065482e977943ba92305febf8c8c163 : SUCCESS in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2763028997454c2a9a55a5780a5fccac : SUCCESS in 16m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a9219b75786342dea43092abe632df9b : SUCCESS in 6m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cf5e366701f149b6b05b89d8d5cb53c2 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d870f9360b044ef3ba6fd92856df69a4 : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1222a5b930a043b18e5303f7d7b6586a : SUCCESS in 11m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d6f0b9242c5c4e0cbb9d34b97bda8592 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d74f00cc80c4fb88430e5f66937999b : SUCCESS in 52m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fd713ea61bc4e1da901087f482729bb : SUCCESS in 1h 08m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce29e37b94c249848e08ce240fa9f204 : SUCCESS in 1h 08m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/851764bcbe4141d089386360be701c23 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":2},{"id":"9b62b2219e790e538bd31f231579036e264712a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 03:00:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b83883c5bf0f155e8ae5243cb336b39407e2324c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 03:01:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69537d7a7078dd15e38bda2c9445ffc885945d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:23:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741d3819211d43ef960c5e0908286c78 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1a4957d04f4fc5a1ebc390741ceb0c : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1130a9c07b34702818c02b5b3938289 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9aaacabe9cfb4c569dec94fd644a4234 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cd0ead97da3444990f840be61be1234 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67f08743ccc14c4cbc02ec6c5273ec38 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15b06aff542a4f4fb4aacef49e3c60e3 : SUCCESS in 7m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d50b90424b764bceba3e4a82f8b6b10d : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb9f7fedbe5b4bdd99ef2f4b97db9133 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5ca66ce9d528485587bcb52294f3aed0 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74f263bb4d9e4636ae29947279529578 : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be8ae498cae741daa1a630cee995b308 : SUCCESS in 17m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f86cee6deab492fa86bde47bc29264e : SUCCESS in 24m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d6b8e15acf734eecbd0eae6ca9a21586 : SUCCESS in 13m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b7f6aa56b3a74b4789df66e52fd9c6fe : SUCCESS in 14m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea1f67863cdd491bb78e8a548a43987f : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0042ec5456794fecb4c5dd17effeeec9 : SUCCESS in 5m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fba3edfeaf9143feb7fffb3fc79e8cb3 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33b5b5f8adfe41aca2f4f463cf1201fc : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1fe7c0adb5764d89bccc523954756868 : SUCCESS in 11m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/231bf1d57a854800a8623f0e790ba8f2 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3810d1514a6a42919faaf36505019703 : SUCCESS in 52m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5457deb6122c4a0fa4e06961df0ff38b : SUCCESS in 1h 02m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a68638b4e1644359aee6060ca1ac447 : SUCCESS in 1h 17m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4ee3ecc163864aa8bb5d8630688b45b8 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8f09f6deeb1bb541f10edc80162a69069e9be868","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:23:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"00a74e82a8549c128c1c129c78005b689a254f04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:24:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c08b00d7e474a6cb1683f14d4556d7e : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9258a3a87892e6806a064afb231ef87bea94ad2","revisions":{"0df3f4f131c4bfd02ef545b5d64058439bde065f":{"kind":"REWORK","_number":1,"created":"2020-02-14 23:08:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/707923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/707923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/707923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/707923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/707923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/707923/1"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 23:07:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 23:07:59.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 5.6.0","message":"update constraint for python-masakariclient to new release 5.6.0\n\nChange-Id: I9f995ba85257342f55f01a11c3bfc38766da1eb2\nmeta:version: 5.6.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Idda492f80f62dd180c809d1256b90e3dda78c5f0\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0df3f4f131c4bfd02ef545b5d64058439bde065f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0df3f4f131c4bfd02ef545b5d64058439bde065f"}]},"branch":"refs/heads/master"},"d9258a3a87892e6806a064afb231ef87bea94ad2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-16 01:48:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/23/707923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/707923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/707923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/707923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/707923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/707923/2"}}},"commit":{"parents":[{"commit":"4979f7d33e1249182f8aab80906ed56cef542c62","subject":"Merge \"update constraint for oslo.privsep to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4979f7d33e1249182f8aab80906ed56cef542c62"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-14 23:07:59.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-16 01:47:43.000000000","tz":-360},"subject":"update constraint for python-masakariclient to new release 5.6.0","message":"update constraint for python-masakariclient to new release 5.6.0\n\nChange-Id: I9f995ba85257342f55f01a11c3bfc38766da1eb2\nmeta:version: 5.6.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Idda492f80f62dd180c809d1256b90e3dda78c5f0\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9258a3a87892e6806a064afb231ef87bea94ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9258a3a87892e6806a064afb231ef87bea94ad2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
