)]}'
{"id":"openstack%2Frequirements~707946","triplet_id":"openstack%2Frequirements~master~I07b4fdf5beedc3480afb9ca0a41252c11084dc12","project":"openstack/requirements","branch":"master","topic":"tempest-py3","hashtags":[],"change_id":"I07b4fdf5beedc3480afb9ca0a41252c11084dc12","subject":"switch tempest job to py3 version","status":"MERGED","created":"2020-02-15 18:53:58.000000000","updated":"2020-02-15 21:47:29.000000000","submitted":"2020-02-15 21:45:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707946-1581803137281-b9790512","meta_rev_id":"bb039d32db6a9342e2e3f6ee5ae97cf5242e4c43","_number":707946,"virtual_id_number":707946,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-15 21:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-15 19:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-15 19:49:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-15 19:49:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-15 21:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1b50b83398ac50c4844d610dc0c902f02880c93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-15 18:53:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c82c46c366ff9bedc009fa9cf62b7c39603b312","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-15 19:47:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af35d00b085ea6d887c7183aee3686caf105db26","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-15 19:49:46.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80a61b0071692acef925cea1991beee8616f3881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 20:23:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc14f7344e2402f9eab838f10fa363d : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e195fe28d64c03a96d5d228a0f4dc5 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c82eb7493ed74233901f58ffd63aa8bb : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a768c86c446a4d94b41d433b4ea040f7 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e444a250aa94ac686f55f6341d3cde8 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41ffde4675004af0b3e41b729e820920 : SUCCESS in 4m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71d23bac4fa14ca9ad61d3287a17248c : SUCCESS in 56m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfeb231b460e4851baa6fbbfebb424f9 : SUCCESS in 1h 06m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/607cbd0e5b4e4335bcd2ae88d954a0fd : SUCCESS in 1h 28m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eae1ad6d85ed4a609da1c00acc9f6129 : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1df422599a69eebbbc4e23b05b7f39a6c4831774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 20:23:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a92c41961ef0e1f3897ed40f9a80ff283872a108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 21:45:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9968a7c891834256b48e84053e707a41 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fd9d1795e64cdd88b587ff1527846b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd62f54acd7485c843872b7f1d767b8 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ed9f33822b149d694f02cb38cf5ad9e : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2101843542341349fbc04882bc2107d : SUCCESS in 4m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3e3c1c69f61479fb37718ae97852d49 : SUCCESS in 56m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa9797aae885406996ca958cf2e89cf2 : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a57d867ad1ee4d7f9266f794ffc231bc : SUCCESS in 1h 16m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9c19de2037264e0b88af2353ef51f486 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":1},{"id":"13d2ff7690c1e197273efb869d37504d125dfa1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 21:45:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4bfe44df91a3760873244727e50137e50188e130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 21:47:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2180cfd83da470187e8cc5dfb652918 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68c7102d639d7ac7f78d920e56b6e8d9945248a5","revisions":{"68c7102d639d7ac7f78d920e56b6e8d9945248a5":{"kind":"REWORK","_number":1,"created":"2020-02-15 18:53:58.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/46/707946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/707946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/707946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/707946/1"}}},"commit":{"parents":[{"commit":"0e2a34e79279974a0d69ecf096282c8bf38872f5","subject":"Merge \"update constraint for oslo.upgradecheck to new release 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e2a34e79279974a0d69ecf096282c8bf38872f5"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-15 18:52:01.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-15 18:52:01.000000000","tz":-360},"subject":"switch tempest job to py3 version","message":"switch tempest job to py3 version\n\nold version installed neutron from master (with constraints, but\nneutron is unconstrained in master because plugins need it).  This\nwas fine when neutron/master supported py2, but it doesn\u0027t anymore.\n\nChange-Id: I07b4fdf5beedc3480afb9ca0a41252c11084dc12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68c7102d639d7ac7f78d920e56b6e8d9945248a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68c7102d639d7ac7f78d920e56b6e8d9945248a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
