)]}'
{"id":"openstack%2Frequirements~708266","triplet_id":"openstack%2Frequirements~master~I519694871b4133f8f7e227b9838fc0b912adfd67","project":"openstack/requirements","branch":"master","topic":"blacklist-oslo-policy-3.0.0","hashtags":[],"change_id":"I519694871b4133f8f7e227b9838fc0b912adfd67","subject":"blacklist oslo.policy-3.0.0","status":"MERGED","created":"2020-02-17 22:15:28.000000000","updated":"2020-02-18 08:50:15.000000000","submitted":"2020-02-18 08:48:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708266-1582015702349-824f8b14","meta_rev_id":"5d9c4c646291dddfad764de7c8bcefcc7885360e","_number":708266,"virtual_id_number":708266,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-18 05:30:00.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 08:48:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-17 22:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-17 22:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 22:42:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-18 08:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dee568648bb683cc86ee7b832edcacec66efc913","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-17 22:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"581c4187846674d16ca861449cfd26306affe06d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-17 22:42:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood adding the py2 cap in there.","accounts_in_message":[],"_revision_number":1},{"id":"c021810545279f15ff0bf4b5593c96fb876ba365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 00:59:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb829bed5d14cc68ee68bef61b540d9 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d81cc5c7b154b04a4900e981201a55e : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/274298ce9fe044b9a0e0455ed9db2e2e : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff8670cd081a45a4a1ab3d0250883c70 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b63e48bed5cf4f8ca84e6169fd1884cd : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af44e5e2b2c94e2abf45794d283cfdf8 : SUCCESS in 4m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83eae8c2c22342e0ab8dbdb4c60adf2f : SUCCESS in 1h 03m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/694c33da3e9a4886b1b43f33f6e638dd : SUCCESS in 1h 15m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/586cc10e2a264f0580fdfb1fff81356b : SUCCESS in 1h 16m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b3a103f83fce419cb6d9847be9d4d552 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"79bf4d1467a94a60c0e99175d410b2013abdbb57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 01:00:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f027006092eb5c021643ecf0737bc381e4195e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 05:24:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b8880f0588407580e6db167e8c4f1a : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e59bbab784e4705855b8b8270739d23 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/640df2b71e12470a8301d96177c4ca18 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a6b0941278a49f4968c83f3800cadc5 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4331509991a46c983b82fb6298b8bdd : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a3a23817ba44b62b6ff47a8bb013670 : SUCCESS in 1h 04m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/83028b9e80c3498dbcedbb5aa8c643db : FAILURE in 1h 12m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a0f2dfaebcf48ddb7d5fb0b27955e66 : SUCCESS in 1h 44m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be6c30cacc604ddc8b23918fa94ee317 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"fa44eca6c5c8a41753453aa49a517b7949fa3179","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-18 05:30:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2f6cb16f4d858bf30b65d025643d41404d5ad36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:53:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a033964bcc496298e4d8144c836389 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829eb4e2f807499280ef41b258fbeae5 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace410adcd0b4971b7b6099463e6350a : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c30bd62904c54003b8d167ff6ac77583 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aaaa946f77d4482960b563b6e936216 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd106f0c6c3f4cb6b8684492e288abe3 : SUCCESS in 4m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f7142edb94648b7b0463a8e79983aac : SUCCESS in 55m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23acf93b52dc42f6a048753a4d0b953b : SUCCESS in 1h 10m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54fcc5b9ea1e48ecb9c1377998b7a3d0 : SUCCESS in 1h 19m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e879fe93ac314f3ba28dc8bf47909339 : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f961691b534d838069c0388c62e37d8f95f8e8f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:54:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d822cf9f5e793062a30c124520190309678220ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:48:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7d076608d1b4557d23a0e801d890ae65f30a7f41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:48:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a09aae349d4421bdcd48f520a31a83 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba27cee739447ba8f3a96996404f813 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c05665ffd4e4503ae886679757fe02e : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94c3137ab9fc4b7f9a433f71cd3bcdb9 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/726558eb5ca94be7b313fd8195213d11 : SUCCESS in 6m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c06113924c9841b4b651e3e953539de2 : SUCCESS in 1h 06m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d37a6843693e4c63907329d42565f266 : SUCCESS in 1h 08m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99fb34dd19a24aa09d57a6b6ddaec444 : SUCCESS in 1h 43m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/248ce3396a8d4ef88e393d144627f820 : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2904ddfea480b974fec1e0fb41008675c0c0ba7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:50:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1a2245783c401d874c301fd1752b31 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8c2205742c9dbec12fbda2f2dcad8d9db636c57","revisions":{"b8c2205742c9dbec12fbda2f2dcad8d9db636c57":{"kind":"REWORK","_number":1,"created":"2020-02-17 22:15:28.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/66/708266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/708266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/708266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/708266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/708266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/708266/1"}}},"commit":{"parents":[{"commit":"8481a77811de6cf23ae69f3eea8b93a24b3be4e7","subject":"Fix oslo.privsep global-requirements with py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8481a77811de6cf23ae69f3eea8b93a24b3be4e7"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-17 22:13:02.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-17 22:13:02.000000000","tz":-360},"subject":"blacklist oslo.policy-3.0.0","message":"blacklist oslo.policy-3.0.0\n\nhttps://bugs.launchpad.net/oslo.policy/+bug/1863637\n\nChange-Id: I519694871b4133f8f7e227b9838fc0b912adfd67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8c2205742c9dbec12fbda2f2dcad8d9db636c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8c2205742c9dbec12fbda2f2dcad8d9db636c57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
