)]}'
{"id":"openstack%2Frequirements~708327","triplet_id":"openstack%2Frequirements~master~I57f958ca93a2dd424ce988cd28816550bbd6e2d9","project":"openstack/requirements","branch":"master","topic":"bug/1745823","hashtags":[],"change_id":"I57f958ca93a2dd424ce988cd28816550bbd6e2d9","subject":"Add pywinrm to global-requirements.txt","status":"MERGED","created":"2020-02-18 10:09:02.000000000","updated":"2020-02-19 23:12:39.000000000","submitted":"2020-02-19 23:11:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708327-1582153869828-5f1f21a0","meta_rev_id":"6f9ae71d51792db743f538b91f895bc82062b1b1","_number":708327,"virtual_id_number":708327,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 23:11:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-18 19:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-19 20:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-19 20:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 19:34:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-19 20:32:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-19 23:11:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3756e0a81360508bf4e533d9b15939e540c5256","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2020-02-18 10:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61b9fd64b532930e0133cada8e8553e1eccc9836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 12:00:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc6b49a7f0543e88a61c4c832eba016 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4b49b218f741568dd7f806affa48a2 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aedd5c18d8354cb581b368657cfe6120 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdb21952d9624313b180b5cba1673412 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bcbe71edcb1408ca4a53ea89a08933a : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4eda120bffc04bb0b656a1929bd8d4c3 : FAILURE in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0efd8ce547ca4de1b3bc8306f91a0083 : SUCCESS in 1h 05m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/650fca91093c4aa8901143adf550d9ac : SUCCESS in 1h 20m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c5a891f593a46c8b3c778cc50b89544 : SUCCESS in 1h 46m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9e389153ac894a00b3dd08c27c9f2b42 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a0239d58815fbe21dfd83c6a29f69694172e20e1","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2020-02-18 12:09:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a58cc7e7567f0652e82d4fb8448de4718e5b4a0a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 13:07:08.000000000","message":"Patch Set 2:\n\nHey Lucien. This looks fine, but can you add the Q\u0026A from here to the commit message:\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements","accounts_in_message":[],"_revision_number":2},{"id":"205d46a848e166272999c8228b4da249a246d12b","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2020-02-18 13:13:05.000000000","message":"Patch Set 2:\n\n\u003e Hey Lucien. This looks fine, but can you add the Q\u0026A from here to\n \u003e the commit message:\n \u003e \n \u003e https://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements\n\nRight away.","accounts_in_message":[],"_revision_number":2},{"id":"3043f0b5b4bf6022e1d8c4e001c3c9b33242f78a","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2020-02-18 13:34:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87b70b6327a841d783f4ad7333bfd5d388faa4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:15:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54a14a0aca7a4fd3b83cce55e1e4bedb : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091010579db047439ce4ed3cf462e1aa : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea4e778704741849e3860dc726f020d : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c65313182f74cf68ccffaf60e99fc62 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e5aa1832b684d4697a7e160ea77589b : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfc7ddf502b0419f8ca4ec6f67c75d42 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d13269f0be14c8ca0665e7c19d5a1db : SUCCESS in 5m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d3dfb8b91c7b4435a7e107b0f320d6e8 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e9fd75717f746a896497108b9e7cbff : SUCCESS in 12m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3b0430d417e4523ad487598c3b2977a : SUCCESS in 8m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1eb834686834457b84d764f111de6711 : SUCCESS in 14m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75daf5e9f1c54b0b8c590271ebe476b2 : SUCCESS in 11m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5323e85efbf840bfb96a35342167e746 : SUCCESS in 20m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/669cc39d6f6c42b9b664e5d98a7df1ef : SUCCESS in 25m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3f22908e89d4719baa2d077652175e4 : SUCCESS in 15m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8cf14426a5ce4104a864b601dc549fed : SUCCESS in 16m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f900d3ea8624a2695bd8e7a938ff445 : SUCCESS in 16m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e306c4a5022840aaa5d19cb668654d3e : SUCCESS in 8m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/540a896dd48744edacc7132a0cdece7b : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5548ab64ba64f22beb570513c73c174 : SUCCESS in 9m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5ebe814d0123456f9891b22819908610 : SUCCESS in 14m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de32b38e9ec0449bafca8238670ca2a7 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef692ab202d542478536300ffc7f5f03 : SUCCESS in 59m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff8589275cb847dd961d76886787f4dd : SUCCESS in 1h 14m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c1cd233940b4c56954b8c4922cdf886 : SUCCESS in 1h 21m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/00820a5d17aa46d2a7971c2fd0204429 : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c53ebdb4f7c25a1dbe57d47206119018f3802ba9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 19:34:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a20a6e08ff303892fe6d2d96483bdaa8526ad95","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-19 20:32:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0ff3d01a8c24499e5676f73482bde9de698a455a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 20:33:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"026d0992cdc7e73967b61f3bd8c4034f9603c31f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:11:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4aa0e6ef5af4437a64aaa3fb9eca0f6 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c45eaaea9b940429dc37d1c879ab2ed : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/271cd284835948498579690b499c88ef : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63085c554df04de38ada763e28b19f9d : SUCCESS in 8m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9a4a6f1a8fa4c2b8fa065646375dcf8 : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/493c8e1506bc4fa59d1d19fe73a082e6 : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73e5e9bade7d4cb0b32eb2bfb4594c6e : SUCCESS in 11m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5348d793811e45d9a111c3181989e612 : SUCCESS in 12m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/66e0efba696842db88e2246b5bce8e5a : SUCCESS in 7m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/13c1843a353b4d048013ad0b8e933bd5 : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44e71e3d29df485fbbc804c7e05eece8 : SUCCESS in 11m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fedc41b9eaf24ae7a0f96934ed6c2f51 : SUCCESS in 18m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0c141a2071d24e0d9c5d52c235c27695 : SUCCESS in 24m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68523fefb77b4a7fa5cb2ac4a08415f8 : SUCCESS in 16m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5f61dc571e9944298d1c1c3c4ccd3767 : SUCCESS in 16m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2930b3399a54366a7ce61ade602645e : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/456c033e65f7442a876e3d765d932483 : SUCCESS in 6m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/963fe2b45a674b43b9ff8990ce2811cf : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/779e155efeb0401da7c0d5126230a8a6 : SUCCESS in 9m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bb287797545a4c0394e74eab9657e0f2 : SUCCESS in 12m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9c1496a7c09b4e49a14b231263a05d56 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24ea154a404c4675a4186d6c1410035b : SUCCESS in 58m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4c4956a86ef481199249786ec582b97 : SUCCESS in 1h 15m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6df7f7718554e009c0f43b2d6a04304 : SUCCESS in 1h 23m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d3542a5d278b45209de2bb00686ffdde : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c2e68c5fa0e4dbe9d29bcdf327ee557a74223e2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:11:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e862e1619cfba7e2b095f56ca7e1be341b41ec7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:12:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3cff808278e4263860f550b3d876d8c : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2db459b19d6ce31447a34fe95f94d02ae6cc8fb3","revisions":{"6875bb8d4179351ba458f735e0033d03360c229d":{"kind":"REWORK","_number":1,"created":"2020-02-18 10:09:02.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/27/708327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/708327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/708327/1"}}},"commit":{"parents":[{"commit":"dc7719e8b446eb1812833ab914ef6a78fc88bc1a","subject":"Merge \"blacklist oslo.policy-3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7719e8b446eb1812833ab914ef6a78fc88bc1a"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-02-18 09:59:37.000000000","tz":120},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-02-18 10:08:44.000000000","tz":120},"subject":"Add pywinrm to global-requirements.txt","message":"Add pywinrm to global-requirements.txt\n\nThe \"oswin-tempest-plugin\" is using the pywinrm library, yet it\u0027s\nnot included in the global-requirements.txt file.\n\nAlthough not set as as a hard dependency, the Manila Windows driver\n(similar to the generic Linux driver but using Windows instances)\nis also using pywinrm.\n\npywinrm allows sending commands to remote Windows hosts via WinRM,\npretty much resembling SSH.\n\nChange-Id: I57f958ca93a2dd424ce988cd28816550bbd6e2d9\nCloses-Bug: #1745823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6875bb8d4179351ba458f735e0033d03360c229d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6875bb8d4179351ba458f735e0033d03360c229d"}]},"branch":"refs/heads/master"},"97f03ff6d35e69598ad83099875e0845af7ed379":{"kind":"REWORK","_number":2,"created":"2020-02-18 12:09:24.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/27/708327/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/708327/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/708327/2"}}},"commit":{"parents":[{"commit":"dc7719e8b446eb1812833ab914ef6a78fc88bc1a","subject":"Merge \"blacklist oslo.policy-3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7719e8b446eb1812833ab914ef6a78fc88bc1a"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-02-18 09:59:37.000000000","tz":120},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-02-18 12:09:06.000000000","tz":120},"subject":"Add pywinrm to global-requirements.txt","message":"Add pywinrm to global-requirements.txt\n\nThe \"oswin-tempest-plugin\" is using the pywinrm library, yet it\u0027s\nnot included in the global-requirements.txt file.\n\nAlthough not set as as a hard dependency, the Manila Windows driver\n(similar to the generic Linux driver but using Windows instances)\nis also using pywinrm.\n\npywinrm allows sending commands to remote Windows hosts via WinRM,\npretty much resembling SSH.\n\nChange-Id: I57f958ca93a2dd424ce988cd28816550bbd6e2d9\nCloses-Bug: #1745823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97f03ff6d35e69598ad83099875e0845af7ed379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97f03ff6d35e69598ad83099875e0845af7ed379"}]},"branch":"refs/heads/master"},"2db459b19d6ce31447a34fe95f94d02ae6cc8fb3":{"kind":"REWORK","_number":3,"created":"2020-02-18 13:34:30.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/27/708327/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/708327/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/708327/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/708327/3"}}},"commit":{"parents":[{"commit":"dc7719e8b446eb1812833ab914ef6a78fc88bc1a","subject":"Merge \"blacklist oslo.policy-3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7719e8b446eb1812833ab914ef6a78fc88bc1a"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-02-18 09:59:37.000000000","tz":120},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2020-02-18 13:33:58.000000000","tz":120},"subject":"Add pywinrm to global-requirements.txt","message":"Add pywinrm to global-requirements.txt\n\npywinrm allows sending commands to remote Windows hosts via WinRM,\npretty much resembling SSH.\n\n- Is the library actively maintained?\n\nYes, there are a few quite recent releases:\nhttps://github.com/diyan/pywinrm/releases\n\n- Is the library good code?\n\nYes. Despite being pre-1.0, it provides a stable and well documented API. The\ncode is easy to understand and well tested. It\u0027s a simple library that\nimplements the WinRM SOAP protocol.\n\n- Is the library python 3 compatible?\n\nYes.\n\n- Is the library license compatible?\n\nYes, it\u0027s MIT licensed.\n\n- Is the library already packaged in the distros we target (Ubuntu latest /\nFedora latest)?\n\nYes, it\u0027s available on PyPI as well as Ubuntu upstream repositories.\n\n- Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nNo.\n\n- Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nYes, it\u0027s already used by the Manila Windows driver (similar to the Linux\ngeneric one but using Windows instances) as well as the oswin-tempest-plugin\nproject.\n\n- If the library release is managed by the Openstack release process does\nit use the cycle-with-intermediary release type?\n\nNo, it\u0027s not an openstack project.\n\n- Do I need to update anything else?\n\nNo.\n\nChange-Id: I57f958ca93a2dd424ce988cd28816550bbd6e2d9\nCloses-Bug: #1745823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2db459b19d6ce31447a34fe95f94d02ae6cc8fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2db459b19d6ce31447a34fe95f94d02ae6cc8fb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
