)]}'
{"id":"openstack%2Frequirements~708355","triplet_id":"openstack%2Frequirements~master~I6dfd45783f3193a3885e50df6c80b293ef90e035","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6dfd45783f3193a3885e50df6c80b293ef90e035","subject":"update constraint for paunch to new release 6.0.1","status":"MERGED","created":"2020-02-18 11:28:59.000000000","updated":"2020-02-18 21:56:44.000000000","submitted":"2020-02-18 21:52:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708355-1582062778406-7c03db43","meta_rev_id":"4e4564308642d4923bd9d42cda079b30cc5802c2","_number":708355,"virtual_id_number":708355,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 21:52:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-18 15:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-18 16:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-18 16:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 15:24:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-18 16:09:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-18 21:52:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc0900e560055dfbfe0cbb25384071deeb2413bd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-18 11:28:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a93f799088b2db5127da466bbf6a6f1475e3db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 14:03:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/975a4fc97a6f446790c55048074dfbc2 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f63172e0ef24ace97d40ed881353868 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac9d16318eee401fa058b07cc614d9dd : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71989ac88aaa40e68bd569ecd459acab : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7602f53a0d2b442792af1bb0389b9015 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c5c016fab7a448ab88afac92448f8c4 : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ce70949a29e4f47820478d37ba74b12 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b129734cf71e44ffbde52f792c259f04 : SUCCESS in 11m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3fb197874275434095e514d765bef36a : SUCCESS in 11m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f0e4e01f69848109fc61ddaa7eacf7e : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3deaa945348f4aafa760e75b42a5638c : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb663f4680a94b3d80f6e5a04d7e1ae0 : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/20680c7b2a774c39b69911c830918b06 : SUCCESS in 21m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a90a83036ab54a328d0cb6d101795a2f : SUCCESS in 29m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d2776ab514564e118b9816a47a981db3 : SUCCESS in 16m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a4fa03d6658347d78556403137822e87 : SUCCESS in 16m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/250b66c4efa041ad863ae8ef2d7dcf59 : SUCCESS in 17m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0338631b6e0a4d2ebc2fc632d18d8426 : SUCCESS in 6m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/161b5487d7dd4309a2f022f50b8077c2 : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/222e8e64bc3342799ff0f1e6f3917fe1 : SUCCESS in 9m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/14e5a17178d841cca4c0e06bd08424ba : SUCCESS in 10m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d74374da256140f58a637615418aa637 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/541ac8ee9e624841b8d87bec9573e6e0 : SUCCESS in 1h 09m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbf1c27d0a49445fb65f20872eb7cfdd : SUCCESS in 1h 07m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d5c02d7569e4acd81c55ed2023e8f2a : SUCCESS in 1h 20m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c511f53d1f3f4372944bf96dc0b6575c : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9d4b28174cf801b29f66389884fab4e0dc0c0533","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 15:24:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7981f2b0f6e68853fa439a49311cca8eadf87e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-18 16:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c29f14489fb16d2aef46acb398cc63a57dd047c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:22:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bab0d4d8a8e1be16f53e9340680c6c1f2cc53e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:52:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86b7584dd1fb31582a0f29b5e1e164d98d91ba9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:52:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4821abee112646af8358698fa3f85598 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bd97846d9046bea8c3d8b778479034 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e623c09609a4469f89b28ea11d8ea4ab : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8c6f6be41f146b0b223404500b64322 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/934c1ccee44d40e399c3b8153961783b : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a83f1d42496e4a33b9123b62ba84a48f : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6364115e7ef0417895e7cf2e0465f027 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5cfbbbdeceb04fa688abed1d7703836e : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2a7738a555f448ceb9cff9c2ec93545b : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f80840a18344d1e863945993dc51aac : SUCCESS in 13m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d86003cacf84619a7879ba725d1e3dd : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ca18a05d56da4db4976e2dca216a1d5b : SUCCESS in 14m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa37b3dc7acc4e2390283d9938a5520f : SUCCESS in 27m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/344343f86fd44f328d6db314d6d29c1d : SUCCESS in 14m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b290f6f6f3094fe3ab280be4966f9457 : SUCCESS in 15m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3300a3859524da0a7e54be6b3eb47a9 : SUCCESS in 20m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e0228dd217a84e3ebde9dcbc1adadb73 : SUCCESS in 7m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/451b7681cefb4287a792b59e967f82d5 : SUCCESS in 8m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63c03005bf28403fb5955222767f58aa : SUCCESS in 8m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fef08832d1d44b1d83adb21fe0d04582 : SUCCESS in 12m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/14b4d4c2d388483d94e6b7f0f8a8d263 : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f48af292d40417fba26f9e762048a85 : SUCCESS in 1h 03m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b17f5e22ec1b40bfaae5dac52250202b : SUCCESS in 1h 09m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cece0bfa15848f2838a1e521bd04275 : SUCCESS in 1h 27m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c8796db10b14d7d9a8a29e10220ac49 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1},{"id":"19a6480804fb09baf0edf56ec3f882f7a19c938a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:56:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353ad61388e14ee5a35984a477b7896c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a81ea7debfc6dc5a61046ecea3e474b40e0a2a9","revisions":{"7a81ea7debfc6dc5a61046ecea3e474b40e0a2a9":{"kind":"REWORK","_number":1,"created":"2020-02-18 11:28:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/708355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/708355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/708355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/708355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/708355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/708355/1"}}},"commit":{"parents":[{"commit":"dc7719e8b446eb1812833ab914ef6a78fc88bc1a","subject":"Merge \"blacklist oslo.policy-3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7719e8b446eb1812833ab914ef6a78fc88bc1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 11:28:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 11:28:58.000000000","tz":0},"subject":"update constraint for paunch to new release 6.0.1","message":"update constraint for paunch to new release 6.0.1\n\nChange-Id: I6dfd45783f3193a3885e50df6c80b293ef90e035\nmeta:version: 6.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: I0a67b51c4df89a15af6a0167cfde2043ecf4d344\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a81ea7debfc6dc5a61046ecea3e474b40e0a2a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a81ea7debfc6dc5a61046ecea3e474b40e0a2a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
