)]}'
{"id":"openstack%2Frequirements~708359","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I546d9bc9ca1ef992d713071e54c939654ef831e9","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I546d9bc9ca1ef992d713071e54c939654ef831e9","subject":"update constraint for ovsdbapp to new release 0.17.2","status":"MERGED","created":"2020-02-18 11:39:18.000000000","updated":"2020-02-18 21:54:44.000000000","submitted":"2020-02-18 21:54:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708359-1582062884379-3390009c","meta_rev_id":"db5c74879bcc4ee754c0c7193f7e3c9c63231010","_number":708359,"virtual_id_number":708359,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 21:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-18 15:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-18 16:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-18 16:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 15:25:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-18 16:09:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-18 21:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30bec60fe97c755979d71b06f76d409fad5d1065","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-18 11:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cb22d41160cc92a3bd5997c1eb737fcabd04584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 14:47:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0d291da1e0db407cb09e7612ef5a665e : FAILURE in 5m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb6b3640b1f470b837b3e2a2d10ad39 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb81abb05fc8488c9130d73a88eb0f25 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b3f462d92b84ad18b8dacffcde10caa : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a84e7e382a148b1abfa366f39e474b9 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9585881b986348f5ae9f3a022ba69433 : SUCCESS in 7m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7785d305ed124c0794fac28c876c5e4b : SUCCESS in 7m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e0621c843d8d47c1bf767adbe8455d7b : SUCCESS in 11m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/776d68ff0e554afcae9de6af5dfaa2a0 : SUCCESS in 11m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/538a28ff132d43109d5d4f1cb03ba443 : SUCCESS in 12m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f735670c988e405eadd48862c884a60b : SUCCESS in 9m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/59706caa0b234db1a962bd01a9a9e95e : SUCCESS in 8m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a899b4fbbc6746e0a0e725b864731bcd : SUCCESS in 6m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/048a2ed7dea54861b02593f246eb8f2a : SUCCESS in 13m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f3a86be8751b48aeb23e4566c4ab65a9 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f9d8c198d604b2d9e87c85a45c98713 : SUCCESS in 9m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2298fb7db4004c55b0e43f3a6c6b2cf0 : SUCCESS in 15m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6684231db4a14eb6b6396bc06746be49 : SUCCESS in 14m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3284c8956f9c44cc960c981d76574eb2 : SUCCESS in 30m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0333a33d848d408baac82db8173bf3da : SUCCESS in 23m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/15a2ee9e546b4fc59326e232dd0157ac : SUCCESS in 14m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/179e140f00aa437d969dcb1c9b70dc4e : SUCCESS in 15m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8c36810c7c924e69ba2ad3b79bd85139 : SUCCESS in 17m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7315550d12fb4de2935b471d6eb20ed4 : SUCCESS in 23m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f754a35109a74e98b74161f6210cad64 : SUCCESS in 9m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/28ba2634235d450fba4c5a3ee6f304cd : SUCCESS in 13m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1d263adfefa4869b8d5aaf79bb54dc3 : SUCCESS in 13m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb0e9075667c4072811187b6d7bee444 : SUCCESS in 1h 15m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/57b517d7d981486682fcfe9bebf0aadb : SUCCESS in 1h 43m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e5cf1cf6ea34d4ab613e2ed9c82cab7 : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"663ccc151de5e1f647858edd50117767bf9e8027","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 15:25:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb65f91148075e1927709e8ebf4a904baef57b38","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-18 16:09:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2815c78bfd4ba6d5c864120c131c2280a805f52d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:22:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d761ba0161bd8bf8333f1a5ea5bb6c32aebd4d90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:54:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28f87b2252e4b81a79c3c3f244536777086293fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:54:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f131f7b5e604949b45d55f268a08b5e : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77ee509bc7634f93ac45fb5e999e3e07 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55638a26e3cc45edbd139cc9ee0c53e4 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0f2e7843ca54f1e8e3cc5ace66d13ee : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1881cb7f77b54c96af16255021df85e7 : SUCCESS in 7m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6fdf4973fea4ae0811ccfa69b46ee0d : SUCCESS in 4m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/44549505d04449d18e535bd47fba8c43 : SUCCESS in 10m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c1fb52cbda6f473c85a1c87f908064eb : SUCCESS in 9m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5f18f4cef55e4e6a8962d9f92261fbfb : SUCCESS in 11m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f0d08c9c01f4465598441b721ddd74c1 : SUCCESS in 12m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8dc5ab80fe5b4d2b8e8e28a88559fbf5 : SUCCESS in 6m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3dc99cad879241d48d82d467971fa7ab : SUCCESS in 8m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d79efae8d72a4c4d8a3ab7daf59bea56 : SUCCESS in 10m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/16ae99d7d5d940478149bccca8a83735 : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec7a0ea6b42c4911b0fd93f9a9ac503d : SUCCESS in 8m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d52eff50920b43049ad669afa59dd4f3 : SUCCESS in 15m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a36f2773b61648aea91ba4192d83f5c9 : SUCCESS in 20m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c513ef3c0e1f42429d87d82ffc538246 : SUCCESS in 26m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d6efe18fce7f431aaa7ce6a49de1e86b : SUCCESS in 23m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2a30dd228ca04a15bd4ea55616a8124b : SUCCESS in 14m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b28fc46982b40ec874344016c9307be : SUCCESS in 14m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/310bbcd7d1474aedb5babfebe193b219 : SUCCESS in 15m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b8c30b184524e06833f9cde34133136 : SUCCESS in 24m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13771c1c0770432aa5400bbd84f04d3f : SUCCESS in 10m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/880cb0199e6b43f6a4dcab622022e95c : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6d7297969d464258bde177a5079b5d5b : SUCCESS in 13m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a0f7005d56a4d13bc26bd93aaa53834 : SUCCESS in 1h 13m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f8ec996961d4ee2890c3fe7038353a3 : SUCCESS in 1h 31m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fba6ffabcbdf4b9a82dd774d58cf9e95 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce4b61b3ffeb249747a5bd8491e5e4b7d8c17fa9","revisions":{"ce4b61b3ffeb249747a5bd8491e5e4b7d8c17fa9":{"kind":"REWORK","_number":1,"created":"2020-02-18 11:39:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/708359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/708359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/708359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/708359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/708359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/708359/1"}}},"commit":{"parents":[{"commit":"2da5c5045118b0e36fb14427872e4b9b37335071","subject":"Merge \"update constraint for metalsmith to new release 0.15.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da5c5045118b0e36fb14427872e4b9b37335071"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 11:39:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 11:39:16.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.17.2","message":"update constraint for ovsdbapp to new release 0.17.2\n\nChange-Id: I546d9bc9ca1ef992d713071e54c939654ef831e9\nmeta:version: 0.17.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: I370cbbeab027c4983cb322cc01fde0bcf3e10472\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce4b61b3ffeb249747a5bd8491e5e4b7d8c17fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce4b61b3ffeb249747a5bd8491e5e4b7d8c17fa9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
