)]}'
{"id":"openstack%2Frequirements~708369","triplet_id":"openstack%2Frequirements~master~I4380aaa7fbad005a1b5081c36afec30d58806580","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I4380aaa7fbad005a1b5081c36afec30d58806580","subject":"Add oslo.limit into global requirements","status":"MERGED","created":"2020-02-18 12:13:22.000000000","updated":"2020-02-28 02:03:12.000000000","submitted":"2020-02-28 02:00:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708369-1582855259241-6731fd5b","meta_rev_id":"096f8ed26314a4ff881b24e904837592ae3457af","_number":708369,"virtual_id_number":708369,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 02:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-26 22:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-26 23:00:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-26 23:00:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 16:54:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-26 22:21:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-26 23:00:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-28 02:00:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bdc814e44d672b541f9db85ace5261f320d5f40","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-02-18 12:13:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32550e0911c42efcaa6feb141f7afcc3d1cb3199","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-02-18 12:15:18.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nNote quite sure I know what I am doing here, I really only want to use oslo.limit 1.0.0 in python 3.x in Nova.","accounts_in_message":[],"_revision_number":1},{"id":"5761f19842bffdcdc4137ee7a2afdca4f52d8db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5886f5120c044731a63ae66fb58b421d : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56cea87682624cd285990b85d41f9f01 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b933f528e08241359eee976371d85ce9 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a0b658a2dbb4d4ab41e3b259dc58b8f : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c36373be696b4516a041c6f28304a2e5 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57ce9596a7154d84af20b865c3807283 : SUCCESS in 7m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34075fd0ce3e40c6a01bb3e3e0799e81 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e94d86f28e8d40f3868f25e98152e2e8 : SUCCESS in 11m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/65348fc8d16f44bfa21d621a0c640751 : SUCCESS in 13m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1b35145466ae4d9f9e23b33b1959631b : SUCCESS in 8m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/690190eba03c43a7aad4c6d0604a2a8d : SUCCESS in 15m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57c014bcc76f46f9ad96db6a0eb50925 : SUCCESS in 8m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e67721f3e81a4d8db8aedd92a09ae3a4 : SUCCESS in 24m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9e8376a0ba004a21a0be186bdf7c29d8 : SUCCESS in 30m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7c76feeb778844b89aa2d351ba3e2a4a : SUCCESS in 15m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d1557741ac2f4b1a8d64f476bb26f2f9 : SUCCESS in 16m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e15ac32a3de24057bdb6f025e3d81c12 : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a1dfecf6bcc34752b9b27fb5cc352e0f : SUCCESS in 7m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ab99e1ec50894194b2d47c8205284aa6 : SUCCESS in 7m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdd42d1b00d345a89ea3f96b5034a7f0 : SUCCESS in 8m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f7fdd340cd2442e2a9c756a058475400 : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ce325cfec7a342c98aee112fc8fb926f : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27314a6235d64594986f25ca13b1c24d : SUCCESS in 58m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d9fe4e2f97b4571b60a3667012c711f : SUCCESS in 1h 08m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/091da0ddcaca4249a460478e1614dc41 : SUCCESS in 1h 49m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9bf8c8378e541adae5166cf363620b1 : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bf0e44619ea66d1e1711be4ab3228f7e1f29cb06","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 16:08:15.000000000","message":"Patch Set 1:\n\nNot sure if we want to do this until the discussion is concluded in the mailing list.","accounts_in_message":[],"_revision_number":1},{"id":"b5d9933061f943ea04e089242c64015bbd09e956","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-02-18 22:27:22.000000000","message":"Patch Set 1:\n\nah, I missed that, I was going to ask a similar question, but just went with it","accounts_in_message":[],"_revision_number":1},{"id":"f0567e83ba7ef14f15c01aaf61ae54b5078f04d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-19 03:02:10.000000000","message":"Patch Set 1: Workflow-1\n\nmarking this -W til we know","accounts_in_message":[],"_revision_number":1},{"id":"21119cd57d63de5131269b918bc464d3e9bac163","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-21 16:54:22.000000000","message":"Patch Set 1:\n\n\u003e Not sure if we want to do this until the discussion is concluded in\n \u003e the mailing list.\n\nYes correct could be worth to wait a bit before moving forward with this topic.\n\nFYI I proposed a specific releasenote [1] to highlight this issue/topic to end users.\n\nSee the Stephen\u0027s comment about nova in my proposed changes [1].\n\n[1] https://review.opendev.org/#/c/708326/","accounts_in_message":[],"_revision_number":1},{"id":"4d8da2848e24a4fded54d37b3fee9245912780c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 20:37:55.000000000","message":"Patch Set 1: Code-Review+2 -Workflow\n\n1.0.0 hasn\u0027t been unpublished, and I don\u0027t think we are going to unpublish things, it\u0027s too bad that 1.0.0 was too early, but meh","accounts_in_message":[],"_revision_number":1},{"id":"2f2af07627461a433aefef080fa5f4f91521946d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 20:42:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c8493091be8b81f1f2756f3d9b38d2ba9f8cdc1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 22:21:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"308a057a1e6e0fb6a9956c111d636f556e4e4ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:58:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e404ef574da4fa381eb1d64e9061696 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8d026506d8545d09af98cfb0748e080 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/275f055c088b4a56b0a9a41099493bc3 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a60959051e26477ba648b9d4dfe59c4e : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a561fca1f344710a67dee39c2bb0b54 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a439d8a9a2f54ddeb94e8bb883534564 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c63dfa0f88cb445d999ad92faef59419 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a2b649467464279b46b7bba2695ca09 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8866e7ad46a049d69071132768266b21 : SUCCESS in 11m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/65716542405c4de2801ed970da0a512d : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4fbcff574ed04777ae446ac6bc1313de : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a6635ba09ea44e99cdd67cf502d1e34 : SUCCESS in 11m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f246a4069694d19a3a3dcd3b288ddab : SUCCESS in 16m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5fdee55dc26d4a39a65e8bf30ac13a61 : SUCCESS in 28m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef84ecd6af7743048a91fc2bbfceeb94 : SUCCESS in 14m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fd8125e57765451793e79f85e9c57687 : SUCCESS in 15m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51db572612cd4e10ab381fea80fc66f8 : SUCCESS in 18m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9310e831e19b4005a2455aa87c32e4c7 : SUCCESS in 7m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0edec60a6cfc446ba3c25b824275763d : SUCCESS in 7m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e95941d929d14323be85dc2f1901c5f9 : SUCCESS in 9m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/126dd528511b4fe081811f4e39509fa9 : SUCCESS in 15m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0cb091ef28b4aeba67182be17492faf : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bfd582e700846ebbee0569a7583a0c3 : SUCCESS in 1h 04m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72a41c3f4baf4cb6b8502197a1ec7eba : SUCCESS in 1h 09m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7df574d66be64f11a4f1aa26c823a4c0 : SUCCESS in 1h 26m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e17410187caf45e1913af4a70d124de3 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3c40b6a2e8f99d37114901bc3f7263654c7dd4a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-26 23:00:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50f0aa6a1ee7b8e3e6f4f92f83776ca2504fa1f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 23:01:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60781c923ea8496e1e1bc2d1ac1a62522668b6cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 05:56:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af179c27b29541858b89e43ddca3d611 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7914f32ec9e4a49847465a87ebd58a9 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ad109c0103c4a78af51a51dd95f5ce7 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21d12a048fd3403485f614fbb389cf4d : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db80133211d742adbeaf17ddf3e60e3b : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2902b2ccbde41c79c059e2bd2da3efb : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87f4051909fa4191b5601cb9d27ec422 : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/763e5357a69246a6a52ab85430000c1a : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cea6fe7d44d743109c931eafd7de1e59 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4adc4186b7514afca2b1daedea446f8b : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f8061400db7243a68c18054f5f3ced8e : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b86e297d6c94d80892156eae29cf289 : SUCCESS in 17m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/13c4c35225f04ecf9d9de6abeb48e1b5 : SUCCESS in 28m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6c3c6dff56644d6cb054d32c9b4c3d66 : SUCCESS in 12m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/335f6cc12b04484b9ec9bb738034586d : SUCCESS in 15m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/318e762c92e840f0b6f5c61f9cdbc6ad : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d8cc278d610e4f939d0c9b3b6c75a265 : SUCCESS in 5m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dfaaa20d31844b9181c248540f8b6160 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa53ca77db7144c3969cfc5b4437cb4a : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/50700dc1ac9a48f7a18a75fe14c1a1bd : SUCCESS in 13m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b529952441444e59910268d9d8e6d37 : SUCCESS in 9m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64b36f5eb61b4dda83e6a95bec1ed4cc : SUCCESS in 57m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc5aac56abba497795bc3e2485f69ea3 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/918cf938441948528ecc8f12b70d53d5 : FAILURE in 1h 19m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/729a0fc170d44bd1b639b3d621aab867 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0a59272730eb5254039528d24863403d6e6c9d56","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 06:03:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"15e1de09bd83de4d6f93a9c70066ffa7d78c81e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:23:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90107405d92f4612aeba931433dde4aa : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45a359ef4484c25b83cbd3fe0284627 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0284f8a3e24e6a901f9664141c8ad9 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37b089d12dcc4c2eb24e890dee4cbfcd : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d552a7d2295342549d3f106a3be51013 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88655e1ee9324b6db6dd1bbaaa4b7dc7 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e85dc947bd464709842c4934ed01f4ce : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8937a93f49cf40c5a141e333af5c4192 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/beb1c9c19ff1423280d24a8ddce52e78 : SUCCESS in 12m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ba585dba1d74610bb550a2f6b3b6fa4 : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f9a2a1c0d34b4cae8ac67e63493e514c : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46ce463ad3be435388fc043f88eef85c : SUCCESS in 9m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42f86e5fcd8147749f38ee4f447dbaf6 : SUCCESS in 24m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b3e60e90f4414dd18cf3df53ba1e74ae : SUCCESS in 26m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/515686f466ba4f24b8fe9e1b9da533b0 : SUCCESS in 15m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aaf2d7f4310045ff8ec087af5ae65392 : SUCCESS in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/879da8cf0f8a4552b6e97ff00b7fe746 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/491615454e1c44a18f5ac36c87e833df : SUCCESS in 6m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cf2bd111654b4f1888a55a5d3cca3e2a : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec07aac10cdb46e2952a684805b7092a : SUCCESS in 9m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5dfb3df15db84c2783b474f8d7629ae3 : SUCCESS in 15m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b8846594b164fd2be861443fb703d43 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fa8f60bedce4c36b745039043a9c487 : SUCCESS in 1h 06m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbd7d879360249b199b7d45223c8c9ca : RETRY_LIMIT in 43m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4edec90be4ae4215be83e6051d440088 : RETRY_LIMIT in 1h 18m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e0ef191ab9141e18c10425927117c56 : RETRY_LIMIT in 44m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6f80736f88a203e348e7baf0eb283cec48a58e5b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 15:30:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d0a781addd49fca5a0fd8c4d92d2c2b91ca5556b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-27 22:04:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c0cea6b8214811478edbbf86ed9059289bc1ea2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 00:37:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b9df85464d42aeb011140f447dc7d9 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae6080018e441b4bcc0a6b43550a482 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79e326fcf86a4d1c957d16ed1f920531 : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d120018af7b4521afb1a341e76c0b73 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd356881fa5407f88fff01d58341106 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b024686a719949ef94057a5625ae319d : SUCCESS in 7m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a86833c7cb24527beaa9548af5c0a16 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db4d616d37b4424ebfc347e624bb7540 : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4028ad135434dfd8f68381f04584987 : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3bcdbd1fa72e4fd08be064ea709b102c : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bb454faf50ff45a1a43c86fd171b40b1 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3b4ce7325474bfc943d2b8cc4d0b20b : SUCCESS in 8m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/31afe0f1deb34258afbd0c675ff50729 : SUCCESS in 24m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/762bf948578d47b7a053cd24c4949361 : SUCCESS in 27m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a0599f19397f4237a3787685fcd1888b : SUCCESS in 16m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7b39f56fb9cb4c28844f2c2f1e962137 : SUCCESS in 15m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75cbfc7c4a9f4c01b046d6dd185c09a8 : SUCCESS in 18m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/baa752fb6f5c47bea11316d2aebeeb9b : SUCCESS in 5m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a9c3c8bbaf74456686472709f3913281 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17b27f01a70342629e272b093c41cd6e : SUCCESS in 9m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d1b5a6f932714ccf8f548442c36d6abf : SUCCESS in 13m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65be315ef3864ba4bdc45e2b7b132d85 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc38c701195645fb9988b932e157afca : SUCCESS in 59m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e3f95c0301849f3ae0ebdee40eafe55 : SUCCESS in 1h 11m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb8b6cfd00d747d390180ecbb13c40c0 : SUCCESS in 1h 38m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b1e7b5ab20f54d619cccb7c1b7ccc2c6 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":2},{"id":"848e76d5e83ab6b675f38397583443e8fe8c8a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 00:37:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8cc0ae4395527e03c1727c24892d46bb5b65499a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 02:00:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c71e962a254a6297f91456dfe4fcbf : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6981cd007e244a6a92c59b37ef1b8b5e : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e26bccf0b664b5ea202feffb815971d : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cde33b66f6e349dc9a9c7565b7185fdb : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e11f36dc275425684ddb04fd671f435 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1dfad5db6c841e18b4760ef1ca6ed3a : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b5a740be6ab14229bb347639de9de071 : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d518a86ac5b74e03b0c91cfb47bb9e40 : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d34790a15584efa9a8a06963dc9603e : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1874303cb7624b6289e4279448cf46fb : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b07ee04a064b42b1b89e4c78dc4d6cdd : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d69918c7c88d40c3bea4f4b06121928c : SUCCESS in 20m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8e2d243135594b779572a37c7ec66192 : SUCCESS in 24m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/091f28e87fe2470dbc9061af8af41575 : SUCCESS in 14m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/66aada2bafdf4a3d8ad7f5fc9ba354dc : SUCCESS in 19m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7610391feee340ce8d8cf9fff4ea5847 : SUCCESS in 20m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fda6dafeb26b482d920b59dc35cb894f : SUCCESS in 6m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/172f122f5b7c4da1ad5b86e0c08915f4 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1e91b2eba9f4fac8bc791af36a0a0ec : SUCCESS in 8m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/869ae56ec8944d14a8953943d0095b94 : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c961db8417d640b2a1bd0e5f6e12e6a6 : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1f42aed736e4d528ed33143bb4c406d : SUCCESS in 52m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4ab1ea08e4b4409a5b8d45448c831b0 : SUCCESS in 1h 12m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a53ba805e354bab83c45f74beaeaa40 : SUCCESS in 1h 14m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b0d4f36c42dd4fb984effde3bdaecf22 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e1d8430dd51f43279a00d0170195e770094c475a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 02:00:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"096f8ed26314a4ff881b24e904837592ae3457af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 02:03:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f466498bbc463ba1962468d8c9c2d2 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5493558e86784455285c9c6ec90af5f753fa069","revisions":{"35a13f4105e06b926f031bf6fbd2199480f75bdc":{"kind":"REWORK","_number":1,"created":"2020-02-18 12:13:22.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/69/708369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/708369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/708369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/708369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/708369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/708369/1"}}},"commit":{"parents":[{"commit":"dc7719e8b446eb1812833ab914ef6a78fc88bc1a","subject":"Merge \"blacklist oslo.policy-3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7719e8b446eb1812833ab914ef6a78fc88bc1a"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-02-18 12:12:50.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-02-18 12:12:50.000000000","tz":0},"subject":"Add oslo.limit into global requirements","message":"Add oslo.limit into global requirements\n\nNova is trying to use this, so we need to add it here.\n\nChange-Id: I4380aaa7fbad005a1b5081c36afec30d58806580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35a13f4105e06b926f031bf6fbd2199480f75bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35a13f4105e06b926f031bf6fbd2199480f75bdc"}]},"branch":"refs/heads/master"},"a5493558e86784455285c9c6ec90af5f753fa069":{"kind":"REWORK","_number":2,"created":"2020-02-26 20:42:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/69/708369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/708369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/708369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/708369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/708369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/708369/2"}}},"commit":{"parents":[{"commit":"f7aa34359c03ffecfe74a5377839e7f8904b32cd","subject":"Merge \"Set tox_constraints_file for cross-jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa34359c03ffecfe74a5377839e7f8904b32cd"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-02-18 12:12:50.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 20:42:14.000000000","tz":-360},"subject":"Add oslo.limit into global requirements","message":"Add oslo.limit into global requirements\n\nNova is trying to use this, so we need to add it here.\n\nChange-Id: I4380aaa7fbad005a1b5081c36afec30d58806580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5493558e86784455285c9c6ec90af5f753fa069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5493558e86784455285c9c6ec90af5f753fa069"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
