)]}'
{"id":"openstack%2Frequirements~708396","triplet_id":"openstack%2Frequirements~stable%2Fstein~I1e73ebd0717e157aa0196e6c352536479118bc55","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I1e73ebd0717e157aa0196e6c352536479118bc55","subject":"update constraint for neutron-lib to new release 1.25.1","status":"MERGED","created":"2020-02-18 14:25:07.000000000","updated":"2020-02-19 01:13:33.000000000","submitted":"2020-02-19 01:13:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708396-1582074813569-3390009c","meta_rev_id":"0f86342d2cffdec709824fee3ac6d61f8da7c702","_number":708396,"virtual_id_number":708396,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 01:13:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-18 20:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-18 20:50:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 20:50:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-19 01:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf63a8c653a3e905e3ebe895b9ad7141ed003386","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-18 14:25:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"169c2dc641f400a8ec433fb731f2c3e37e20f736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:44:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/063b907ff8334746af114c918edf6ce2 : FAILURE in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc0e41218944c38b4949f9e294ce6e7 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9722212b02194bea9062144e8bf6f5b0 : SUCCESS in 5m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8f51073ef924c3f8049cfc3d7ed98ba : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/394007a8bca548859e321b7eb7eafd24 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f56a42b1818c485b907aaa925dd9c46b : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/151a2fa7ffe14f7eb5085c47f15f1bbf : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/185c9f64588a444ea678bffe96b0fe4f : SUCCESS in 9m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/206c085cd94f4a2ab834fbd2158e22ca : SUCCESS in 10m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0d05c06bc5640d8863d8816d21e15cf : SUCCESS in 9m 09s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5672a84b22bf4478992b4ed7eb4b6292 : SUCCESS in 12m 09s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bbc81b5db44b4883a402d523b0f8730d : SUCCESS in 12m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0e8318215a394ca3825196154f3cc986 : SUCCESS in 6m 39s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4d7659d1b45f45c7bf0c5d30d7570055 : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d6413795f7e04dc6887e386ac9c70dbc : SUCCESS in 12m 11s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5af27141f70a4fbcb721e19f9e779ec0 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe1ba5aebe114c5a92f819a67a8c45cc : SUCCESS in 8m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/04d918abd29244e89c338f5c87ecb0f6 : SUCCESS in 26m 21s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e703f38fcfbf40e4a9ef1c2bbffdb318 : SUCCESS in 24m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/be965c8a6d904dc2807efaddd9628819 : SUCCESS in 20m 38s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/3e26d1f2eee144ed8c949166bdf377f6 : SUCCESS in 28m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/00bf575804644d28b4cfbcce7a138b6f : SUCCESS in 18m 08s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d4c37975f768402096e532bd6f48e236 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c876629e9b54475b84c5846aaa4eda3c : SUCCESS in 17m 10s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a092ba0f5a0a4de7a28f585e8fe725ad : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3bb0def1685849a0be41b4a1d6e1ea7b : SUCCESS in 13m 13s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9377a8f8809044e88ecf464cfeedbb49 : SUCCESS in 10m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4d96455031443069480edf2f9dca117 : SUCCESS in 1h 07m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d84249c3dd14063b72a1b94a9709abd : SUCCESS in 1h 35m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4c48fc3bc8474d238f8af64169dc76b6 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b0686b0acabf81d4d92b8f3edde51307b607aa15","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-18 20:50:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc6400ffdce70a9e793d00243ba2a27ee2483887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:51:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77713dec534fc066692467556a35f7953d79b689","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 01:13:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d30946b2a588189624bc8f043cfebf277c85b7cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 01:13:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8bf6466054424d8d3d3acf6c738e18 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8ee293b0e214a26ad9b6fe5ed025ea8 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6fee2f25156b4bc1bcc24ab43594d230 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e3706344d7486f8de4ffee793f62a9 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1340c738102f45b9b07f6f91a48bfa59 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb523ba8854b4f11ba9d0cd24242ac15 : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/08a2258c8c6e4cf893dddb1dd90a9a9d : SUCCESS in 9m 37s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4f445761c09f442c9447c97ce100e24d : SUCCESS in 9m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/96467595cf104c8daa22761e46c9866b : SUCCESS in 11m 06s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3eb87d1bbfc74bcc8bd2e73b6a821eb2 : SUCCESS in 10m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3f0b333c4e4c4406b5beed008152d445 : SUCCESS in 7m 11s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b7684589db3b405eb56ce1a2d7816952 : SUCCESS in 7m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/01f9b2658ff244ff9efb7d4bd331a14d : SUCCESS in 14m 51s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d9ca7765ffea4aad94e9cc7885927a1e : SUCCESS in 13m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa4d5395dadb426c852bb96dc4f2c1de : SUCCESS in 10m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/19afcec371e94ce99fc16296f23b2600 : SUCCESS in 24m 31s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1fa53c509a764182ad8397709ac70a38 : SUCCESS in 26m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f4b7dbb9ae3346eaa0f99052dbc5cb30 : SUCCESS in 26m 04s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/161b9fb17ca946d88d1b89f02a6e2e1e : SUCCESS in 25m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/441dab68edc5447d8e083c8bcb4bc08e : SUCCESS in 16m 22s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/33f0d57815ae4bf6a07d368b385aef4d : SUCCESS in 19m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d51acbef41b9490899b5bf38b7a50ce1 : SUCCESS in 22m 18s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/967ba3c744264ce7a6252a42d39b7cf5 : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7dcc70d5104742119edfa12a82e67095 : SUCCESS in 11m 27s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/dcc512708a6c49298f74ecd5c6a9c5bf : SUCCESS in 10m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a36866b61b42407b9a93949927e64562 : SUCCESS in 1h 05m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc3d4e1c752141c1b24a972d0272ed4b : SUCCESS in 1h 41m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/21c67c535afa42f6ad23f3bf171f807f : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49396e20ea3427313e176c7519755fcf9be6844e","revisions":{"49396e20ea3427313e176c7519755fcf9be6844e":{"kind":"REWORK","_number":1,"created":"2020-02-18 14:25:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/708396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/708396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/708396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/708396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/708396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/708396/1"}}},"commit":{"parents":[{"commit":"6f8536ee0b2c4a2f53cf21ba94380dd3fa0ff81b","subject":"Merge \"Add neutron to stable upper-constraints\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f8536ee0b2c4a2f53cf21ba94380dd3fa0ff81b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 14:25:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 14:25:04.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 1.25.1","message":"update constraint for neutron-lib to new release 1.25.1\n\nChange-Id: I1e73ebd0717e157aa0196e6c352536479118bc55\nmeta:version: 1.25.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: Ia95f2c867226870214e4b2d41c78ce7ebed6b338\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49396e20ea3427313e176c7519755fcf9be6844e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49396e20ea3427313e176c7519755fcf9be6844e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
