)]}'
{"id":"openstack%2Frequirements~708399","triplet_id":"openstack%2Frequirements~stable%2Fstein~I9110eecf9866fd3bccf92d17b9bd10c53aae974a","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I9110eecf9866fd3bccf92d17b9bd10c53aae974a","subject":"update constraint for python-neutronclient to new release 6.12.1","status":"MERGED","created":"2020-02-18 14:28:58.000000000","updated":"2020-02-19 01:13:37.000000000","submitted":"2020-02-19 01:13:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708399-1582074817461-2176cf01","meta_rev_id":"5597fdddf018e519b88c10b80fcbdeaeaad80a76","_number":708399,"virtual_id_number":708399,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 01:13:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-18 21:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-18 21:30:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 21:30:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-19 01:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc50d43d1c0401c6f2df3546faaa22fe370f18cb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-18 14:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0365236fde6aa9596227d27d31b68200a232c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:16:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/21966b052a6e4a5cb12bbf4ebec4aa9a : FAILURE in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3a0e69bee44d38b3fb2b2a668e2a22 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/191b9cc0b44742a78404a6210dc95bd3 : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5864cc19450f461ca33bc6cf6e3cf2ad : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10841936ebf940b6a6c6488fbff22eba : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fb33df6394c43988f4f77cc25b0529f : SUCCESS in 7m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10e68dd784c941628095ff9dedf9c780 : SUCCESS in 6m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f946d53271124217a94b9a1c31491c2e : SUCCESS in 19m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5b21e40736c54a5ca820d9a1410fc278 : SUCCESS in 20m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d3af2350dbf486e989f6ac133b3377c : SUCCESS in 19m 00s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d2c9c0b150cc4332accc17c3d3f49a08 : SUCCESS in 15m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/411dfdb8110c463c96d84462fb796e74 : SUCCESS in 15m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/daae400121e247f7a041f0dcc0fbafc6 : SUCCESS in 6m 33s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c5652b2e78c64649bdac9d4b1eb0c2e0 : SUCCESS in 6m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ea43dd9c22e34bcf809362015e23124a : SUCCESS in 12m 00s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8ec825d028104227b3bc557676378091 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54f5e79bfa7a40b38457ad897ae821b2 : SUCCESS in 9m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a7be8a2768254b2fbf86dafc0c2229eb : SUCCESS in 24m 11s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/82b52d69fcbe46af9c1c69130d888d7a : SUCCESS in 35m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1c3f88a4d22545dda87888d00beb85c5 : SUCCESS in 23m 06s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/997594b3adc04ca78aa7e6a4fd9ecdbd : SUCCESS in 29m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6ec331818c3c4396ba383b4a0880d8ba : SUCCESS in 18m 20s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b104bc2455f247d1bbac3340beb31788 : SUCCESS in 19m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42b76397eeb7455b954d3d9731055444 : SUCCESS in 25m 03s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f3292092dd6246bd93a1552989b5454e : SUCCESS in 7m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7e5503dbdc848e1beee64466e1e797f : SUCCESS in 10m 57s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e99b537e131d4a6a92f9a0d4d798fcd6 : SUCCESS in 10m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7ae41b0faa04d14a9a80802d05a22f3 : SUCCESS in 1h 06m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3aab68f26244fd796261cf996bc20e4 : SUCCESS in 1h 43m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/13930b2942c64f37a9ee6273517cbb73 : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a2a671b1d9f673aa714479c15dc5cccc2026cbac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-18 21:30:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"145c5301a5a5c9eecc8b1c2be7c3d55ccb3f4153","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:30:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4564599599237416cfe7ac98ed4bb9d38b011a84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 01:13:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6026e3c5c811ebcaf0522e28298e6a755ad01aa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 01:13:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6031692780439caf0571b79b14680e : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1a63eb0ea1949d2b19b9d441940cd4e : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/14e6653f05824e64a9db1d1a0a8ff225 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0e9be51216542259023db986c8e3d13 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62bfd6ccf2f6445480a9d28b0a190885 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e90778fbf71c4701b47bb80966e07851 : SUCCESS in 7m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a3353ca323314e9fa4fa31a98f8d6572 : SUCCESS in 10m 19s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4991d9dc35f94e5ba953c59035125f2a : SUCCESS in 10m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c60550e148e24abe9d144712612f90a7 : SUCCESS in 12m 01s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7aae30ef3aef4f94b0e6d7eb756d5642 : SUCCESS in 10m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cc9df5f981084c86a4fe5101267c6c3a : SUCCESS in 7m 07s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bd6334df170c4f1da1d5b96dc5e79bb4 : SUCCESS in 6m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f9e03aa7b48b4b56b28ebd6d413da95d : SUCCESS in 12m 44s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/999f83d1a5c84ce5b7ef410f2e3ee853 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/830ce737301a4dda90bbcf2c310f40ee : SUCCESS in 9m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bf55f7b5d9f94d49ba5aa3474d8678cd : SUCCESS in 21m 58s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5d8756fed1164881b293fecfece2a5c5 : SUCCESS in 32m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7929c1daf8204861a71230bf71fdcc00 : SUCCESS in 31m 38s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/28b438e53cc14942a04e04258b00b922 : SUCCESS in 28m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d1e7523c413a49e0b31b9ff8a2246785 : SUCCESS in 16m 05s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1064de1bf9b34cc0a40e4648acf9c76a : SUCCESS in 14m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57c2ed79e0c3470894482acdb8ae27f8 : SUCCESS in 18m 32s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2eb6c985708e4caa8099492cc985597f : SUCCESS in 6m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18311383ef3f4a5ab976fb1981018dcc : SUCCESS in 14m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/40438b141a174392acee1b1d34fb2c83 : SUCCESS in 10m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0908bbaf4a56481a87a902aa4f5bf741 : SUCCESS in 1h 17m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/35a14fef03cb4fc3b280d51e737802e0 : SUCCESS in 1h 26m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b318495c7fde4cb0bb880b1bf4f71de6 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8f4ecf4f3124677638380de9e2cd6c34e737cd3","revisions":{"b8f4ecf4f3124677638380de9e2cd6c34e737cd3":{"kind":"REWORK","_number":1,"created":"2020-02-18 14:28:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/708399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/708399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/708399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/708399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/708399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/708399/1"}}},"commit":{"parents":[{"commit":"6f8536ee0b2c4a2f53cf21ba94380dd3fa0ff81b","subject":"Merge \"Add neutron to stable upper-constraints\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f8536ee0b2c4a2f53cf21ba94380dd3fa0ff81b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 14:28:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-18 14:28:57.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 6.12.1","message":"update constraint for python-neutronclient to new release 6.12.1\n\nChange-Id: I9110eecf9866fd3bccf92d17b9bd10c53aae974a\nmeta:version: 6.12.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: Ia95f2c867226870214e4b2d41c78ce7ebed6b338\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8f4ecf4f3124677638380de9e2cd6c34e737cd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8f4ecf4f3124677638380de9e2cd6c34e737cd3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
