)]}'
{"id":"openstack%2Frequirements~709033","triplet_id":"openstack%2Frequirements~master~I1641885203ff1ed395c7947eec85acfc0ef2cae9","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I1641885203ff1ed395c7947eec85acfc0ef2cae9","subject":"manual generate-constraints update","status":"MERGED","created":"2020-02-21 03:12:47.000000000","updated":"2020-02-21 07:44:33.000000000","submitted":"2020-02-21 07:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709033-1582270931821-d0314534","meta_rev_id":"a53c6cb882ec1061266ea9514a1323e47af41cd4","_number":709033,"virtual_id_number":709033,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-21 07:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-02-21 05:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-02-21 05:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 07:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dcdc555461678709babcd888d22e4c3329481314","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 03:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8fbe3dbfd5b5a42d92707035ee991f7bef32158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 04:07:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7421e9861ee349c5ba13acfff5f698b4 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe984873d6974dc5bceec20107f09497 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1e4127e90224f4c9bc63a65cb4aa7e6 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1321d151a134fd38b6df0da4609ee92 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309a5075ee0d4f2f82cdf12beada0f44 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37867fcd6e8944c4a10958e087298428 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bad6e564fa654f0b8b20d06554f88be6 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d83e3fa97f9b4aa1932680a7df4c0cff : FAILURE in 7m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/512f7438f4534750948c48782d8a010e : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0fdcbcd069784ee2b2211bca9d524234 : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e4542466a604329af376b1673ef6a59 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2371c22a74154c82a9cd19d747a560b6 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a873dbe9aff547da9f5d0cf0aed8e773 : SUCCESS in 42m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f7625762c05a4169a40e4a660a453848 : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1e22e4573b4941e3afae4a9083afca31 : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c9bec0d24af54915a69a40126443ee98 : SUCCESS in 14m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/227c9f9575f041ee82a5c1b707407b86 : SUCCESS in 17m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2b4f9ad004cf444ab5c40e07188f69d2 : SUCCESS in 4m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e393acf9fdb40ccaf54bb3829e0b20c : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2873fb57a3524ae58781de4642478e9c : SUCCESS in 8m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/83c27ca8349045298d858b2cf77acc90 : SUCCESS in 12m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9459dae12e7a4024be3a58d5dca7f2b7 : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e056b580ec0544f9a92f2f3a97f354fd : FAILURE in 30m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a65b7cdbadd4d0eaa24a69b010f56b0 : FAILURE in 52m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11bfac2f3bc44c8eb5efd21dc5113f7d : FAILURE in 22m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/391f946def824aa78cbb47699ae1611f : FAILURE in 47m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b9d5b703d08ed48caf29d59c8316c516eb471e13","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 04:08:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83d31721e9cb8012c02e5a13b136d2ce8e77c742","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 04:27:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c49d3c408aae843ea9afd5598904e6c18452432a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-21 05:59:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"830f2ea0aad98c705fbfdba9dd25015e48dbbd56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 06:12:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce008aa57008491caaad9f900ebeff02 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16dd8ebaf92944debffba5f4c7bb1446 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66ebf1d0ad764a67b23350e9a7e1ef0c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f59d93b0f0014cc3825222ad5bc69e9e : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e69778f8d2d4a97ad80923f2a36c65f : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/190c70bac2c14bb3989ce79765c4ae55 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15d173a070754a3b90442fcab1072727 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8954d147de114e74958c49a48e0f45e9 : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82f78d219bda4582aacdc496a95ad035 : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/19137b3e794345ce8f90856d92867b7a : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2afc77fa6d69457ba88e4434a2eb96a5 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/daf403bcbcea4e2b8d45e51cf70c60e8 : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a47ef247c1e94d7a90a937f57ea7f147 : SUCCESS in 13m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5e6bf9ad81a4210852fc2bb207c38d8 : SUCCESS in 25m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f540373ed444cb192d6c21f369f0555 : SUCCESS in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ec1c7de3e1994a1b9e9c5b5536e53530 : SUCCESS in 12m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6305a0f074c347e9ab2a1765a3074168 : SUCCESS in 17m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/048b750944bb47eaa728f534cb20eee1 : SUCCESS in 8m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/32248b20a8904b8e961129198d820afa : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5da47b261a704d90a9d448f3adf8e0e3 : SUCCESS in 8m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/72a1389aa31642fea1fb96846be4edad : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44752007a31f43afa305c42a3df969d8 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48f1830905e24c94a58e7c065774f7af : SUCCESS in 57m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8988c66825d4acd88d8f32fd674ff18 : SUCCESS in 1h 43m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b8ad2ea79494a199b2ee7417a5bc357 : SUCCESS in 1h 24m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/68865b33db364827a74471691497e9d6 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":3},{"id":"c4b5974a493d5cbbc4ca60146c87347df6ecc5bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 06:12:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"440161afdc4aab23f8cd28e874bbeb1a79bbc66c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 07:42:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b76d959de7e188fd09897a26e09400d3be4fd9b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 07:42:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c30e32a60724e04bf2b34aab924b00d : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a97c87cf44488fb5ff77760f97492d : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34311382f00b47b2958d64212493672f : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18e547cbb7284b609a1d93faee2ddee1 : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f1375fc14874e5384c336dd1b9ab2f7 : SUCCESS in 6m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07d59bd361874e1695f508c102a565d4 : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc4cc091567c44f0a602667f925b1091 : SUCCESS in 17m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba1b5f0336d947209670593d2d11fe83 : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/054faf1a890a4843bb2d9f5e4b09e609 : SUCCESS in 7m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/511ccdfcb17348f99e58435a22202508 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b63ad6430eb1492dbc22232d3ea119f5 : SUCCESS in 9m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/faf91361b03743d6a2d8837a18ba3159 : SUCCESS in 18m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bad5082dcdc84cecb8149b67f2927e8b : SUCCESS in 29m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/30984d8a490e42f6b17240a6251aa024 : SUCCESS in 15m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/07ca02c4cc514cba9fb188f2398407ed : SUCCESS in 15m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ea702e05a2a495b97924d438c3e4fa9 : SUCCESS in 21m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9b831cc304ed4c5185acfe8b2acab717 : SUCCESS in 10m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f235f33f54ab4de6b26f4c02a882a640 : SUCCESS in 8m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09e0d1371dd84a0690df7a880029b4c6 : SUCCESS in 9m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dd37e8c6206946f79a711fa95158653d : SUCCESS in 12m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a7ae90dd44d0452abdd152aabe51b452 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/642e51514ba34044a0bc0185f28fff3b : SUCCESS in 57m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/46ddae705ce14d1a8dcecf376a24bfe9 : SUCCESS in 1h 18m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70ae440062534d24b4c1ed2575cf5975 : SUCCESS in 1h 17m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0403ba467630461aba3a54f0a5e1310d : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a53c6cb882ec1061266ea9514a1323e47af41cd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 07:44:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852098cc07b8423aaa72697ab06c986a : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6","revisions":{"e67142481a4c4457f718c5fb6997d6a18575ec44":{"kind":"REWORK","_number":1,"created":"2020-02-21 03:12:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/33/709033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/709033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/709033/1"}}},"commit":{"parents":[{"commit":"34c402d5e537c6ec8aa640162744d4a77fc0edea","subject":"update constraint for python-neutronclient to new release 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34c402d5e537c6ec8aa640162744d4a77fc0edea"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-21 03:08:05.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-21 03:11:18.000000000","tz":-360},"subject":"manual generate-constraints update","message":"manual generate-constraints update\n\nstorpool trys to do stuff only root should\nhttps://github.com/storpool/python-storpool/commit/9f072befed9850643f5c31ee789f55b1d6cfa7f4\n\nhold back alembic for nova https://review.opendev.org/708706\nhold back tinyrpc for https://review.opendev.org/692915\n\nChange-Id: I1641885203ff1ed395c7947eec85acfc0ef2cae9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e67142481a4c4457f718c5fb6997d6a18575ec44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e67142481a4c4457f718c5fb6997d6a18575ec44"}]},"branch":"refs/heads/master"},"3e97d8bc743c4808aedaaddae1cfac1a412b7fcb":{"kind":"REWORK","_number":2,"created":"2020-02-21 04:08:54.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/33/709033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/709033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/709033/2"}}},"commit":{"parents":[{"commit":"34c402d5e537c6ec8aa640162744d4a77fc0edea","subject":"update constraint for python-neutronclient to new release 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34c402d5e537c6ec8aa640162744d4a77fc0edea"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-21 03:08:05.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-21 04:08:13.000000000","tz":-360},"subject":"manual generate-constraints update","message":"manual generate-constraints update\n\nstorpool trys to do stuff only root should\nhttps://github.com/storpool/python-storpool/commit/9f072befed9850643f5c31ee789f55b1d6cfa7f4\n\nhold back alembic for nova https://review.opendev.org/708706\nhold back tinyrpc for https://review.opendev.org/692915\nhold back storpool because it break tests\n  - error: could not create \u0027/usr/share/doc/python-storpool\u0027: Permission denied\n\nChange-Id: I1641885203ff1ed395c7947eec85acfc0ef2cae9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e97d8bc743c4808aedaaddae1cfac1a412b7fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e97d8bc743c4808aedaaddae1cfac1a412b7fcb"}]},"branch":"refs/heads/master"},"825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6":{"kind":"REWORK","_number":3,"created":"2020-02-21 04:27:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/33/709033/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/709033/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/709033/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/709033/3"}}},"commit":{"parents":[{"commit":"34c402d5e537c6ec8aa640162744d4a77fc0edea","subject":"update constraint for python-neutronclient to new release 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34c402d5e537c6ec8aa640162744d4a77fc0edea"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-21 03:08:05.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-02-21 04:25:43.000000000","tz":-360},"subject":"manual generate-constraints update","message":"manual generate-constraints update\n\nstorpool trys to do stuff only root should\nhttps://github.com/storpool/python-storpool/commit/9f072befed9850643f5c31ee789f55b1d6cfa7f4\n\nhold back alembic for nova https://review.opendev.org/708706\nhold back tinyrpc for https://review.opendev.org/692915\nhold back storpool because it break tests\n  - error: could not create \u0027/usr/share/doc/python-storpool\u0027: Permission denied\nhold back msgpack because salt...\n  - ContextualVersionConflict: (msgpack 1.0.0, Requirement.parse(\u0027msgpack!\u003d0.5.5,\u003c1.0.0,\u003e\u003d0.5\u0027), {\u0027salt\u0027})\nhold back docutils because boto, again\n  - ContextualVersionConflict: (docutils 0.16, Requirement.parse(\u0027docutils\u003c0.16,\u003e\u003d0.10\u0027), {\u0027botocore\u0027})\n\nChange-Id: I1641885203ff1ed395c7947eec85acfc0ef2cae9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/825c615e5ae2d71f0bf0fbeaf9a5de377c09adf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
